)]}'
{"id":"openstack%2Frequirements~896018","triplet_id":"openstack%2Frequirements~master~I5b4c6f7b2399d3a9d6c8414806c1cf35677e50fe","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-09-23 10:14:30.000000000","reason":"Change was abandoned"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-09-23 10:14:30.000000000","reason":"Change was abandoned"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-09-23 10:14:30.000000000","reason":"Change was abandoned"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-09-23 10:14:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5b4c6f7b2399d3a9d6c8414806c1cf35677e50fe","subject":"Revert \"Bump Sphinx to the latest version\"","status":"ABANDONED","created":"2023-09-21 10:29:31.000000000","updated":"2023-09-23 10:14:30.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7d83cbf1bf7ede87c7b9bb990249ce959279f9cd","_number":896018,"virtual_id_number":896018,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-21 15:10:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":-2,"date":"2023-09-21 16:29:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2023-09-21 16:45:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":-1,"date":"2023-09-21 14:31:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-09-21 12:33:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 11:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 12:33:29.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-09-21 14:31:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-09-21 16:29:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-21 16:45:43.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2023-09-22 11:35:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-22 11:35:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"14dd857fe8a80b55c6602eb82041af139e7cabcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-21 10:29:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"224ec49665caf58eff921d27169f74b795b330a3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-21 10:53:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d048d5774d7f059a9c74cfca1355e500a1f5ee94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 11:34:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11d4634f260f47a0a88b80142edf7b7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/accc97aae6834197b71d35bda00e3800 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c3f810e4e0541d4a4f10d7644ae7425 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2469d8e2d4c643d7bafc6188f6e99398 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42d299990af144d0bc7a79c0a5e969a1 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16d7f1ab29cc41c0a00cfbbeb000e4c6 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5dcf79522b6847be9e044d12eb679849 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/899b8ea2e39a486bb08c5e764cbc67b3 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9400ac605b44bbb8d82a1d5ef30ee69 : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0fa5377579de4d12b15c7be6fb17dc15 : SUCCESS in 9m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e320bd83b274c32a6ec527b925b83a7 : SUCCESS in 16m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/05adba400be74b5eaff27578fe4f6266 : SUCCESS in 9m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4784ee4882834ae495f08b60784593f0 : SUCCESS in 16m 08s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/282e096da62941fe882f3ea557a65706 : SUCCESS in 7m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7d951eed131f48088ee8f4fd1d0fd764 : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/96a38858adf644c38f704d85fd1c1492 : SUCCESS in 6m 09s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a44a9fedc7ce4bc990a6869ed334f634 : SUCCESS in 12m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a3bc306e8c7d4479a0383be46cbf9500 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9c0e8ed0cabe40cf969f1644843a2184 : SUCCESS in 5m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/56d5198b70004ad1a2a5b28969650a44 : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ce6960b01517419d8e9bc1c912c45f03 : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d7688e4b5534860bfd629fac144f515 : SUCCESS in 9m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0364be4d63b74106aab2461f823a4ad9 : SUCCESS in 8m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b59fc1a838104d20879ec55a08e79aff : SUCCESS in 16m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b852c105c23d42db908d1c130bd9cf8f : SUCCESS in 4m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d7ae0657856b456c87cfd058f23b104a : SUCCESS in 18m 38s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/140b37e7ddc345cda265382ade70f686 : SUCCESS in 5m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d7ee08e69f5c4f2ab0ee344b8561e812 : SUCCESS in 9m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5df5f0b3280f4c538fe9f8977ae6928a : SUCCESS in 25m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b5b609742d9c471485613af10cb4449f : SUCCESS in 10m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e835ade720084c91b51f4a2a958379b1 : SUCCESS in 13m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/df00eea6a5d34407a6deecf725f68e63 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51e9d86e9843434eba2772c3fff1d997 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f012d63d144a4a0f8fd20590bbb5986a : SUCCESS in 26m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7c6492efeb7e49be8628b11e38711fe7 : SUCCESS in 6m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/203dd3309ee74ce49a782a28179960fa : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec4b8f619eff4736af874c212920399c : SUCCESS in 7m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/26a53a21befc44f0af9f1d6453b9f078 : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0b913084db94620bf7b1dada5ce187f : SUCCESS in 57m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dc305dac88748ab892abcaf4a56a51f : FAILURE in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bc832aef0a23d15d3b37ed14afdf61eb1fa1fa38","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-21 11:36:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a01ffd32a1d341f2969332faa77cb22435d14a0","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-09-21 12:33:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd19554d7e7afd39574f6930c16367fa68d3e941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:16:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d242e29b04e04d8eb1a07159dc3f66a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7123922401fe4a4299ef94f2bafe9961 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf89bbbea204ab69b39cfd17decab1f : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd27a66565945028e84661d4bd02383 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9f502bf52ea495b8fac35d18f51c80f : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9977027a2acc4fb1a8cdc470b75d1dc3 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eac7494a8b4a4d75aff6736ae65cb5b6 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/352c12217ce24b1da99d53d82b67d0ae : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e78b1034d606447d831d2fc6e0fd1a3e : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28c3dc76b1734f53a5a437b4a22a174a : SUCCESS in 10m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7482317f6bda47bf8f36dceddad2f0fc : POST_FAILURE in 13m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dac63d1df8774225a7732cb30e5d5f1d : SUCCESS in 8m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c1b8c764b37c4c14a0fcdcdeea037b02 : SUCCESS in 14m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ee759c8da6b846b0934f72a1a5d66fc3 : SUCCESS in 7m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3a01e6a9291843cb9a638408c2214f45 : SUCCESS in 6m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4560c25dca4a4ed798aa4b3554304080 : SUCCESS in 6m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3fe1ba94883a44fc9c165186c63ba63c : SUCCESS in 11m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/50a7693851764a6d9ce17c026274f33b : SUCCESS in 5m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fcd4a1ad01fc446ea74cd51316128da2 : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9e106fd90c824a89ae7266593a580a68 : SUCCESS in 8m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b22c8ce3a51d44b3bf8b002c0445bb12 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dae6bec0611c4e0a8f8df95cadbf7a84 : SUCCESS in 8m 13s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/15a3baba86da4262a6ebd63070691e1c : POST_FAILURE in 8m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0452d7bd11c84488b99e92abc76c6b4e : SUCCESS in 13m 39s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b65a9b22c9d0477cb7b15d699a6eb174 : SUCCESS in 4m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/dfeb93d9b1294420bf9311a61b4e14fb : SUCCESS in 11m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d0c9735749804b71af70c6bf07f3da82 : SUCCESS in 5m 39s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1c47598f0efe4f5d9942d66697e0a911 : SUCCESS in 12m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5869d33e5e144fe78a0c62e87d55df63 : SUCCESS in 24m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9030622f68cb45d59cc65ba20fe96b57 : SUCCESS in 9m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3bf341a9b1f8401e90b36f1be3edb729 : SUCCESS in 16m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/45f9b53d500a4a03b843e5d9c575f24c : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f081b4677e75452f98e40e380f6130fe : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79072d08b1e043459805e60b7a151caa : SUCCESS in 22m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ed8be2399eb14c55a5fc9965b9c85046 : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3b6fd09a34de4fc8b55887c2cd816ba4 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e60c593c38ec4c4bbf94824e5078c6cf : SUCCESS in 6m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d69937ac0b66464da86a513f153fcaa6 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bea365e2d28416795787a6be46d96f6 : SUCCESS in 1h 03m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4e583756e6640d6b464c7a7490ed6c9 : SUCCESS in 1h 34m 26s","accounts_in_message":[],"_revision_number":1},{"id":"64909479383db2fe067457ec68a3df15d22cf4d7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-21 13:34:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"011ba42f9c840107110a5d3e4e723a4b6e7661d2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 14:31:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"074b1530bfd0078e08a433434c67e5d3e712843d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:10:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab3fb4c40bd445fabaed6492a90611e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f149e1c1596e427b8baa7bc9e9e8b7e7 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9d338bdb9b4d5186e7f4dca8e5d294 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a408abb92c65428c9b07504562e8d7e5 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e7ec217d02e49bea12faf5ccb69c30c : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/014379e096694755b6950248527c50b5 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf18888ca4d64a0ca9b98f03bf95c31c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08e5b5bfe6f243c3afa55c8ad8168b89 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be7692d96b31467fa2a948743b6a170c : SUCCESS in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0a0f24b5128f420fbdc39f604b1252f3 : SUCCESS in 12m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38dab6db30b74509b8ba99485558ccfe : SUCCESS in 13m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9a72a6d7dc1b402f8604dfa349dc98c6 : SUCCESS in 13m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ab0c81b88494b3088020bf3befce584 : SUCCESS in 11m 06s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3da76cd4df5642f192d3857ab043b6d0 : SUCCESS in 8m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/02e921f9fa3f44228d01c13d599a20b0 : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1abe2493a3b64fb68d95de304e46b36a : SUCCESS in 5m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fb87b7e83e75479fb4016a0675e85306 : SUCCESS in 10m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d4932ddf0f204cb387f2f08175b07601 : SUCCESS in 6m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/91e9bec9ccc1427dac18baa708a0637d : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dd439dd36c1e4175acc3be27c3862638 : SUCCESS in 8m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/88b964448a9c4daca3403365f96a693b : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d57b1757b968456597565d9a35953486 : SUCCESS in 9m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/96095558b9ed4ece811e9297cf767f76 : SUCCESS in 8m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bee3f8da44104617ab3a1e23e293ef70 : SUCCESS in 13m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/69db620eb1ab492d80a79dc43ba79b12 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/93a4f8c9cc5b4206ad7eab3e6e605b48 : SUCCESS in 8m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/03747594b6714fcfb3b8a66359322c53 : SUCCESS in 4m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/06535d34c15c493c8b2398735cc40309 : SUCCESS in 10m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ec847c2ed6cf4305957c6f8cab0cba51 : SUCCESS in 32m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/153ef677eb9546318a34a6503f480fc0 : SUCCESS in 12m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/24b8d5520c9d4f88bee2ac2f2e312ff6 : SUCCESS in 18m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f632c552faa546988f65f9901aeb7655 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23ced79df7d44674aa7fcd3c28571419 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f8bc4bf928eb42baafafd4408ece05de : SUCCESS in 25m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a68063e651304ea18e55a496c31bacd3 : SUCCESS in 6m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ee5c436a25a849a09b405c55a36c72c0 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85b32901b1cf45a6bedfed16cfb01996 : SUCCESS in 6m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6ae7dd16fc5046539eedad1ce8b24517 : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98389b4fa3d146c4a33c8a1f0a5381a6 : SUCCESS in 1h 11m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/feb9303da2ad4831ad6b040618709203 : SUCCESS in 1h 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f94644088e031eb7090ac4f370055f9c49870be5","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-21 15:41:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7384dde9b5756880eb36edd4d18f0ac03e45170","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-21 16:29:47.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b052967636ac59b681658eb4849a601eaa534e9e","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-21 16:45:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e494c9bc457586171a5a03cefbb18e04ef7d8735","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 09:29:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1b331288452a797495b278d572cf55b8bc5b376","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 09:59:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5e6363d3b3bbe502e7704978a7711bcf649f22d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-22 11:35:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d83cbf1bf7ede87c7b9bb990249ce959279f9cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-23 10:14:30.000000000","message":"Abandoned\n\nAbandoned in favour of https://review.opendev.org/c/openstack/requirements/+/896195","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19096177ea7df044e8e4cddedae05297b711a399","revisions":{"19096177ea7df044e8e4cddedae05297b711a399":{"kind":"REWORK","_number":1,"created":"2023-09-21 10:29:31.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/18/896018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/896018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/896018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/896018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/896018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/896018/1"}}},"commit":{"parents":[{"commit":"b23a8724d7a891b7d1312950968ab2ffb3af54f1","subject":"Merge \"update constraint for python-openstackclient to new release 6.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b23a8724d7a891b7d1312950968ab2ffb3af54f1"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-09-21 10:27:34.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-09-21 10:27:34.000000000","tz":60},"subject":"Revert \"Bump Sphinx to the latest version\"","message":"Revert \"Bump Sphinx to the latest version\"\n\nThis bump breaks docs gates, for example Aodh [0]\n\nwe should wait until wsme fix a45e0af97e2e8362277f160d0f2b737590e367d1\nto support sphinx \u003e5.0 is released before bumping Sphinx\n\n[0] https://zuul.opendev.org/t/openstack/build/5df111c602944396bf6e334e1bec4e42\n\nThis reverts commit 289feed31747cbd922ef9d4a8df9ca87084b10a7.\n\nChange-Id: I5b4c6f7b2399d3a9d6c8414806c1cf35677e50fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19096177ea7df044e8e4cddedae05297b711a399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19096177ea7df044e8e4cddedae05297b711a399"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
