)]}'
{"id":"openstack%2Frequirements~896053","triplet_id":"openstack%2Frequirements~master~Ia77ae93ce59f4e175aaee1b030ea21e6c394480e","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-09-26 18:07:40.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-26 18:07:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia77ae93ce59f4e175aaee1b030ea21e6c394480e","subject":"update constraint for oslo.db to new release 14.1.0","status":"MERGED","created":"2023-09-21 11:35:45.000000000","updated":"2023-09-26 18:08:32.000000000","submitted":"2023-09-26 18:07:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896053-new-release","meta_rev_id":"9d67fc122e8359aba07ee727a56138389d2cc3a9","_number":896053,"virtual_id_number":896053,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 18:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-26 15:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-09-26 14:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-09-26 14:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-26 15:58:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 12:46:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-21 13:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 13:52:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-22 21:56:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-25 16:19:21.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2023-09-26 15:58:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3ec1d92e421fbcdd029d54e3ea9113447c1eaa5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-21 11:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8957a587b7e23584f12caeec983c0eae9988304","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-21 12:46:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c53bcfd1397ab2b62382d4e0d41394231390b355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:01:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/830a78dd656f462fba4b777c8bdac646\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87838c8a1a214ac8b957849491ff4287 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0807be37a62448058e497aad3ef940e0 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd2a32fd53144873802c1b9006b00a73 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35b98c47de6d4f9b8ad67659a17c35da : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aecfdc4f33eb4618b8b9250184ce96dc : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cf118d9ed814a449830c7d7ded4e5e4 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a47605b091c4020aa6f10ec271a87bc : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0de9c5f1d84a4f7eba3fe3f18058997b : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fccf71f5bf924f3d905c80cb4b02c5a7 : SUCCESS in 9m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d8dd277eb9b343188360e39a3a6fda43 : SUCCESS in 15m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bedf3ffb2a2e42178199337efeb6fb8c : POST_FAILURE in 10m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e036ca253dd044e98d66e74c5b28070f : SUCCESS in 10m 45s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d1a3c9cafe3d48d6a0f963e97605aa40 : SUCCESS in 8m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/436107f0d8e0401c8b1df87dd6f487ad : SUCCESS in 5m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/516ee397f72146818a22e15840ac1d55 : SUCCESS in 5m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bfd6d42f6444438e9a9748e935ffe238 : SUCCESS in 11m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3887d920127d49a58e2e3db440cd8062 : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9a08414ececf4ad29e857c139a395ef3 : SUCCESS in 6m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/aa5a474eda75433cbcd430788d4b438f : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a78c9b8cbe0b4dcf87f2c91373107d76 : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5483bab176734730959d9bd5206020ab : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6808ab120d8d4483bb3e96548dc08964 : SUCCESS in 7m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f802192c96bd4bec80b3948761297e8a : POST_FAILURE in 12m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3cf93c3e8f3f46f9aeb2e7dc16167a69 : SUCCESS in 4m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/054317ca09f14fbaa7e14dd1bddb4904 : SUCCESS in 10m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/79244b0498424abc946b25eb583ed1b8 : FAILURE in 4m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e19e797bbe4b420ab445650bc628e0ac : SUCCESS in 11m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/23db3f318eb7483d860e272bf0da6e01 : SUCCESS in 29m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/976986beb8ce44c2b8dd4be07258e6ca : SUCCESS in 11m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/af4ffc09c7144f69be475060bd3aaa90 : SUCCESS in 16m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/53300e80592e450aaaeb305386c6cb63 : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0fa28e45de994f2abc30c46463a2e606 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d389325ac5148f4ad98f0a936048deb : SUCCESS in 24m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/263e4f1a33d4438a8f46ed25e604fb2c : SUCCESS in 6m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c044d5104b1b430f8062e162893c55be : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3135462a3594c309cdc3f79675529b4 : SUCCESS in 6m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/294efecd31e6482ea4aaf9cdb3fd70d0 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a2852c02f8147ed98bb3a1a43632f8f : SUCCESS in 46m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c456ad8140c433cac04aa0f9ff181bf : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"31259ce463f59c87197473cb28098ab396d288eb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-21 13:52:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3d159f494b4d972c6da3da1e2a23b232e3a652b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:26:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53fc5820d61747f7be1a3b7f194c4e82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74468b4fe4f84259b07b087e5ebad4cf : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d075d0daefb43268993fad2c1cf03b5 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ada6c5a511dd4c32a726557641c93310 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26185c00244a4892b949b4c1ba057c5b : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4491382cf28944bd87f7b02832e768b2 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a84d5a3d86fc4dd0929512585374f138 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/573e976b012a46d6afad3a4d75938d19 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a5b97fbfe3d49cbb3633f13b9ad9b10 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b6dd257d79a3426b83e11e5ca5218eb2 : SUCCESS in 8m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7f2d91939fba4679be46f9fc64518571 : SUCCESS in 13m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4e027d63e21747888053eff00ec2def7 : SUCCESS in 11m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/038a67471d5b410ea375e3acd7a82ab7 : SUCCESS in 12m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/019cd431c52148bebc4482fa65649045 : SUCCESS in 8m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7848a37ea97a4b1eafdae96036fceb09 : SUCCESS in 4m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/14e5def07fa441cb9ca83fcd667595cb : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/95e68ae8c31e42719b6cde9da2e6e313 : SUCCESS in 11m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d1b0f58b0cbc4e5699454e4cb0766f69 : SUCCESS in 5m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/41517bdc28a94788a40eb7c0c6100a85 : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ef7e2afa163f4e82ba4eae659e3cd56e : SUCCESS in 7m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/832c8c67a6f74fa0aff514dd02d22ac6 : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3e3ebbd520f480cba4c7852b7e9f10a : SUCCESS in 9m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0945ed502e8d4a60907a5856c9cb4ea2 : SUCCESS in 6m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0e2017a98217450bbcc181f50fbdf2e0 : SUCCESS in 11m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/34e069bea326485fa2286c79d4f35bf7 : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2b42fe186a5249579d5df88084347c33 : SUCCESS in 9m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ac35762719774f7bb8a91b8049a0874d : FAILURE in 4m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/88e554260a634df08c826f69490155cb : SUCCESS in 11m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2294f868c5074fc79b5d69d3e6a31032 : SUCCESS in 27m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a7d6468d33a04949aa8b3cde25dee053 : SUCCESS in 12m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c36a973015364d5e9501555e8ee0a382 : SUCCESS in 16m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/16a282302cd141f6afad127c32cf6c06 : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1ceae63451e349158d97bd80f6d62585 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e65cdcbfd0fb4da5b95bf73415bdddc1 : SUCCESS in 21m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f53ed8ed342341fe8aa095ca0a34dbc9 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4f015db72e4647d09d1560ae9fb877a0 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c7be3beb0274dd8ab8a4c5a54426a98 : SUCCESS in 5m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5369e6f4c756406991b61b58781bbb7d : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f9b9f761ea64fe0b2e8a88d946a9a3e : SUCCESS in 58m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c96e9c203814e73bf1c0268df8ab953 : SUCCESS in 1h 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"45b057f123b02f46800d0fcc8866bdbc7328f58f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 21:56:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a3501d2a51bb6ebe8d38f08fed1e89d51a7b261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:44:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c11b4bd965f441178e0b19da54e3d2ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304cb0c836154d9a91414c2fe7436672 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2e448b8bc44d34a01082777f81fb1b : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72cdf84fbf464590a7481972da181aa7 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a8d10865bc3441d987f09b244663b90 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d240d286c895418d909cfb021ad8a1a3 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3d9cb0c07ef4cdba5fe8c97f72389f3 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bd234964c3f4165a5305591b0fcd7a9 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c01aacf099ab44a7b181b7affad6ad2a : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdb78773e18b4837a5bb2052f44dbd22 : SUCCESS in 10m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/af24dcdc82934c9f9a942dfe8ad6023b : SUCCESS in 13m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/431289175d1e47c98a1c5b1fd2058ff7 : SUCCESS in 11m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5a3b05fee1bc44109a3e26ecb6820fd1 : SUCCESS in 15m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7f61ad1bf1a6486b8efc11a2e3fccbcb : SUCCESS in 7m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bda8b8e62be940f1b61e9cb0c940faa2 : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b3328bf91add42d08d94e8eeaa38390f : SUCCESS in 5m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e76d99ba25cd4d1aab59d3ea9ee85a9b : SUCCESS in 9m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fc7280fd9b5340788a66c694cbfba8ac : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4780cda2ce504241998166df149e0500 : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/088bec16610b4276a481d95c746b4c04 : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/86229ab05c434ae881575207c789e4bf : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c35d37799b51405783d949ce4725085b : SUCCESS in 7m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e22c704df2a3424fad7ca37bb3538484 : SUCCESS in 9m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5659367622ba4938a355b5518f40b76b : SUCCESS in 9m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7ea5aa8158c340418a50b2cbcefc3174 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/949715c8f8334b68899cb4e54b32927a : SUCCESS in 12m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4345d286b60041f09a324ca58b7f9e28 : FAILURE in 3m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0325ef37b81f4c0e96f2a507b0edc9ee : SUCCESS in 9m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9ad8964913224047b715b3c2dbb06049 : SUCCESS in 26m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/04284f4c962348b0a1988fb58cec5e5d : SUCCESS in 9m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2ba9bf8132d44e64a3dba5a692b30e67 : SUCCESS in 15m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/027790d97d554aedbe529cbde2ba0828 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9880e8afbd0e4f3abab722ddf3669130 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83c7ae93061741f190166b828319dc1a : SUCCESS in 24m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d14b6d18466b477c84d74cd94b22ffbd : SUCCESS in 5m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e400af75ecc448b2a3d3a8d5463298b7 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0cf3365a00444a1ba58184ed55dc807 : SUCCESS in 8m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/00f6ac1be63b49be8a8251c63311a552 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/282c4525e7964f53b1969d1d95d8a88d : SUCCESS in 58m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85373afd35684bd1b46d7beaf493b709 : SUCCESS in 1h 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0ccfea0b73a765d059f1d65da09c4eba200aa942","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 16:19:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3238274930245251d4587a4ae9b380f40e00dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-26 14:54:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d500c2f1e444ef00676fd176004bd0f0971db95e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-26 15:58:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"77f40c9b19e1099c8676359989c6c3da80398212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:36:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc4990eb935b42ccbdc04f4b1bebd879\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436bce12009c40a790c246be64e7a064 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db38eaaaf9ca4215b20a34ef2ffaa40f : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46803cc69f4940a8af5d3afe1b08b581 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d84303231de84ad8823aeace81abc2ec : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63434101058f4e37b2232772dbbcfada : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd69d9705f924528989286b67932d261 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1aeaaef1c07f42048d7466ff8bebd9b5 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0fdab115effa45e388925ad3dfd48c98 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83bde29d85ad4508b2d74226241eeec0 : SUCCESS in 12m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e4890675a0734256b1cad238ac754755 : SUCCESS in 14m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/53443f9c0f3441e7877928ea0b4c7399 : SUCCESS in 9m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/26fd7dd4036049cca68a4aee731cc848 : SUCCESS in 11m 09s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/480377f7cf24403ea9fb9800955010a1 : SUCCESS in 7m 23s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ed9b387e3ec545eea0f4144e784152f5 : SUCCESS in 4m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e4781d39cb814b03bc003c64bb9cd989 : SUCCESS in 5m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/048449d2a1c84610ab80c28aa9ce0845 : SUCCESS in 11m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7980bc19187946f682ec834d1b1ea7b3 : SUCCESS in 6m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/77b41052b91f4ee8ba519f68914383fa : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/851264491f3047c88521dd3181e3e266 : SUCCESS in 9m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3ed6d8649294494f8f5555ef1a752c3e : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5451359de0814dccb58ae8cc6f0a8508 : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1d5a928c0a0046c290509833ac8a3229 : SUCCESS in 9m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/50ca7f48cbbc4237be19fd09ef125e59 : SUCCESS in 13m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/aed2203ce6f14f61990a35c2f1f5f753 : SUCCESS in 4m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1444b27370f14835820604e236ca6377 : SUCCESS in 10m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/05d9cbbf0f6444718e66ccbe6d293247 : FAILURE in 3m 47s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/31f7be409a1749c983ead5ab6e4c1f72 : SUCCESS in 10m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6c724a2c71e3476cab60176704d926b7 : SUCCESS in 29m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f161051c257f4d9daf1fddd66a6ff5a7 : SUCCESS in 12m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3d18d609a7bf44349f898d992f0a58bd : SUCCESS in 14m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d9f67c9abf4e4f6f9473f3165248249f : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7f7847c866064e69b9580c8d5849e5a1 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90ab8d6be5304be6aee71e419ee32f5b : SUCCESS in 28m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/997f292598004484b486b90dae97e2b7 : SUCCESS in 6m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/42ec03c3c02f463e846316c999bc24ed : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56cfd8b99f1141be8264e4479f89a8df : SUCCESS in 9m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/baa7ec7ce18f42ecb55f97d964e2c851 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfcf7e04ddcb4ed294f0ed452a05f6fa : SUCCESS in 1h 04m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9c7b5205aab4a0ca9b00e14a125df88 : SUCCESS in 1h 34m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6db32f991152329cd8f9becb30949b765bb14c23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:37:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e97353c211c5e86ec2c6de6c1fb2dbfc15e4d35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 18:07:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef7844f3c684ebaaffd13a027d24003\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2d3de482ee486b9314ed25b9d97c27 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac4686108a845c899e00269f5d2db78 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9afc5c8e1ed64e46a502daedf3c64b63 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/068aaabf6ebe41bb94bdaf311fc64449 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/013af397526049fea5b8506321d88220 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b84f27365d746d4824c83e3f9107436 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7dda2ca8fd24689af3582fd9e03cb4c : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d594e379b1004ae794a96821a0004da1 : SUCCESS in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/abc4d9c1aab94d289cdd6592e61acec5 : SUCCESS in 11m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ba7edf390e3b43e08026f87c30d12b36 : SUCCESS in 8m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1f9f0a60ea974a46a472437ce4c4f4b7 : SUCCESS in 10m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d89f2fd6f53a42cb9b02675363bd8cb0 : SUCCESS in 4m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c922c5ff65144bf8964c79b1e2018359 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5463f1faeea14d4dbd439b525a6270c0 : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/419a481180294b589440373ef672d8f7 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bbc5a2e7d9504682a9f501a8fa66fafe : SUCCESS in 8m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/87663679b67044d784b8d5bb0abd406e : SUCCESS in 7m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cd2bb9c3a0e941dfb8857188bb5d2b4f : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/81fa6d0cd8e944b7b3a0459c6bbfdd9b : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a39e49b5f3414ee7a102aa6dcafca798 : SUCCESS in 10m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/990a8221f33a49c7a1d366243ff49e02 : SUCCESS in 24m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/927375e382ca43aa9e0a17b7a4941d1b : SUCCESS in 12m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d7076ab7211f41a2a9cee97ab943ef88 : SUCCESS in 14m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2a06eb3afb8148d7a1771feba283ea6d : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5b9bf5aab814c4894d000953342abbb : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4a5ee342684465ab16012138c2a3447 : SUCCESS in 30m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a4de718622144a48b6062b69abe66b06 : SUCCESS in 5m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/968a6fe3877f414c8baa8ebbdbc3fd7d : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db4cb7d5e9e442b4b27793adf55ca24a : SUCCESS in 7m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bf6e2970b2fb42b2b9762fa231841ad4 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bb70ec72fec4df284fd5fc788756756 : SUCCESS in 57m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c9b25838e93403b866124875eeb85e9 : SUCCESS in 1h 25m 42s","accounts_in_message":[],"_revision_number":2},{"id":"27ff70032a246def8918554cd215e40f45663c94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 18:07:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9d67fc122e8359aba07ee727a56138389d2cc3a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 18:08:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5231708deea74183bceb6579b4d52ea2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2712bc44154a21afa4e1503b013857 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d842f2f41aad13ec9ab44dd5f334c185d9743afb","revisions":{"eaed915d256ec39a59129e6ca7e0772a9c298853":{"kind":"REWORK","_number":1,"created":"2023-09-21 11:35:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/896053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/896053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/896053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/896053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/896053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/896053/1"}}},"commit":{"parents":[{"commit":"b23a8724d7a891b7d1312950968ab2ffb3af54f1","subject":"Merge \"update constraint for python-openstackclient to new release 6.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b23a8724d7a891b7d1312950968ab2ffb3af54f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-21 11:35:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-21 11:35:44.000000000","tz":0},"subject":"update constraint for oslo.db to new release 14.1.0","message":"update constraint for oslo.db to new release 14.1.0\n\nmeta: version: 14.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: I9bfaed7682c762caf0ed18bdf4f1c42179f28141\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Thomas Goirand \u003cthomas@goirand.fr\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia77ae93ce59f4e175aaee1b030ea21e6c394480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eaed915d256ec39a59129e6ca7e0772a9c298853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eaed915d256ec39a59129e6ca7e0772a9c298853"}]},"branch":"refs/heads/master"},"d842f2f41aad13ec9ab44dd5f334c185d9743afb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-26 14:54:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/53/896053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/896053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/896053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/896053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/896053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/896053/2"}}},"commit":{"parents":[{"commit":"c5d7969111ad76ab30b027d864870896575effb0","subject":"Make cross-masakari-py310 non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5d7969111ad76ab30b027d864870896575effb0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-21 11:35:44.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-09-26 14:54:35.000000000","tz":0},"subject":"update constraint for oslo.db to new release 14.1.0","message":"update constraint for oslo.db to new release 14.1.0\n\nmeta: version: 14.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: I9bfaed7682c762caf0ed18bdf4f1c42179f28141\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Thomas Goirand \u003cthomas@goirand.fr\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia77ae93ce59f4e175aaee1b030ea21e6c394480e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d842f2f41aad13ec9ab44dd5f334c185d9743afb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d842f2f41aad13ec9ab44dd5f334c185d9743afb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
