)]}'
{"id":"openstack%2Frequirements~896264","triplet_id":"openstack%2Frequirements~master~I1e9542beb9cb86e517c7e33023cc4e1e8c0d31d4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1e9542beb9cb86e517c7e33023cc4e1e8c0d31d4","subject":"update constraint for python-barbicanclient to new release 5.6.1","status":"MERGED","created":"2023-09-22 15:20:25.000000000","updated":"2023-09-22 22:01:27.000000000","submitted":"2023-09-22 22:00:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896264-new-release","meta_rev_id":"581041bbe9539f49eea75572b72f86177e8b2061","_number":896264,"virtual_id_number":896264,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 22:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-22 20:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-22 20:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 16:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 17:54:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-22 20:09:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0a9a854ef4ebcd22cbf58615f8740cbdcb08614c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-09-22 15:20:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaa3f33fdcdda08ee7f43c53fa52c2776024b008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:46:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/21b237855f1146639a53cd9bb7c85a9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc68136b9fcd40118ffa58fbaf59dab0 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538cd2ddaa484b42aa43fdcbc171d297 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b78318c8c2ea4d95ad6529a63441ef4b : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f04afe92230c498097c3b71d67713a47 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8de2870bb6440ba89665dd8d3d9a3d6 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86cdbd6e25a84456be8c005d2f46339c : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44ff83515e1e4a8784bd7b0f1663144b : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0ef251681a545e4ba4a0185d41852c5 : SUCCESS in 4m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00ea70f9247d46f6a2e3bd57a9c3bbe7 : SUCCESS in 8m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7643b52ef8f24cc5a5bd862decb82b97 : SUCCESS in 10m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/66a3dac108ef421593de09145e3cdecf : SUCCESS in 7m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a10edb2127704c1bb60c3e366f226836 : SUCCESS in 10m 54s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8d20eee8603a41e1bfc9157fe47dbeda : SUCCESS in 9m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/433eef15c2ac4d31b2118f63d11ccbb5 : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5ee11acae3e24489a918e3860839805f : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a1deedf6e8254a91b8555e03426375be : SUCCESS in 12m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/391f693e68a84a2f8e3a39b56b468a35 : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/72b5822e88d14ae29a12a6a73e78d36e : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c86666e83ee642ecb730675eb6ae1e8c : SUCCESS in 8m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d0c0c68a9ced41d5807977edf51bc9e2 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f15c52658ef478e89f45bbd9293418b : SUCCESS in 7m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0fc7e5343da34782943628529a229ea0 : SUCCESS in 7m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4e7710e5c87f4f148757d00ceea281a1 : SUCCESS in 11m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0657586b208e459c92393fe1b5add9fe : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1df4d62210a24ad7b4123d4ffdb273e9 : SUCCESS in 10m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fca852178de4494581cc1e0bd61b1928 : SUCCESS in 5m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a392a20cf6a14ba99395d16b0d61c67f : SUCCESS in 10m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d82a336a3a8046989d52d39715d0baea : SUCCESS in 24m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6e61bfa27538442489f1a4da5719b705 : SUCCESS in 11m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/14235b30d1cf4c46a47727d095d9e10b : SUCCESS in 15m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d862e57ca6df42bba1ef91954a2d9ddb : SUCCESS in 6m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e99acfb8faf448569950033659cea5f8 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bc7b446d71d4f83984b9d0153ef24e8 : SUCCESS in 24m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c535fa8f925049789089189df5914260 : SUCCESS in 6m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/759bf0c19711471aa97efdbfee49125c : FAILURE in 3m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d3077fda97d401ba5094dfc3f826dd8 : SUCCESS in 7m 13s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/cd95f9cb15694f8fbd950d51d3842a31 : SUCCESS in 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fd457742a764bc49e785d6067fadde9 : SUCCESS in 47m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c52f45bddad8487ba04133b1e6a694d6 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":1},{"id":"92c5abbb2b341ffe5fa42eefc4d409bd078bebb6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-22 17:54:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a91f4a57a97fb8f043c86b799cece771a5436174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:37:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5e0cfac7c94635ae88d0e07270f2c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66223770a86b4d20b78188d8f36d9447 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52add317c4674c40a846720c20b70aa7 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ca32fa024db4acda1083776943b47a5 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cd6828b596a49fe94ddfe50f102a03c : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30e938c9f491468cb9041100da3639dc : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a90a9a8628b84327a8c7b930e8ffce04 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7de2b94e458d46ea89baf1151882c283 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99599e2cdd2a435aab96c3d23541fdad : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85e8863ec1aa4fc0a88ad3f9c7b5d9a2 : SUCCESS in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/05af3c92736842e3bdee2fd2583d262b : SUCCESS in 12m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1ff87b46ff4f42af94c8d8123341354a : SUCCESS in 8m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e599cf4392e74e8e916826a6b530dfdc : SUCCESS in 10m 47s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/608014723807473a85f8e85931fcdb4d : SUCCESS in 11m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a3128de5ce8b4f06a3c11d349e1b8420 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c82291d98a254464a059c8dc3231f510 : SUCCESS in 5m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/354ade38378442cabb8abed6ca793ded : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d82f874a20b74ea481eb2ff353e183e4 : SUCCESS in 5m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/afd5de35edbc4fd3b092bff27a7f681a : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f4e0bdb3cf1b4571b7f8542ac347574e : SUCCESS in 7m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/dfb36973436946f3aee1c3d3a67520e8 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6bacc8975e3043f7a47598dcab1d4aeb : SUCCESS in 7m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b0c2c1b7b3574f99aece8e6881787805 : SUCCESS in 9m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d642198daf294d55b114c5a53d97a08e : SUCCESS in 10m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/48e1ab5b10c44a99a15fd343ff66c1c0 : SUCCESS in 4m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/948512db49e943aaa2153181fd8c8e61 : SUCCESS in 10m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/aceab5ba666a461db70960ddcf8722b1 : SUCCESS in 4m 08s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7be4ad9c5f77464b978ceef947bd4632 : SUCCESS in 10m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d2ee931bba3e441ea744568fd8961c25 : SUCCESS in 23m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/640dabdae458445ba393d84f70326658 : SUCCESS in 9m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b827cca1b6664686a2a8a25620014009 : SUCCESS in 11m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7dcdefe872df48fda0e266b46eb1fdd4 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d021f4c2419848a7be629dfe046b6643 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac9079fd065d42d9bca1eb4c94eeec01 : SUCCESS in 23m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c2481c6fd9c54e498c26a7cbdab3efae : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/693e7022964244bdb3d2f7090464139b : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7718052ea6ba401880915d87579646e5 : SUCCESS in 6m 10s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7d9484a0a91742c8a81882aa22adb4fe : SUCCESS in 9m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/266b2e39adfb4a9a982fdc1afa4834e8 : SUCCESS in 59m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc94e345a271416ea82c0bde80d233be : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5870c95bf8be8e6d3f3618c4f6a3439d08307f29","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 20:09:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1e0af3485dbb891b2e7e427ee00a29cfd53c9b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 20:10:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b96c6854d0f6411811abbac4ef36f96f0d3aca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 22:00:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd4a0f9ae5de4ebabadb3deaadb2d898\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e777cf02e742e8b335e610ed86cb25 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f63cede74b64236b697dd6b86552b83 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8638d8f01b8849d5b7d61047d06a7ca1 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd687d8cddf74525bb0ae48e6b5ac97a : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f2985cca927456ebeb6f14f9b066b89 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f342d84f73f3454baf839303c1bf8d6b : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2dae48931414ff2a8ec1a4ad19ad208 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b1a1ed76fac4b97bc46fbd821d1362d : SUCCESS in 11m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c95fae17f6f42429145cd7ba7710853 : SUCCESS in 11m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1bdf534ed0d34e8980fe7ab35dee2a58 : SUCCESS in 11m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a1b984dc50484b34b5d6395f59ccffa9 : SUCCESS in 11m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/096a561c757d4b6d999af033d79f1f43 : SUCCESS in 5m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/db9b153073204e538421810461c9dd51 : SUCCESS in 5m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/968fe577d2a74f5c953ee7ce075e8b24 : SUCCESS in 8m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7057c6f62e4f4883b23e7a4b050a4e2a : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2bd16add7c7458fa3805361da5c7ddc : SUCCESS in 8m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/894ab06fd2154779899d03653a9091d1 : SUCCESS in 7m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/855c48f91a124fc0814fb812beaa80e1 : SUCCESS in 10m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/837d68748bef4fcd9df51e53e828f2ce : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a3caa8334d624223bf9036f53067cbf9 : SUCCESS in 8m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d297be9852a041568afb25523795a535 : SUCCESS in 4m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a0e9dbc8d03d4c6fa163426457c61876 : SUCCESS in 22m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8bb316a973ba42a699cf9cbaaa1d674b : SUCCESS in 9m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0e9d947e3c2d47da86ea5f8a0ba729ac : SUCCESS in 12m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/50615e503f04495db6bffe5f751559ed : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6cf7d09c7eb4d2cad1f2e176471b49d : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/306d2c2b826447b4a4a6be952869fcce : SUCCESS in 23m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/248c78d3001947669b022b17a2f28700 : SUCCESS in 6m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8b834ab1f7e449cb92ff16353004060f : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ecab9b6f1f2483cb9d81527a8363c5d : SUCCESS in 7m 11s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/31fe09acbfa0463ba0f511a81ecfbeb8 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d4d3376d3854d3e8ee4dee32f91c461 : SUCCESS in 1h 08m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cbfc050e9144a41927ce88eb0d0a56f : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d9d988cc5f652ea0cb66d5b689d933c59b9e5650","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 22:00:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"581041bbe9539f49eea75572b72f86177e8b2061","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 22:01:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a19d44f76d44a1b9e11874126cb7e9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d8df1c85a641d7b66624640fabae56 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03e613cbb730534319755bd2ead5ac21eafd6bcf","revisions":{"03e613cbb730534319755bd2ead5ac21eafd6bcf":{"kind":"REWORK","_number":1,"created":"2023-09-22 15:20:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/896264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/896264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/896264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/896264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/896264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/896264/1"}}},"commit":{"parents":[{"commit":"f63b576e9f6cafe93aeae00a8a4627a9083cb797","subject":"update constraint for virtualbmc to new release 3.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f63b576e9f6cafe93aeae00a8a4627a9083cb797"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-22 15:20:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-09-22 15:20:14.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 5.6.1","message":"update constraint for python-barbicanclient to new release 5.6.1\n\nmeta: version: 5.6.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia667f29ec67cbdb07696a6375d4285fcb786cc9b\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1e9542beb9cb86e517c7e33023cc4e1e8c0d31d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03e613cbb730534319755bd2ead5ac21eafd6bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03e613cbb730534319755bd2ead5ac21eafd6bcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
