)]}'
{"id":"openstack%2Frequirements~896310","triplet_id":"openstack%2Frequirements~master~I0b82ec1b0b9d2395caae50048f01839fbce0ad4d","project":"openstack/requirements","branch":"master","topic":"python3.11","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-02 19:35:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b82ec1b0b9d2395caae50048f01839fbce0ad4d","subject":"Update requirements-tox-py311-check-uc definition","status":"MERGED","created":"2023-09-23 11:51:12.000000000","updated":"2023-10-02 19:38:13.000000000","submitted":"2023-10-02 19:35:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896310-python3.11","meta_rev_id":"85ed0dd780eb0b5ae3043fcfb71b77b5e6e3a0b2","_number":896310,"virtual_id_number":896310,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-02 19:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-10-02 18:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-10-02 18:19:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-23 13:19:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-02 18:19:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6c0c74a3df9d84584937e0412392348f46549026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-23 11:51:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7f75fbd93d72f2973a3a2719d24772d3a8c889c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 13:19:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bbaf9f803244107b153898a7948f196\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88afbfce585344e2a478dce646ae399c : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e018a4f770dc4f8c9f8616187f330cb2 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6341367c9107482b83ae82f15533380d : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77707ff8fda04876b67fb2b419e1ac1d : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/301159a0a1b84b4bb776656e13fc9fcb : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b81ef3b711e74273ba6e222bcc99ec0c : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10142b8bdec6414492d62e425cac9714 : SUCCESS in 4m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/360f1828ae4843fa852aec9e73d7ea70 : SUCCESS in 13m 44s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1facde10b6c445378ef90855bc9fd773 : SUCCESS in 55m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13a426f1041b4f50896e716bdd389b37 : SUCCESS in 1h 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c5cb519593d6af33e4a9ea46673ea2ecea23be3b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-10-02 18:19:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12ac89c60ecb3119d249c2d7f057538151fcdda3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 18:20:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50684f7103161dd081c3f8f018251483fe2455ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 19:35:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cf7a1adab584fb1b19250d7b8342e02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1732efd6f014df09973ec29aecafb3a : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/954041b5eaab4341ae795c4f498a3622 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae93807f218f4f3c932177cf94e1848f : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c516aa6483c44bc1b40cd4cb65e7e71d : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4de1fa39691740e6b9a9ee20fdc87163 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4191e73e16ca4a76b3fdf2590450431f : SUCCESS in 3m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5adef96ebff14869b79a1646daa86116 : SUCCESS in 49m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a4eb7150db44eae9a6ba2cc040bdcb3 : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"00fa72a53e6c8061ddade2ce02ec1e3b47df7a4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 19:35:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"85ed0dd780eb0b5ae3043fcfb71b77b5e6e3a0b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 19:38:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb26eff8fbf2433a9290518ced634dcb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d38749c599f4bf7a57fca781a2b75c4 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adbd46a2c0cf0d197bbba1f57df2544d5638fd18","revisions":{"adbd46a2c0cf0d197bbba1f57df2544d5638fd18":{"kind":"REWORK","_number":1,"created":"2023-09-23 11:51:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/10/896310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/896310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/896310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/896310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/896310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/896310/1"}}},"commit":{"parents":[{"commit":"5eac47701c8d27c4cdc2906197956270bcdce16e","subject":"Merge \"update constraint for tooz to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5eac47701c8d27c4cdc2906197956270bcdce16e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-23 11:49:57.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-23 11:49:57.000000000","tz":120},"subject":"Update requirements-tox-py311-check-uc definition","message":"Update requirements-tox-py311-check-uc definition\n\nWe can now use openstack-tox-py311 as parent.\n\nChange-Id: I0b82ec1b0b9d2395caae50048f01839fbce0ad4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adbd46a2c0cf0d197bbba1f57df2544d5638fd18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adbd46a2c0cf0d197bbba1f57df2544d5638fd18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
