)]}'
{"id":"openstack%2Frequirements~897319","triplet_id":"openstack%2Frequirements~master~Ib26326c6cc0a0ea56771cc69388377bd6a7f5016","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib26326c6cc0a0ea56771cc69388377bd6a7f5016","subject":"update constraint for tap-as-a-service to new release 12.0.0","status":"MERGED","created":"2023-10-04 11:26:36.000000000","updated":"2023-10-04 17:36:19.000000000","submitted":"2023-10-04 17:35:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897319-new-release","meta_rev_id":"82c539ab9ac3823fa504b3a473a0063e20ded105","_number":897319,"virtual_id_number":897319,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 17:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-10-04 15:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-10-04 15:38:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 12:57:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-04 15:38:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"18f4c48b3e112f62f85a802320c74ea60a1fd984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-10-04 11:26:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98e29548b81ff5f2ad51cb721e7804f938ed74ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 12:57:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/951415e790aa4dde937b5d3d3b05a233\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ed0532958b413383615a84faa57baf : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a380f46c3884ab0916a151e67a79e69 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09524d68df1e40efb5bf8cfee70907e4 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/813a42a8412940b8b02431aa9bc68444 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15416476e310433da16cafa2e375ace2 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/981cbacdd773429e9a3374f2af3a09b1 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1969c2ca61fa4a2a86b6ce530f598004 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/249950ad981b4f29be150efa24dc69d6 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21256656f3324b1d8d5dc18979a206e1 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cb943c143ef8498b8f53ecffb4a1c5a4 : SUCCESS in 14m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/87a1d4c99872450bbc9ea02290902d77 : SUCCESS in 9m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0c19c403d399439ca18df53ce5a798d3 : SUCCESS in 13m 50s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6339764a748047b2b0493a316a9a3f44 : SUCCESS in 7m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/84b99ed6864b4ef193877ac4af6d2427 : SUCCESS in 6m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9193876db4834a41b343e84dd5e900f1 : SUCCESS in 5m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1caef6b936f143a0b72d0fe1e006cd8f : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dbe0eae4265947b99d332881ac2f77d2 : SUCCESS in 5m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3c628d0784804919a7e6252d77cb30d9 : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b1b7d7c6dc284fd3b49f9356bc8180b1 : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f4ff3c6d21d54425bc12d69a2397bf5d : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8854e4807bff4e25b45e7ec81a4fde82 : SUCCESS in 7m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5caf952630954155bea652a224dd504c : SUCCESS in 7m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7018417c3d734190a011b200453b80ff : SUCCESS in 13m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5bdce4d3fea046d392d67885603261d6 : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6ef05e9e5f71466c8527bd8f1844d2b2 : SUCCESS in 10m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f44f31f5bed740e89ee0c2a7f7e7b9ac : FAILURE in 4m 14s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f7b44a1d51114c168f113d2b1a148fad : SUCCESS in 12m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7c87b78dab8247058c8689e32cc34c19 : SUCCESS in 27m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d709188fba77437d86e472e02159a81b : SUCCESS in 11m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7d611a3d90094e5ebd759e14732ab5a6 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b39b8f92fd8f4cb4b0239785793a6929 : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b64ae222b20748b2b5b1262aa06a9f0c : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57753bbc52384d6688f949bf01d993ce : SUCCESS in 24m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ad1e379c19eb40fcb6b7f0015d0dcbe5 : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/540a85e06fbe46f79582ec0ca34c3b40 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ee768bbeaf74c6ab0e0ed81f598f116 : SUCCESS in 6m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f12fca96d73d4572bdd195ef4a9567ff : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbff46535a5f45fab0097e9fc1e7ddec : SUCCESS in 56m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41bb3cd5ad2f4aa5833f096a023088ce : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4bb7e1263c88ef8f1e67889ae793698bcbeefec6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-10-04 15:38:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28bcc5a366d7562f595c2e2a7d99d866e38ed984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 15:39:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"416acb4b0965a8852b89624f08659063b636c5dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:35:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1b3306cd52f4ce685f9301f5ff57258\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3034754e574cb48adba3dc6203dbad : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28598d6bcf7644c6bb3019072023dcd5 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5fbd0d158cd4c5abac7c8e216c81727 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/294a4c63a4064ee1ab4c1ae84a729223 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb47753fb5244df38d591e0637490e57 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b390030ae4b4b08a6ac9dc6198066d1 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25548c1e20c441f8972bf1f50ee95854 : SUCCESS in 4m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2fdf993c227d4e89add9434fd632924c : SUCCESS in 11m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4dbdcba51117409d895f1fc9377a4926 : SUCCESS in 15m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7ba3c8e3152741c180f2b894de42c983 : SUCCESS in 11m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6be2a385b07c49569b2484b8af716a9d : SUCCESS in 13m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fb87724c386c4183ae5f6b6bbdb18a2d : SUCCESS in 5m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/17868ad50f624ba68ae9120a6bc47788 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/efe491511d494c5594be18454318c01e : SUCCESS in 8m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5643411439fa464eb6a10d6564024d3e : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79fc1e68ad4e4045a0fccae7c96113e8 : SUCCESS in 9m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/68e60c5fa6ab49a7bcfbe021015ae110 : SUCCESS in 10m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d16330869d9e4099aa3b96a783b87ee4 : SUCCESS in 12m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4370024bc9a647da9ea152dcb1dbcdb5 : SUCCESS in 5m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b87eb0b44ecd49208a0efecf8dbe2157 : SUCCESS in 10m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e91cd312db9e415eb00e53bcdb711f8c : SUCCESS in 26m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/27b91435c1a243dfbea9327190a71471 : SUCCESS in 15m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1c001a2e04024b95a64191587b1844ec : SUCCESS in 14m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6c86b4bb5ac940ad8fb96e414ac4124f : SUCCESS in 7m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a333fb4a1b5434a99d25d911ada5f4b : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c06a64354804437eb77437fc58e9ac8a : SUCCESS in 26m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/56d33cbc48aa467bac2dd71f4a746aa2 : SUCCESS in 6m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/52a0d25ea2b140b98137e03631dccbec : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62499e2c16b34c36be487c63bfb5f2e7 : SUCCESS in 7m 06s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1beaee57d60e431abe27089e7d128bd6 : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ee7ec322a004989b391010528a7b1a5 : SUCCESS in 1h 08m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e7a21f91ef7476f9475254d0adc7bb0 : SUCCESS in 1h 44m 01s","accounts_in_message":[],"_revision_number":1},{"id":"14b760a21f4a0171bbd13d2228672ac97a302fe6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:35:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"82c539ab9ac3823fa504b3a473a0063e20ded105","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:36:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a7c2b245e504f8f8a2a9842c74dd765\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c9c3cd3dab4a62a4afd1ee76cbbae0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38ba3b526a3808a454d2a8cd661c6dbaf145756f","revisions":{"38ba3b526a3808a454d2a8cd661c6dbaf145756f":{"kind":"REWORK","_number":1,"created":"2023-10-04 11:26:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/897319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/897319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/897319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/897319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/897319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/897319/1"}}},"commit":{"parents":[{"commit":"888becdc2f4637a9ab7f02449c011078730c6428","subject":"Merge \"Update requirements-tox-py311-check-uc definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/888becdc2f4637a9ab7f02449c011078730c6428"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-04 11:26:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-04 11:26:35.000000000","tz":0},"subject":"update constraint for tap-as-a-service to new release 12.0.0","message":"update constraint for tap-as-a-service to new release 12.0.0\n\nmeta: version: 12.0.0\nmeta: diff-start: 11.0.0\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Id15a8d02c72129ec32da86d96139bee5d0651278\nmeta: release:Code-Review+1: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Code-Review+1: Jake Yip \u003cjake.yip@unimelb.edu.au\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Pranali Deore \u003cpdeore@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+1: Hao Wang \u003csxmatch1986@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib26326c6cc0a0ea56771cc69388377bd6a7f5016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38ba3b526a3808a454d2a8cd661c6dbaf145756f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38ba3b526a3808a454d2a8cd661c6dbaf145756f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
