)]}'
{"id":"openstack%2Frequirements~897537","triplet_id":"openstack%2Frequirements~master~I2f3e7f892d755f212f1945d1086f4c91306f7c09","project":"openstack/requirements","branch":"master","topic":"revert-pillow-bump","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-10-06 13:50:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f3e7f892d755f212f1945d1086f4c91306f7c09","subject":"Revert Pillow version pin to 9.5.0","status":"MERGED","created":"2023-10-06 10:37:24.000000000","updated":"2023-10-06 13:52:01.000000000","submitted":"2023-10-06 13:50:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897537-revert-pillow-bump","meta_rev_id":"100f8d84ce79495303b3a53962e4c7ae28b7da56","_number":897537,"virtual_id_number":897537,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 13:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-10-06 12:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-10-06 12:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 12:02:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-10-06 12:29:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"333e029de9e22ec5ed7d77fbad838393d7d81c7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-06 10:37:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f821158b0741835e29ca0d39d9cef1841a49cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-06 10:37:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"59a8437d84abc48d9734bf1345b5a4f4f5825a6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-06 10:39:08.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b57b35559eaef8f8afe3261f2eb12fd2e10b29f8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-06 12:02:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"051a7f47df627526dbd467ed002def34d6a46d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 12:29:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0427a813500449bc8cf36d0508df5156\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/666c1b8ce83a4e6bbdb28154d022f5e9 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88747da3e0564e099a3368fa795b5501 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a91a744b6e6409da3a120dc25cff5c9 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0c9e2bd33e4f28a8780ad39c5a497b : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11ca319ca2424abd958dffa49af1d99e : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eee86346725e44a8897c05756845b134 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dda4143fbc5f4ef8a7af92bd260ef79e : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/265a1a04260e48f48627694be6b02584 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b59b0da8e27644a28c71f7f772abcdcb : SUCCESS in 11m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/27afa1b956e14911ae924f0aad0d97d5 : SUCCESS in 16m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d2b905bdda9b4f5e932507b4526d338a : SUCCESS in 10m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/40a73b3cf52d469ba91b543af370e14e : SUCCESS in 18m 21s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4ee8fdd60ee9444b8de53c242f589cd6 : SUCCESS in 7m 16s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f7fdc72df9c44bb3807bdbb9c0f0f9c3 : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/53cebdc60ef14575935dd5ac83a03352 : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5204f6f465294ccba6e8bca3352e694c : SUCCESS in 11m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9b0548cbbe6745e8a8112ea222f2179b : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/24ad0e20aba24c9ea793bb5b44ed49bb : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8b1d918ce6b148c4ae3df5471259af10 : SUCCESS in 10m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8386431c491f458e86d3e55a2184997e : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb2e1996ac784fc5abe5a51aca9dc601 : SUCCESS in 7m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3e13a3ca53ff45bcbb8e6dcffa882843 : SUCCESS in 7m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8656cdc4b78b4b9da3714e3749b30267 : SUCCESS in 16m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b7d61148577c41b999427dddb0d7c910 : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/072c81f97cf04e79a0a1875d44fb4360 : SUCCESS in 11m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cfcf350047824af7a190021c4b1a08dd : FAILURE in 4m 30s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/dcb48c9070e44ec38f6cfe3311b0b436 : SUCCESS in 11m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c2c1ebadcc00489fb7db2fbc1bd76125 : SUCCESS in 28m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d718655116974f62a341aadca7ede248 : SUCCESS in 10m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cf325dc7c8674e6f8494f06c438deb52 : SUCCESS in 11m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6f7062b40f0f465fae8b0921d87ea93a : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e7c42473e8c14ed0a50c8f4b5fde6555 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60d2b61422374fd88347fa9009bc4a5d : SUCCESS in 19m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4b5edeed5f8d4ffbb18dbc106200fc80 : SUCCESS in 4m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fce79f38f3ab4c4aa279290d2c011543 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4c6ee31885d4baf91ea803abbdb0645 : SUCCESS in 6m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4d21bf70ced2441ea7fec40ea872bb00 : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/878560b2e14440b884d6758023683d56 : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c806b95d01a4401fbc9025a9fbb74a3e : SUCCESS in 1h 44m 21s","accounts_in_message":[],"_revision_number":3},{"id":"24c5993704de8085f8dfb99bbc304e2a11c844ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 12:30:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0064c339e7639cd58012ff30eb1356f8f6b822ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 13:50:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1bc26316ed148b99c017c117c46ccfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3836e8a97d241659de55f94b6c0a044 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/145890476a1f4c228836da0cdd4dccdb : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bc5fc55f38f4c949cdfe3e776413e90 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5c0cb9377b441008e01b30c1887bec2 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73b07e321fcd442c9352f5476bb15994 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c83f9b181e5744d69a1127a6c63d7156 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b33c1961dd441e38e6c30d1254aee06 : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b73fb03a79c4f67b7acdb739fdca9f0 : SUCCESS in 10m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17be99de1010445f9c9cba06c89141a9 : SUCCESS in 14m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/86fa2db998864965aac0464d8ea4c542 : SUCCESS in 10m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cfa8cd64334345c79d27988c3a647930 : SUCCESS in 11m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/68a7c1b6cec2482b830c673c0c50f5f4 : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d83b256f94b443dfba0383a018240404 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/aa65b7a4e20d4e8facf19dca45187e3a : SUCCESS in 8m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1cd4c1a4268149668434ef1f392aa8de : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d94a006f52544d39a60b990434fd4e3 : SUCCESS in 8m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/eeaa97e17ad84984ac48c48f1e9ae307 : SUCCESS in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f8e1003a5efd4418a7e04c895213c0a1 : SUCCESS in 11m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7db71150178246bd96a8418a72e26dd9 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ee17f09658594d3d899eb609c77dc8be : SUCCESS in 9m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d4881de38a804c629cd5471561a72528 : SUCCESS in 26m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2ec82720c3ed48308363611ad2152ac2 : SUCCESS in 11m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5d13b21bdaec46dcaa3bc18813f5e182 : SUCCESS in 14m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/df6f5dc5732b4140a71db5fbbe53a55d : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1df68fa5dc904259869122cf04d3261e : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5065c17373214feeb532736b58e6b7bd : SUCCESS in 18m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/92472af7a85640999c3563e977764060 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/33d82ad344b44f89baf4380313eefb35 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/426217c7b24746b498e76fc81b7f1206 : SUCCESS in 6m 18s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b9101d886d9b441b912f4180dc75a7d7 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5b32af84f014425b19224776272401f : SUCCESS in 39m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3d822bc41b44d1cb86c178c5d904857 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":3},{"id":"a520b89ac46e51baaf05bd802903406f8c6be803","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 13:50:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"100f8d84ce79495303b3a53962e4c7ae28b7da56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 13:52:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a10a66cb6544abd83ea7ef3b69bc0ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930cd19eadf043ef8fe836ffe90d5c37 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79faab4a72a5876090e9dff295aec99e3c43176c","revisions":{"63e38fc93cc3f5803fbd015fda9bd55fcf16fdeb":{"kind":"REWORK","_number":1,"created":"2023-10-06 10:37:24.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/37/897537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/897537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/897537/1"}}},"commit":{"parents":[{"commit":"ab0dcbdda28753cca86e162b9ae2fcb1aa99d001","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab0dcbdda28753cca86e162b9ae2fcb1aa99d001"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-10-06 10:33:37.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-10-06 10:33:37.000000000","tz":120},"subject":"Rvert Pillow version pin to 9.5.0","message":"Rvert Pillow version pin to 9.5.0\n\nThe Pillow bump broke the blockdiag library, it is impossible to\nrender diagrams, and that breaks the execution of the docs job\nwhen using tools like seqdiag or blockdiag.\nThis patcj set version of Pillow back to 9.5.0 until a valid\nalternative can be found.\n\nFor more details see [1]\n\n[1] https://bugs.launchpad.net/ironic/+bug/2026345\n\nChange-Id: I2f3e7f892d755f212f1945d1086f4c91306f7c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63e38fc93cc3f5803fbd015fda9bd55fcf16fdeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63e38fc93cc3f5803fbd015fda9bd55fcf16fdeb"}]},"branch":"refs/heads/master"},"d2b8a90309834f417e3759b252b96cd304c84a66":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-06 10:37:58.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/37/897537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/897537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/897537/2"}}},"commit":{"parents":[{"commit":"ab0dcbdda28753cca86e162b9ae2fcb1aa99d001","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab0dcbdda28753cca86e162b9ae2fcb1aa99d001"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-10-06 10:33:37.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-10-06 10:37:48.000000000","tz":120},"subject":"Revert Pillow version pin to 9.5.0","message":"Revert Pillow version pin to 9.5.0\n\nThe Pillow bump broke the blockdiag library, it is impossible to\nrender diagrams, and that breaks the execution of the docs job\nwhen using tools like seqdiag or blockdiag.\nThis patcj set version of Pillow back to 9.5.0 until a valid\nalternative can be found.\n\nFor more details see [1]\n\n[1] https://bugs.launchpad.net/ironic/+bug/2026345\n\nChange-Id: I2f3e7f892d755f212f1945d1086f4c91306f7c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2b8a90309834f417e3759b252b96cd304c84a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2b8a90309834f417e3759b252b96cd304c84a66"}]},"branch":"refs/heads/master"},"79faab4a72a5876090e9dff295aec99e3c43176c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-06 10:39:08.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/37/897537/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/897537/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/897537/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/897537/3"}}},"commit":{"parents":[{"commit":"ab0dcbdda28753cca86e162b9ae2fcb1aa99d001","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab0dcbdda28753cca86e162b9ae2fcb1aa99d001"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-10-06 10:33:37.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-10-06 10:38:52.000000000","tz":120},"subject":"Revert Pillow version pin to 9.5.0","message":"Revert Pillow version pin to 9.5.0\n\nThe Pillow bump broke the blockdiag library, it is impossible to\nrender diagrams, and that breaks the execution of the docs job\nwhen using tools like seqdiag or blockdiag.\nThis patch set version of Pillow back to 9.5.0 until a valid\nalternative can be found.\n\nFor more details see [1]\n\n[1] https://bugs.launchpad.net/ironic/+bug/2026345\n\nChange-Id: I2f3e7f892d755f212f1945d1086f4c91306f7c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79faab4a72a5876090e9dff295aec99e3c43176c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79faab4a72a5876090e9dff295aec99e3c43176c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
