)]}'
{"id":"openstack%2Frequirements~900207","triplet_id":"openstack%2Frequirements~master~I859abd5902dd4311a4c449501a3869445614ffb2","project":"openstack/requirements","branch":"master","topic":"typing_extensions","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-11-07 15:09:11.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-07 16:41:49.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-07 16:41:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I859abd5902dd4311a4c449501a3869445614ffb2","subject":"Add typing_extensions","status":"MERGED","created":"2023-11-06 15:38:41.000000000","updated":"2023-11-07 16:43:00.000000000","submitted":"2023-11-07 16:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900207-typing_extensions","meta_rev_id":"a41c4e9b279826cad78f75d8ee25435699be0c6a","_number":900207,"virtual_id_number":900207,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 16:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-11-07 01:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-11-07 15:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-11-07 15:09:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 16:28:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-06 18:32:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 18:47:33.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2023-11-07 01:53:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"032102fdc6e1b2b28f900039f85bd201fe73430b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 15:38:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64d40737fdccf57517369422721b939afbbcd5de","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-06 16:28:14.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ba5806ba1aeef4beae5b332e9380b8cd273897e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 16:54:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"77fe8f594bbb25e9f3b7fbd3a25a20bda34943b7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 16:54:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b257c86cecd075df9518867556a93f5c89915d34","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 16:55:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"96c861d882a6a6f4b2e185a6351c6ed3b3538527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 18:32:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/409d98e5d5b545d7bc5df45ac070db32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c1166222874f79b76f2375ba916de5 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a98883c2e154a30a21c928660d4efd3 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe5d9422834943aab38a3fb61c277a17 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a96c4efa1844b00a5c9d940c0064710 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee2c8cd8440d4a699c78e0f0b98b1897 : POST_FAILURE in 8m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd96c98d3b104627b71d1a67039d969c : SUCCESS in 1h 02m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0de0dc7538b84c968185e075c917447f : SUCCESS in 1h 31m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ec567e9845e16aeeb433662b90141e0bf4d2db9b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-06 18:47:33.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e2c62b2671a28e4a55d856a68c9cb29e24d25a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 20:30:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/200fe9ba669246d1bb12c7495d22866c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6c4c4dda1d4a6abbe5f5361c05314f : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0463f93a4f324ca48b9894e755fa90c2 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4fd6d6f67a34e4db061ea90b2702cc5 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b0d609c862a4da3a4f5bc86138bdcaa : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27351045a0694aa7b2193d91b4259302 : FAILURE in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce4507e0aac548dfbeaa2ae977517438 : POST_FAILURE in 1h 03m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81842a025aa441b2842023da4383fa11 : SUCCESS in 1h 34m 59s","accounts_in_message":[],"_revision_number":2},{"id":"64b5d83f310244779de45f4b6aa848b2b5a5ab91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-06 22:03:59.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"587be575185fa2e7f38e8131ffdb0a9bb7a2f614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 23:17:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668e6b03f9c64fc28329a8a21bb9ce29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3ed7ac8db6434abbd781fb3a7664b4 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9025098eb4d64fd9be2eb1a82ee4c7f4 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c832940184d944d4bbcb7ed25f826a88 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12d1e33287e84e0db21de86a9f1094c5 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e9947622f8a4ccd8d7a28947fc06d8d : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f006b6d406e4ad7bbc04a8c18752153 : SUCCESS in 1h 05m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0efd5e6b670142989d63eb63106280aa : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a8a850268bf154e9585e85ed038aa87b4047764c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-07 01:53:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e426974b5add7b8b3c2720160522e91431da6282","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-07 15:09:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c03c08b5457ffd599584e898cbd5a33e4c47154d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 15:09:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0ed12faac52b58ed604577e9c0452a73c7ab9b4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 16:41:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d3dc6f6071b474d8b272220849dc0f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057789f47775437dad8c6f6ca50de226 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c59f957a52e4407aeefa54dfac3408e : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed436daffbc24a5ebeeb9eca40bc91a5 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74a6f38a5cad486b9cad619d193c5550 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de5c650fb81a4c39b9e3f169644cf5af : SUCCESS in 4m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faa21b0ee90d4393aee93f462d613cc1 : SUCCESS in 53m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0682c6f6d25427480b5dbb82ab15c8c : SUCCESS in 1h 31m 59s","accounts_in_message":[],"_revision_number":3},{"id":"e49fde3eda858e09bbabc71f0fcbe0af4ff8ac51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 16:41:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a41c4e9b279826cad78f75d8ee25435699be0c6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 16:43:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d9e500d97104d4483dbbc3b302c5515\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681eb214d8cf43f0b507caa686a8b97d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88840c648261b5ce8e82a9de30794d0696ed71ad","revisions":{"daf7c40734bac91d0c5ef8dce6e18fc50127557f":{"kind":"REWORK","_number":1,"created":"2023-11-06 15:38:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/900207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/900207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/900207/1"}}},"commit":{"parents":[{"commit":"d2f135f2a9a9e57ef5d95989e12e9fd986bba441","subject":"Add python-observabilityclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2f135f2a9a9e57ef5d95989e12e9fd986bba441"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-06 15:35:17.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-06 15:38:09.000000000","tz":0},"subject":"Add python-observabilityclient","message":"Add python-observabilityclient\n\nThis patch adds typing_extensions for use in openstacksdk and other\nprojects.\n\n- Is the library actively maintained?\nYes.\n\n- Is the library good code?\nYes.\n\n- Is the library python 3 compatible?\nYes.\n\n- Is the library license compatible?\nYes (PSF).\n\n- Is the library already packaged in the distros we target\n  (Ubuntu latest / Fedora latest)?\nYes (as python3-typing-extensions).\n\n- Is the function of this library already covered by other libraries\n  in global-requirements.txt?\nNo.\n\n- Is the library required for OpenStack project or related dev\n  or infrastructure setup? (Answer to this should be Yes, of course) Which?\nYes, openstacksdk.\n\n- If the library release is managed by the Openstack release process\n  does it use the cycle-with-intermediary release type?\nN/A.\n\nChange-Id: I859abd5902dd4311a4c449501a3869445614ffb2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daf7c40734bac91d0c5ef8dce6e18fc50127557f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daf7c40734bac91d0c5ef8dce6e18fc50127557f"}]},"branch":"refs/heads/master"},"845fe9448b41a72079734ace3691b41b3f783cf0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-06 16:54:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/900207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/900207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/900207/2"}}},"commit":{"parents":[{"commit":"d2f135f2a9a9e57ef5d95989e12e9fd986bba441","subject":"Add python-observabilityclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2f135f2a9a9e57ef5d95989e12e9fd986bba441"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-06 15:35:17.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-06 16:54:30.000000000","tz":0},"subject":"Add typing_extensions","message":"Add typing_extensions\n\nThis patch adds typing_extensions for use in openstacksdk and other\nprojects.\n\n- Is the library actively maintained?\nYes.\n\n- Is the library good code?\nYes.\n\n- Is the library python 3 compatible?\nYes.\n\n- Is the library license compatible?\nYes (PSF).\n\n- Is the library already packaged in the distros we target\n  (Ubuntu / Debian latest LTS)?\nYes (as python3-typing-extensions).\n\n- Is the function of this library already covered by other libraries\n  in global-requirements.txt?\nNo.\n\n- Is the library required for OpenStack project or related dev\n  or infrastructure setup? (Answer to this should be Yes, of course) Which?\nYes, openstacksdk.\n\n- If the library release is managed by the Openstack release process\n  does it use the cycle-with-intermediary release type?\nN/A.\n\nChange-Id: I859abd5902dd4311a4c449501a3869445614ffb2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/845fe9448b41a72079734ace3691b41b3f783cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/845fe9448b41a72079734ace3691b41b3f783cf0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"88840c648261b5ce8e82a9de30794d0696ed71ad":{"kind":"REWORK","_number":3,"created":"2023-11-06 22:03:59.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/07/900207/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/900207/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/900207/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/900207/3"}}},"commit":{"parents":[{"commit":"d2f135f2a9a9e57ef5d95989e12e9fd986bba441","subject":"Add python-observabilityclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2f135f2a9a9e57ef5d95989e12e9fd986bba441"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-06 15:35:17.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-06 22:03:55.000000000","tz":0},"subject":"Add typing_extensions","message":"Add typing_extensions\n\nThis patch adds typing_extensions for use in openstacksdk and other\nprojects.\n\n- Is the library actively maintained?\nYes.\n\n- Is the library good code?\nYes.\n\n- Is the library python 3 compatible?\nYes.\n\n- Is the library license compatible?\nYes (PSF).\n\n- Is the library already packaged in the distros we target\n  (Ubuntu / Debian latest LTS)?\nYes (as python3-typing-extensions).\n\n- Is the function of this library already covered by other libraries\n  in global-requirements.txt?\nNo.\n\n- Is the library required for OpenStack project or related dev\n  or infrastructure setup? (Answer to this should be Yes, of course) Which?\nYes, openstacksdk.\n\n- If the library release is managed by the Openstack release process\n  does it use the cycle-with-intermediary release type?\nN/A.\n\nChange-Id: I859abd5902dd4311a4c449501a3869445614ffb2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88840c648261b5ce8e82a9de30794d0696ed71ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88840c648261b5ce8e82a9de30794d0696ed71ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
