)]}'
{"id":"openstack%2Frequirements~900435","triplet_id":"openstack%2Frequirements~master~I72338abeb617ff554d862fa5e4fd230545145ca0","project":"openstack/requirements","branch":"master","topic":"uc-testing-patch9","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-01-20 13:16:24.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-20 13:16:24.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-20 11:35:15.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-20 13:16:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72338abeb617ff554d862fa5e4fd230545145ca0","subject":"Manually generated constraints","status":"MERGED","created":"2023-11-08 16:20:51.000000000","updated":"2024-01-21 17:54:38.000000000","submitted":"2024-01-20 13:16:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"900435-uc-testing-patch9","meta_rev_id":"16bbf0934215e771dfdd193ffe1eba2703abdd8c","_number":900435,"virtual_id_number":900435,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-20 13:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-01-20 11:35:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-01-20 11:35:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 16:46:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2023-11-08 17:18:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-08 17:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 06:26:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-11-17 19:23:31.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-19 09:32:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"0578420e1fc89ec8d15ba956ca42ea616a959822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 16:20:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27979176260b2135a9f9ea2e38fc11ea8bc5f920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 16:41:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f51138f80c7498461ecf4a44ea882e697a26a5c2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-08 16:46:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92de63524523ee2282591e6e72c54a5e49362822","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:02:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9df7041d38e5fd9ae34560cdc8278cc52f2bea46","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:02:29.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"282f4b4b4b1a7c03159d3ebe3f384724d97c6873","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-08 17:18:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4090de13730df93212657573adbd53e57f795d81","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:29:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8f25ff13c21a730ae55ba61a2360f27ece8d0806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 17:44:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9adc5eacab794d258ac7c3962d80f506\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3e629726074979a0ec0e0f9c004133 : FAILURE in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1dd1b63dac44cd99fafd3596600aab : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03f13271e50345ac9921a10d7ff42b47 : FAILURE in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8a04a71d09c4634acd3de9fd6fde225 : FAILURE in 2m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4aff1a50692d4f45864908e8d8bf6314 : FAILURE in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf476283cf0a4f49a67942d4d01db627 : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1fb48d336524443f90c5cd40dd118544 : FAILURE in 2m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/590eceee879f484599c2186300dba5b1 : FAILURE in 2m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/000ca5e8db694fb0816c2b85f203025d : FAILURE in 3m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/227da9668f8247219b4856ad7e43f1ea : FAILURE in 2m 36s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/73cf0ea18b8040cc85f2c31da910eea6 : FAILURE in 3m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d1376ddcf2624ebc8faefaec33d66ab9 : FAILURE in 2m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fa0e9e74727a464a9a20b596f361f7cd : FAILURE in 2m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ce2ba713bdc341bfabb386fc30de4c89 : FAILURE in 3m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/524ad7def4fb49ce84abd1eda4bffd11 : FAILURE in 2m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/93182894ea2249a08c37a494a5a14226 : FAILURE in 2m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ebd1154bc8ff45f2b1c6ed2c192aaf51 : FAILURE in 3m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/892f770740db44ef8632ae9b70f3c2c4 : FAILURE in 2m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d58095cac4d24046aad2215604945197 : FAILURE in 5m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/87e52241cf534c8f856f488a51ffeebd : FAILURE in 3m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ac959cedfb484554b56c5e8ae0f84087 : FAILURE in 3m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bd8369a811ec4893b77e72b8d6471356 : FAILURE in 2m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a8bc756b12c54f36bfd433b9cb9c0531 : FAILURE in 3m 55s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/077550116be545c39e5df5ee622b8c57 : FAILURE in 4m 02s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2937b04d642849239e6bbafe6319d57b : FAILURE in 3m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a4d766b9eb1e4a76ace05333f85508cb : FAILURE in 2m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/65348c6b71a84490863ac123a34acc6e : FAILURE in 3m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fca67211a97349c286914ce7cb7044fd : FAILURE in 4m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8b25e1f26a854ba387d542911b70e5ba : FAILURE in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0595451145c45f0b18823d64178b7f8 : FAILURE in 2m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0f0b468b76644599eec63e081f76c52 : FAILURE in 4m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ae6144e3db3c440e9040771ce3fee130 : FAILURE in 4m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f5af90a965f2499ca26beb7ca45c11cd : FAILURE in 2m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46b95becd3f94a83abf66aae0926c487 : FAILURE in 2m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/24d7b170a1ec4096a6e4fa6a03966289 : FAILURE in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2bd67e4c04446bc95f58fbf1a5fe922 : POST_FAILURE in 8m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6da1d644201746e19e27ead48f48bc62 : FAILURE in 6m 51s","accounts_in_message":[],"_revision_number":3},{"id":"80bba016d1a15b56adafa499551859c0b8fcb8c2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:58:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b854f1aebdf86abf107ec6fbe38f50ebe5abd907","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 21:06:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"043a542678b263099a78ef26df998266385a2b53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 22:37:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f24c87dc14e84b6fb3f523457fecd592\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0dc8a8916147019efb2e6095608f12 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9350ff16d3be454eb170458ced7562a9 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526a2c44e0c54c0cb5c5945fb28ec333 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2cfede24a8034e10b83f7aee1693d390 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cd63266782e4975abc752dad40921c5 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0deefe6c9178405983b3633544437c97 : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/72f66776a8b5446aaf0abe5ce1cddb36 : SUCCESS in 9m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bf29be14aa2b4214a91b08811a452b7d : SUCCESS in 17m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1008ad7f5186435392eb01d0fd299729 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ff88004d355428bbbb54e6f0756caa6 : SUCCESS in 15m 43s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2fbb1ef613b9424783c21bfcafe26d40 : FAILURE in 4m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e506ab97a6c5445080397b6162518ae7 : FAILURE in 5m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0faab6d01c8f4031915cdeb713179105 : SUCCESS in 4m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e2a87995efaf471fa880142153cd19cb : FAILURE in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cf56ec7351394a67b32ebeea658ecaf6 : SUCCESS in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ad6dc3c5fb3747d18ee82afa1415350b : SUCCESS in 5m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/64ffbdb006e24ea187404da7a1d949c4 : SUCCESS in 7m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/be4edd1c2b48470c9e56859b979a17e5 : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b18e20e9407d47a8911091d90f0a10a0 : SUCCESS in 7m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b57b2620c45f4c2e9a8510454a6596e7 : SUCCESS in 7m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/26d86acf16d14d0badc9b08cc30214a6 : SUCCESS in 10m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ba2203a686394308be776ef9e63af4b2 : SUCCESS in 4m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/253edb8a01a746daa6fad9c31060cc77 : FAILURE in 9m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f7e983ebd711488993b98edda4c019fd : FAILURE in 5m 17s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5e572e0f82784279b489ea827b1b90b4 : SUCCESS in 12m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4b6a75e74d4c481185db42ad4d358ec2 : SUCCESS in 28m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3348b0f729ee48c985998d18256f3a11 : FAILURE in 4m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8d5574f9a73642e7919aca5e06d05ba0 : SUCCESS in 13m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/30e491b45180492cb6af3fbb3b9e1ba4 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3482d306631944ce975cbce641236db3 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5423fccf43e4b2084e170d472261a11 : SUCCESS in 23m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d45e42abf7a04c23a70b774c567d6cc9 : SUCCESS in 6m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/262b98938e9f44569cd3bc46bb6a5835 : FAILURE in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fec0e02f31924cbfa3246a307427910a : SUCCESS in 6m 27s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4c2774b8e0254ff281adb9c9117f98fb : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7aadbe340e34273b88292eb078f5903 : FAILURE in 1h 03m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52cc294d04af44fba28661e270e7b4f4 : SUCCESS in 1h 27m 43s\n\nWarning:\n  Job cross-nova-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/virt/libvirt/utils.py in the commit\u0027\n  Comments left for invalid file nova/virt/libvirt/utils.py","accounts_in_message":[],"_revision_number":4},{"id":"9e9e1d4f5ff699cc6ea1b630ae4b3fe9a8ffaaf4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 15:41:00.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"453574d76aec77e7959a98cf7b0456ff347a29f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:39:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a882d94db2614499923007a27e617580\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4da02e16984ee2b65bd606f3aaa674 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed254d8438f4c5f975a606d0bad9085 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/908cfbc9095b49b897e17f08ef97e578 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a864c90a89fc43bfa4c95eda7c1bfeb0 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a5192b14f0c426cbd667317be2149dd : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/042fb6823d7c401c8699934244d3fc59 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/93d5a0373b8b49548bcdb33188746479 : FAILURE in 4m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/da51247bff544bb18fc4b516e55aca39 : SUCCESS in 16m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/06cdc32550004659b1cab336888b94fb : SUCCESS in 11m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2dd586a7f6ab470ea18b167d748fd154 : SUCCESS in 23m 00s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0f6e0eac3e90491cbe6d515bc4d439ee : SUCCESS in 8m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/90b75c52caed465dbd768a087d15e666 : FAILURE in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c448a99b751b463693c0b0bb2b142e46 : SUCCESS in 5m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bd50306b9a8b458da75749d81df5180b : SUCCESS in 12m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ee05e826c65c41d19eae45be7ee6c39b : SUCCESS in 5m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/08d11e175ed649139411f06198e3e66d : SUCCESS in 6m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/472343be32eb45d9a9242727913bf1fe : SUCCESS in 10m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3cac9b7dcf53498b8d0e1b077b37d040 : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecba8095b5c043f6ba4c93884aea0d88 : SUCCESS in 7m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a2fbbcdfac1743579f9c2aa7c230f06f : FAILURE in 9m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/02dc035e81034a1280c1feaacf672000 : SUCCESS in 15m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/29eab1acb2de4522b353a57148b3acd8 : SUCCESS in 4m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e4a8bf3478334cdeb9db2cb296fb0d63 : FAILURE in 8m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7691bec6c5ac4cab88fe403d4ebc53b0 : FAILURE in 4m 29s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9c3eb39078be46a28105ce29e2ef5f74 : SUCCESS in 10m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e939cc4cd4d845cbb72e8238a50367ef : SUCCESS in 23m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e02fbf475d964f788e3ba52eb0095cca : FAILURE in 5m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/73e9e2961ec945e1a647cfddfd4a0bdc : FAILURE in 14m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4fb64c196b104443b20606e43e9be302 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8c8da8446914253bdac40c01be6e48a : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/210d8106b2b1423a828b5f553695e440 : SUCCESS in 21m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1ecc775f9fe24f3ca383d8f165c5988a : SUCCESS in 6m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/13ce89e3b3984026887def02b3094376 : FAILURE in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6a0de0d2e9f4ca4b52e8bc80bac1d97 : SUCCESS in 6m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1da7c545fd744076ab5d10eb3a16ae79 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ebf3851d9cf4f8eacbd68ebb9cb4bc5 : SUCCESS in 48m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7a37bf14ae94416bea1e5feb9dc5bae : SUCCESS in 1h 50m 58s\n\nWarning:\n  Job cross-nova-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/virt/libvirt/utils.py in the commit\u0027\n  Comments left for invalid file nova/virt/libvirt/utils.py","accounts_in_message":[],"_revision_number":5},{"id":"1039ea80512be9b40c3969418976cf98ce61e27c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-13 19:19:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"dfdc56465685bc23727fed298e760257e689badb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:36:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4f55b6c26574542a219d86182f7e3bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6796a86389b74a15bb0dbe94b5c38cba : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3574a79d30543bc8a15bc763a765d39 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06ed5d58e8ef49c7adca90e1dd3d1b76 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1e105f1224b419a80355830bff11953 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10c13717665344de98c481557247a2ec : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1785c28b97d4144b1a54055e70a04c8 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/16fda7a622e74d76b717e3edf280d901 : SUCCESS in 8m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d9f832b6426843daa8870c9e7f7aa852 : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/76e2b49294e34c9b9c0dea50c9e00a67 : SUCCESS in 8m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/70da2b575b4e4b21bde8e0ec296cf57b : SUCCESS in 16m 41s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0b3d40f3e8b149518811d120217438ca : SUCCESS in 8m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/51fdd9e4d2a94bc9954a6a7b45d1e4b8 : SUCCESS in 5m 38s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3d1494f2143f4c829e8d44f057bd5694 : SUCCESS in 3m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/737d226d7e5e4b3c87e827772d01cd46 : SUCCESS in 13m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d88615ba214a4feba313b2c99df1e7cf : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/724b56e92a1f41beaa4c0fe2b8dbd945 : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e497e620c2f24fe3b0550a0e738bb574 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7f6d5b5095b84d8e81558662ca43a4a9 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53a4e2d689134f5e99ced7333004aae5 : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b7b29a9470f64357a07e3ef6fdb4454e : SUCCESS in 8m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3550e8706a244477903fc1eb51781ba6 : SUCCESS in 10m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/af49a4527237429e869902ab198c83c9 : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1f90b5225ee047f1ae6a324ef5a64078 : FAILURE in 8m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4e395b050a8f4d90b813afc13c80b3c1 : FAILURE in 4m 25s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7e4219237d64418ba62628ff6afa76c8 : SUCCESS in 13m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3f0cf372cd14483a848cf6f233313fd1 : SUCCESS in 25m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ba9ec491ff404063885ee349ac12138b : FAILURE in 5m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9f31191d461549848e37a4e355784d4c : SUCCESS in 13m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/93a3989ba2ab4ae4a938e154a6b4c345 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a3114ab3cd64d78ab92eb60cd7bd962 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a6334ad3577421b89eb46a2c73243d0 : SUCCESS in 25m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9e1125798d4e488790798d7220ba25b2 : SUCCESS in 6m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8ffaa978b84a4b3b83095533883646b0 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5aa8033ca7f24bb994bdf6eaaf09a5ff : SUCCESS in 6m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/33cf0a5d738547adac70439e8479eaf4 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1f5bd47ccc647a997f9eee6414a0819 : SUCCESS in 49m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97c55680b62a4e3187b85f89faceedc1 : SUCCESS in 1h 13m 44s\n\nWarning:\n  Job cross-nova-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/virt/libvirt/utils.py in the commit\u0027\n  Comments left for invalid file nova/virt/libvirt/utils.py","accounts_in_message":[],"_revision_number":6},{"id":"a5eaa004730077e670d0a8c635637227ae1473d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-14 16:21:36.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d70325bc4e9bf85f8ab39bf007e28a919fb71c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 18:35:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/215ea902dea7456aa3ead68764d2b2f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd22ac5425ab49d995dc40e9dfe75dc5 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec8ebb1baa74ab39ad46867f87dcd56 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc8d44fe95af4a0b8512f5ecb7383927 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/714f175e44ef42658fcf7d7709ce775b : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdfd009364be44428352662400512aaa : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c8ab1b648fa47439d9f5eb6a2e92d0a : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00a48d66c0c84d3ab12712dadbb59276 : SUCCESS in 9m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/be62b972df614956b9e552a406f969a1 : SUCCESS in 18m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/504059ddb7e8443399463efdd757d72c : SUCCESS in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dfbb202aee114e3792110f31f9ff2db4 : SUCCESS in 16m 55s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6ac90ab90c0c4f9c89eebc45d218152f : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c07bd062e97a4360a8249108345840f4 : SUCCESS in 6m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2b5c1e70e1a9470ca51e13ae244d0aa5 : SUCCESS in 5m 09s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1aa2a430aadb4b8db8da2d626e9a0490 : SUCCESS in 12m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a7550d42d0d843b6be93c42349698858 : SUCCESS in 5m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/325c80bbde234df98eb32117a59bee9d : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/72392190f3db4d4399ed16e018b70cbd : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/00911c0177a043ddb8b5dae68f753343 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bb6894f7f394bbc9386cfc8bba47620 : SUCCESS in 7m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/975edf3c1a264f94ab839f9ef02ee44d : SUCCESS in 11m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1b1bb0fdc423496282f37bc8d43f1059 : SUCCESS in 14m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9a975b45066e4600b74a67079dcefa59 : SUCCESS in 4m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2591f9f04ffd4aa29af170d95665f2de : FAILURE in 8m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0f1f6f8914c5415fa01bac59272055be : FAILURE in 4m 28s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cd7f6fbf63fc440eb0e4710e1fb0b26b : SUCCESS in 12m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d743d38ef16f4c3e8b82f4acd1a42d48 : SUCCESS in 24m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6086e5f9fa4f447997778f0b5d6558da : SUCCESS in 12m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/88ca4e335dab4df8b70f9f461ebd7fa5 : SUCCESS in 13m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b5637f3cdd7c4008afae90cea5497e15 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4abc11d40978492da2c1b0d913de07fb : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40fe491c8d51454a985e61e759a6147c : SUCCESS in 24m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3c414187b040405ebeea34e2afca79d2 : SUCCESS in 7m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4dbab297b96444dca93775782ce9c42a : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e829ce22614541fd903e0a6b62f15bfd : SUCCESS in 7m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c2c40653f06d4dec8446ece9440be12a : SUCCESS in 14m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0397b6abfcd94a3e8b6f2eef615c5d7e : SUCCESS in 1h 05m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ea7bb7568444327a1b194f838ad4c80 : TIMED_OUT in 2h 04m 14s","accounts_in_message":[],"_revision_number":7},{"id":"b51268e50cf9db35aa142861a13143e1952eb1be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-16 01:02:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"5b1a35297d6c286b6cd40d457412b59e3cc8ea22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 02:57:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/374501e4f91f4a45b932d594c5034649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc1732ecd934b5ebde9221503cbf565 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11ee0e019054b37bbca6eb2720d7a61 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6585ef77c8f48838d81be5ed35b51fe : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8d10a0496cc4c489ddf4cfaf0390c82 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/881a774e07e34dc79ee9a916816cb039 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be1e15d41a8d4fbcbe6612979379374f : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c3fcf39beaa145b9b8b23a9a6e86de57 : SUCCESS in 11m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6a88ce7e5904475288e83e7ef17a94c3 : SUCCESS in 16m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/87e0c19367194aaab6e5388e48db96c1 : SUCCESS in 9m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3f329a05496f48b396bb8868ae93167c : SUCCESS in 16m 58s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/33cf3fdf115b4bf694dae20ce091eeed : SUCCESS in 7m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/361290cac8fc4bc5ad1ee4cd263ec4d1 : SUCCESS in 4m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5a4d799752ea4cd4be383b6142461ac2 : SUCCESS in 3m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4a694c37480147778b2c2ba2a51d50bb : FAILURE in 9m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0561a75fbed3416d8b455fc921b5e760 : SUCCESS in 3m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7231a554cf2647c98a6349770f59017b : SUCCESS in 4m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bf3995b60daa46cc8c4d47c5dbe31a4c : SUCCESS in 7m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/53300618711943f7b777584038372e69 : SUCCESS in 8m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ff92d86ce484a77ba15a89421001267 : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fb0902a6c7694147ac8efaf7953cc616 : SUCCESS in 9m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4490b5fa170d4670a1a4fbe5996ed521 : FAILURE in 9m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ab168ae0158245eda0349682d51f8541 : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a96eee90390540b4832551cdb646a277 : SUCCESS in 7m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ebc79ed401024a5b9b466c634bf6a8a9 : FAILURE in 5m 29s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0bc1137bb3bb431994d6c623b2562fdb : SUCCESS in 9m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/43efe23ee94a42a98a334c64e364f552 : SUCCESS in 24m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0df5524ea0b047c986991764c28776f3 : SUCCESS in 9m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/735eb76c01754ab1b16887b829851962 : SUCCESS in 13m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b767db1cf8024690903da9db2839a66e : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99006ad4c44a459abccbf1abfcc17eef : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29cddb5a146e4c90bb217687b8d4ab48 : SUCCESS in 21m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/508a0e47f1ba4dcfb6db31f81bc13556 : SUCCESS in 6m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e48853c0c53d42d0b47ee3aa4858fe72 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d440cda11694f35974285448bbe52dd : SUCCESS in 7m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c27f9b27bdd240758ea1f58f7d2b69a2 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/778a894cc6b54f80bad69773773dd627 : SUCCESS in 1h 07m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7df2d0ba0244104940f51ac477dd8c3 : SUCCESS in 1h 49m 14s","accounts_in_message":[],"_revision_number":8},{"id":"c0096ff52bfb3bd1e0d816667ffc2641494bc31f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-16 06:26:05.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b33f7efd97acb6effef7ed26d1275ddf94b46d6b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-16 13:22:40.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"986afa345fc1a7012c50f8bc97171297fe26eb0a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-17 19:23:10.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"36e983a67a42ebf5309693657efa26a09002aacb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:23:27.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 900435,9","accounts_in_message":[],"_revision_number":9},{"id":"b15c5b571927430c42a24c2e4b6df34dec61011c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:23:29.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 900435,9","accounts_in_message":[],"_revision_number":9},{"id":"0e511bcbb2d3ae853f29b96dbd29d4f19245ede9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-17 19:23:31.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 900435,9","accounts_in_message":[],"_revision_number":9},{"id":"cdb54064c35401f55e1fe9f78b369c14c0a55c50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-20 05:34:00.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"475a8a0db3d7336b166f4137b66eb593d35b1933","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-20 05:34:35.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f2aa05af6696ac06bb1cf7af2bfc1aeabcf9af5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 05:56:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa435e452250444582efcbdd60efcad1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2beb33575b0f4fed8335e2c296089770 : FAILURE in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68019da3b944437922afad2c8308c9b : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b342a2c8194fcaaca7d624db09b361 : FAILURE in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f62e3b3776e4475abdcf1df42cd1dd3d : FAILURE in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/802a9787d67149a9b79154615e6215d2 : FAILURE in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cd26a42910042ab8f29427fd0323a2d : SUCCESS in 2m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2b42e2c3c899460c98923f510e0cf484 : FAILURE in 2m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6ca89810a8df4c19b5f1300219536143 : FAILURE in 3m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/848894c157c84e798ce6c9650ac7ca09 : FAILURE in 2m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/26438639ba834c39be189fb90167239d : FAILURE in 3m 32s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6ea8f3bea0574cc2aad85c26c023837f : FAILURE in 3m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9b3cfd6d373d45afaa6fcb1778e5890a : FAILURE in 2m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1952b42b339847a1a14adf9079e9e675 : FAILURE in 2m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f9094d540cd84dada474dc2220020d40 : FAILURE in 3m 02s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/f6124f2806504d7c8563df091e7a1229 : FAILURE in 3m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e090c6900a444e1195c4167daa941fa7 : FAILURE in 2m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c2c00cb337d54e12b428436fb493dd17 : FAILURE in 2m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/04a40bae6fad483585fd600ba83b2a5e : FAILURE in 3m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/65e294ff52084e0998566756c7e2ec0f : FAILURE in 2m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52acc7f866c34a708073b1c9bdef22f7 : FAILURE in 5m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0409e1f65360412fb2fba1fe33367a5b : FAILURE in 3m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/684a04d566194c7692ddc0efd6187bf8 : FAILURE in 3m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ea1c88e11b3d435fb2a3833ef13deebe : FAILURE in 2m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cd1da6f8211546de99d43c4e83e19c2b : FAILURE in 2m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e64f711ba12e4787a4587e3e09f207ef : FAILURE in 2m 40s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9ec128ae0d6d4d369664e2da9735a8c8 : FAILURE in 2m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a1db801704e544dda21496d1048171b7 : FAILURE in 2m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/955e8736d005401bac792eccc3c46787 : FAILURE in 3m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/155a5ee82cdb4b9ba9364059879cba60 : FAILURE in 3m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ba8c386fc6ee4eb391e0bfc04d10618c : FAILURE in 3m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5b3f45ff3164b04851f7985a0bb626a : FAILURE in 2m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53efb99fdf1545539aade76248842f62 : FAILURE in 4m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e3ec21b421714aac89e35bbfad432b37 : FAILURE in 4m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2e6dae966b0944b9a660eb1b4fcb700c : FAILURE in 2m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22bc55de5e0a4228bf073a9e6a44fb50 : FAILURE in 3m 19s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/23a638b806be409989fef8367b39bc48 : FAILURE in 2m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a165b6c231f46eea9fe56b29725d31d : POST_FAILURE in 6m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e44ba2ef2634af19a4076cf97020659 : FAILURE in 16m 26s","accounts_in_message":[],"_revision_number":10},{"id":"6765598d88acda017d40570f02e65a51494f9f8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-22 20:04:10.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"f19cc5087f6fc57def31a9f9a0aeaf3ccb9ad4f4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-22 20:04:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9b0c68a32fa8cc74be1dec8dbc9972e83b80700d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 21:24:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c8e91da2bb42b6b4685f32230a8db5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552b31d140614d67a8e67697e551fb5d : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7665a88f9584ff588c2f6bfb9de0437 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23a49680461f42a58033a75009f666c7 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef777577b9cd4e2da99b99be846a02e1 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36c3c5811605487fb66e7b227de8a3c7 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12865ac8594d4936b35913ef86c549e2 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/81a22bf59f1f405ba989fce4bab97040 : SUCCESS in 9m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cbacefbc9de4466a93c16f006c315aa8 : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/478a339e813e4a989a1743d94c10df9a : SUCCESS in 9m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e38abe93d5a440fc8ea752bb6a781997 : SUCCESS in 16m 07s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/03453c43338444f5b90a95a20bf4db05 : SUCCESS in 7m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/16ff6648caca447c99b234592599dcb0 : SUCCESS in 4m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/42ac8ce6728642ed95ec7413237b460a : SUCCESS in 4m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/493740944bcb49709a4c5e9c4dbe119e : SUCCESS in 11m 04s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/d03b1ab9320e41f896cbc6dcdf630368 : SUCCESS in 5m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7f00b6f2f8741b7afa9ef262e93b298 : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0f951f145ca541ef882f10802248432b : SUCCESS in 5m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2eeba6a4fe964267831aba01dc220888 : SUCCESS in 8m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c1eba4bb37654cf384ba108d419ffd27 : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a61fa4229e9648dcb9296d4f1dcbcb6e : SUCCESS in 8m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/72ac386207544966896320bf86c1ea06 : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0de96125399b4d3a90bb20749112022d : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/18be849fb0e44896953eb864ee587ca4 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/65782892e917495891bfa560a78ee21b : SUCCESS in 10m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/684d9ec2b255491fb2023e4f5ffda12f : FAILURE in 4m 12s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6e6508e1650c4896818d1efe0e761883 : SUCCESS in 13m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/071b8460ba12462aa289c2b0f003ae46 : SUCCESS in 25m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/17b5d7ab7fe1421688a57fc5d0cd02e6 : SUCCESS in 12m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/73a86059994348aa8a60819f4cfb97f6 : SUCCESS in 15m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dcb3e857cd2d4f77be3853c2b26949ba : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a6775c7065a41af80ba1095afeace5a : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1cff25680e74b8696686e187d538bd5 : SUCCESS in 21m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f6f7a847f9584dcaaff2deb2b2dbc75e : SUCCESS in 5m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c19ecff5e4f64c858c56d579e4fcad4f : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69a53f6103ca498fa0c2bb6f22818a9e : SUCCESS in 6m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bf2a276056dd482086433d03d6d38bc0 : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd8a7d90e5f14cfb8dfea3a4c1a95623 : SUCCESS in 46m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ff8eb36c08d44a9ac8d2b52242db993 : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":11},{"id":"b9fc6acca4af71d82aa01e45df123d8eade8eb34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-19 07:52:32.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"c65c96c19097dabf8b46a533184c2eb1fa1a993c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 09:32:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"278ae68b4946805851c76aa2a2179334d91b72f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 09:54:17.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7df7d3550a244bb39d8361fa8bb34b21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9b122620024b66a4a4fd262e28e2ca : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/004116cfa1b64d92846d4d975f85e980 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6055b75c04a8454a97caabc6def533ba : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3e9fca661464b53952b391e6935623e : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7efccb97145941c7a5103a8f16d84c8d : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f11d629f517e49d3990c274cac4f7ecb : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1499d24f6e8d4e26a2fbd050f6e7d846 : FAILURE in 4m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8d9c5b6c2ca644c79d3a8cc0f7224a9f : FAILURE in 4m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1b44f9ecf09e4f51b073232970c230e6 : FAILURE in 4m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2c69605c527849cba9e2879d95d02e8d : FAILURE in 5m 01s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3a6c4cbe8692466a9fd727074ec05e24 : FAILURE in 3m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/09c534e546bc42e8bb54bf0223bed704 : SUCCESS in 5m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e570f55e0aeb4670a6c87bfbeff60b88 : SUCCESS in 4m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/55cb3f9264c74f93a9954a1fcd026658 : SUCCESS in 11m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/24227816346e4f58ba86544ea3976bd1 : SUCCESS in 4m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/83c79b8001514f4f868db2ff4e7908d2 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dba6613470804b0d894eba6bf2884ac7 : SUCCESS in 7m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8c0ad5b11c78478c9e316a56f8fc2a09 : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18f5fefa1b494a79a6fd0c673df3b062 : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d3dab5687299427b9a18e412ed648dfe : SUCCESS in 11m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/039cc0640fb84500bf48595d7c28e7f1 : SUCCESS in 12m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ec32adbab1d24691a3c16bc2235dfb89 : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a95141b501704484acb1f16ec9b6764b : SUCCESS in 12m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8ac528d1900b4062a191348edd03f964 : SUCCESS in 4m 18s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2106d53d160d4fbf99ec1dcb1d2bc604 : SUCCESS in 14m 00s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4bd744d7047f42de99a26d89ef981765 : SUCCESS in 22m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1c8ecf5e165c4789bd4b31343356f8ee : SUCCESS in 7m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cd85461ad797403f97f9567bd5353f57 : SUCCESS in 13m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/06daf93c2e594a0ba1c637f31d058976 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9bce6068be14fce8d5afb0271b8dde5 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f09726dfef1c4d1e95686f29d26054b4 : SUCCESS in 27m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fe06da941f544077befeaaaeb0ff6aa5 : SUCCESS in 4m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/33ab8c37c0e14b3699680eb09fbc4969 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e03fd8afafe45f69592cdd43a5c6281 : SUCCESS in 6m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/04aabeae24fe4312bb369101e36ce9a3 : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/633a413015064e9485269559dceebc42 : SUCCESS in 29m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88aefb277092488b83971f80b2633145 : SUCCESS in 1h 52m 32s","accounts_in_message":[],"_revision_number":12},{"id":"2be93cae3a2a35c52486c3dbe7ee904b7153f886","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-19 15:35:09.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"20fbbf006ec08778d28cc024b9791b68a9a9a6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 17:26:28.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51c0a6c0d5464c8397d90526ee4dcacc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ec398915a04d9286aa18e74f5850b7 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1cd83961aed4b6c99cb3755df428d17 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e78a5031350140858999df742640fa8e : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1be2dace3124aa8a3c9c725b5e52523 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/588f5a072aec4fcea366fdaaefacc4e1 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdd5dc7a36ce4fc8b2e2826de9e76424 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/72f0b8c5e9de424981dc3f45933ea88c : SUCCESS in 10m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9da8c7563ee64468a82be6b5e0c6d74f : SUCCESS in 14m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce439dd8fbb44c95acf67e43b3886dc5 : SUCCESS in 8m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8984cec647034c03aaa7b72251d237ed : SUCCESS in 17m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/46d836d0ce3246ffa2f3c1ede73ee3b5 : SUCCESS in 6m 17s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d7ffd7f63be24e478e2e8026b3e416ea : SUCCESS in 4m 38s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3877978ab5b14b16b073550c7aa988b4 : SUCCESS in 3m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d0a8e6951c4747b8a4258412b89737bb : SUCCESS in 11m 58s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b7a69f1de3b542a6bc57b363f3c63c25 : SUCCESS in 6m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/43c1e963867d420ebba873e34a9c714b : SUCCESS in 6m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/905f57418d9243d0b85eb687125f1f08 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/de4ab4ae5bc74b148ee8e3e2900855e0 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58c166cd42a94be8951201272a425409 : SUCCESS in 7m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9f749764f2084fb68d7c48401a85d5bc : SUCCESS in 7m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d352f9524eab48a3a570555f53487bef : SUCCESS in 11m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/aff7f31ad74f4d869937a3ee95d9c179 : SUCCESS in 4m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7e0958bf86324748827af16541c4a09f : SUCCESS in 9m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7515faab287c4d42ba1aba6131aa4fa2 : SUCCESS in 5m 17s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d51961680fdb45d7ab9e7e7ebf2d8c52 : SUCCESS in 15m 29s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7c7d0488e2534e9784bb0b9ffb64d7bc : SUCCESS in 25m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ba08341982ae4a26884c253d4565e300 : SUCCESS in 7m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/33ebd95666c947caa451fccc879b17d9 : SUCCESS in 13m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/386b2d95aa1b4d1bb90609c19040432e : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8080939b62e4184a9ca4cb2f84621b5 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95992fc6c603404599f8149a35f37fa4 : SUCCESS in 24m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/746a78cac6aa4c97b35507817fea8607 : SUCCESS in 6m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ff63ab8fc3284d8ca696fe633a34c66d : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9f20581c85944d4a56a0492edb7e71f : SUCCESS in 7m 11s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/cf1e4f96a0ec405b8f922e97479811d2 : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0dd706e513f4f1ebcbb25d5b5f355a8 : SUCCESS in 38m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/150c81a7534948278e3ebbb085e7b126 : SUCCESS in 1h 29m 56s","accounts_in_message":[],"_revision_number":13},{"id":"15d900ac98da650e9efc556b741803402a50738a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-19 23:18:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":13},{"id":"a1da2cc4db666dda93a56d9439b4c69a10d817c2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-20 11:35:15.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"149b86234d7b978f6fa38cdaaee033089d76c6d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-20 11:35:56.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"043f4bbc59224e842dbf3d25c5ae183eaf3e1fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-20 13:16:24.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d949337145f4b67917092c91964b073\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e806b51a1bd4dcbad0be45e244b1b91 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8a6359c9844f3390ae6991ede6dd13 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/229107a9bae8430a87e5e690199e9171 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d0eca70510f45c3bdd812111eb0ece7 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76e730eba2cf4a31b71c2d49e074cd13 : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aec0faf4dfc54b009c058821cc5a0e0a : SUCCESS in 4m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5df4da713814060ab7bfd2bef54a68c : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dcef98f112694087a242ea71c400d417 : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/763ecc7b52d546f1925709c2c1cd82e6 : SUCCESS in 9m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/15247e55181f4791bb5097c72400cc25 : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/10d134c1996942eeb25952218ea5ca01 : SUCCESS in 5m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7643711373d948b69b4e78bd7575073c : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1d4ed4b7aabd4a1bad3e60e7bee7ff1f : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/47f2a8c91ae8415587c2d274561917d7 : SUCCESS in 14m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6baecfb2f4fa4b06a65b0eb7a1f0b4b2 : SUCCESS in 7m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7af10db4236f4f48b580468369daed9a : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b569a1bf58dc42148f0f60775e5e232f : SUCCESS in 11m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/28872824db1444f0a5d2162f0a1812ba : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5c232b5a796d4e71bb74eb1c6bd93a49 : SUCCESS in 11m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e0b3a1214482408791275ce22ed1f7e8 : SUCCESS in 25m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4d0bc16458c543b68c2d95e182a8f5cc : SUCCESS in 7m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6855951104f841468647bb774e0e5f19 : SUCCESS in 12m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/292da229a29c4ce8a72d5846d2da09d3 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/416a88f2d3684b7283af69e8261769cd : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/371e6892bc8046e4acc4179282ee6c6b : SUCCESS in 22m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e8e9c295fcd24c439415cdf355ecbd22 : SUCCESS in 5m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/507739b1b1b9446b93468874adddc255 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2a3e775f6cd44cda8bda65c82df36a7 : SUCCESS in 6m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c31f04df0f3c44db98b99253b20a461a : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/325615ba60b14ca0b192a97b4540d85d : SUCCESS in 23m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b79912fe2d6433b94c314128538e478 : SUCCESS in 1h 34m 41s","accounts_in_message":[],"_revision_number":13},{"id":"97c3e9334d65656ab276cb108d8dce83c73fd875","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-20 13:16:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"f9ab26f2ab851fd4d3360ee4693006a5a05e3f30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-20 13:17:18.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f73d9c361444788be77630aff3cb8fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb15477f14e7422ea7de591875421153 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":13},{"id":"16bbf0934215e771dfdd193ffe1eba2703abdd8c","tag":"autogenerated:gerrit:revert","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-21 17:54:38.000000000","message":"Created a revert of this change as Ib7f8191c980972764bc3a07136603a5cc10ea732","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"eb2b2f2765933473802150f3da4af386256e02b1","revisions":{"46247cefaa873cd273c52038943a12446c7eea56":{"kind":"REWORK","_number":1,"created":"2023-11-08 16:20:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/1"}}},"commit":{"parents":[{"commit":"8b18b77560c641b80464692028ef14d9743bf4a4","subject":"Merge \"Bump Sphinx + deps to latest version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b18b77560c641b80464692028ef14d9743bf4a4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46247cefaa873cd273c52038943a12446c7eea56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46247cefaa873cd273c52038943a12446c7eea56"}]},"branch":"refs/heads/master"},"674ed5a82350ee78f7e895432d16edee0a11ee4c":{"kind":"REWORK","_number":2,"created":"2023-11-08 16:41:24.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/2"}}},"commit":{"parents":[{"commit":"8b18b77560c641b80464692028ef14d9743bf4a4","subject":"Merge \"Bump Sphinx + deps to latest version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b18b77560c641b80464692028ef14d9743bf4a4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:41:08.000000000","tz":0},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/674ed5a82350ee78f7e895432d16edee0a11ee4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/674ed5a82350ee78f7e895432d16edee0a11ee4c"}]},"branch":"refs/heads/master"},"f2c73a01376d4aae25abaadd82bb8fe0806f765d":{"kind":"REWORK","_number":3,"created":"2023-11-08 17:29:32.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/3"}}},"commit":{"parents":[{"commit":"e8425bc831c159b873cdc46a21ed68ba18b2eb04","subject":"Merge \"Add typing_extensions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8425bc831c159b873cdc46a21ed68ba18b2eb04"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 17:29:04.000000000","tz":-360},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2c73a01376d4aae25abaadd82bb8fe0806f765d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2c73a01376d4aae25abaadd82bb8fe0806f765d"}]},"branch":"refs/heads/master"},"c106e2a1c031e47f5cc1a8b55ad6f42bfed35b80":{"kind":"REWORK","_number":4,"created":"2023-11-08 21:06:44.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/4"}}},"commit":{"parents":[{"commit":"e8425bc831c159b873cdc46a21ed68ba18b2eb04","subject":"Merge \"Add typing_extensions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8425bc831c159b873cdc46a21ed68ba18b2eb04"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 21:06:30.000000000","tz":-360},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c106e2a1c031e47f5cc1a8b55ad6f42bfed35b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c106e2a1c031e47f5cc1a8b55ad6f42bfed35b80"}]},"branch":"refs/heads/master"},"58890139c408aaeb23e51436abf32967947ba766":{"kind":"REWORK","_number":5,"created":"2023-11-09 15:41:00.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/5"}}},"commit":{"parents":[{"commit":"1a2d1998e655e5b5b0afb96018f1ee04f289642f","subject":"update constraint for pbr to new release 6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2d1998e655e5b5b0afb96018f1ee04f289642f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-09 15:40:05.000000000","tz":-360},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nEnsure we stay on SQLAlchemy \u003c 2;\nAvoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n * https://github.com/pyca/pyopenssl/pull/1223/commits\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58890139c408aaeb23e51436abf32967947ba766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58890139c408aaeb23e51436abf32967947ba766"}]},"branch":"refs/heads/master"},"550fe583a9f8a68b16f1f6a17c381c3863e3c247":{"kind":"REWORK","_number":6,"created":"2023-11-13 19:19:12.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/6"}}},"commit":{"parents":[{"commit":"82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66","subject":"Merge \"update constraint for testtools to 2.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-13 19:17:28.000000000","tz":-360},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nEnsure we stay on SQLAlchemy \u003c 2;\nAvoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n * https://github.com/pyca/pyopenssl/pull/1223/commits\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/550fe583a9f8a68b16f1f6a17c381c3863e3c247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/550fe583a9f8a68b16f1f6a17c381c3863e3c247"}]},"branch":"refs/heads/master"},"53ab4439c06d00b10518584321f6b0bdd29af084":{"kind":"REWORK","_number":7,"created":"2023-11-14 16:21:36.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/7"}}},"commit":{"parents":[{"commit":"82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66","subject":"Merge \"update constraint for testtools to 2.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-14 15:57:34.000000000","tz":-360},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nI have included the following changes:\n\n- Ensure we stay on SQLAlchemy \u003c 2;\n- Avoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n  * https://github.com/pyca/pyopenssl/pull/1223/commits\n- mypy \u003c 1.5\n  * pep8: commands[0]\u003e bash tools/mypywrap.sh\n    nova/virt/libvirt/utils.py: note: In function \"create_image\":\n    nova/virt/libvirt/utils.py:161:21: error: Argument 1 to \"write\" of \"_TemporaryFileWrapper\" has incompatible type \"Any | None\"; expected \"str\"  [arg-type]\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53ab4439c06d00b10518584321f6b0bdd29af084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53ab4439c06d00b10518584321f6b0bdd29af084"}]},"branch":"refs/heads/master"},"22246cdf640de6bb4da9e67c192edaf6449c9dda":{"kind":"REWORK","_number":8,"created":"2023-11-16 01:02:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/8"}}},"commit":{"parents":[{"commit":"799ba44062027b90a88849d1afa3c549ae396044","subject":"Merge \"update constraint for metalsmith to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/799ba44062027b90a88849d1afa3c549ae396044"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-16 01:02:02.000000000","tz":-360},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nI have included the following changes:\n\n- Ensure we stay on SQLAlchemy \u003c 2;\n- Avoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n  * https://github.com/pyca/pyopenssl/pull/1223/commits\n- mypy \u003c 1.5\n  * pep8: commands[0]\u003e bash tools/mypywrap.sh\n    nova/virt/libvirt/utils.py: note: In function \"create_image\":\n    nova/virt/libvirt/utils.py:161:21: error: Argument 1 to \"write\" of \"_TemporaryFileWrapper\" has incompatible type \"Any | None\"; expected \"str\"  [arg-type]\n- alembic \u003c 1.11\n  * manilla:\n    - AttributeError: \u0027BatchOperations\u0027 object has no attribute \u0027rename_table\u0027\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22246cdf640de6bb4da9e67c192edaf6449c9dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22246cdf640de6bb4da9e67c192edaf6449c9dda"}]},"branch":"refs/heads/master"},"f1576dea48037edb4dcfd68019fee02d0efce253":{"kind":"REWORK","_number":9,"created":"2023-11-17 19:23:10.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/900435/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/9"}}},"commit":{"parents":[{"commit":"799ba44062027b90a88849d1afa3c549ae396044","subject":"Merge \"update constraint for metalsmith to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/799ba44062027b90a88849d1afa3c549ae396044"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-17 19:23:04.000000000","tz":0},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nI have included the following changes:\n\n- Ensure we stay on SQLAlchemy \u003c 2;\n- Avoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n  * https://github.com/pyca/pyopenssl/pull/1223/commits\n- mypy \u003c 1.5\n  * pep8: commands[0]\u003e bash tools/mypywrap.sh\n    nova/virt/libvirt/utils.py: note: In function \"create_image\":\n    nova/virt/libvirt/utils.py:161:21: error: Argument 1 to \"write\" of \"_TemporaryFileWrapper\" has incompatible type \"Any | None\"; expected \"str\"  [arg-type]\n- alembic \u003c 1.11\n  * manilla:\n    - AttributeError: \u0027BatchOperations\u0027 object has no attribute \u0027rename_table\u0027\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1576dea48037edb4dcfd68019fee02d0efce253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1576dea48037edb4dcfd68019fee02d0efce253"}]},"branch":"refs/heads/master"},"2015c8d11bf53da0700bafe89bf4e2dda2497e92":{"kind":"REWORK","_number":10,"created":"2023-11-20 05:34:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/900435/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/10"}}},"commit":{"parents":[{"commit":"3a122203dd3aecd85436469587d8048083d8fcca","subject":"Merge \"Revert \"update constraint for tooz to new release 4.3.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a122203dd3aecd85436469587d8048083d8fcca"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-11-20 05:30:43.000000000","tz":60},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nI have included the following changes:\n\n- Ensure we stay on SQLAlchemy \u003c 2;\n- Avoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n  * https://github.com/pyca/pyopenssl/pull/1223/commits\n- mypy \u003c 1.5\n  * pep8: commands[0]\u003e bash tools/mypywrap.sh\n    nova/virt/libvirt/utils.py: note: In function \"create_image\":\n    nova/virt/libvirt/utils.py:161:21: error: Argument 1 to \"write\" of \"_TemporaryFileWrapper\" has incompatible type \"Any | None\"; expected \"str\"  [arg-type]\n- alembic \u003c 1.11\n  * manilla:\n    - AttributeError: \u0027BatchOperations\u0027 object has no attribute \u0027rename_table\u0027\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2015c8d11bf53da0700bafe89bf4e2dda2497e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2015c8d11bf53da0700bafe89bf4e2dda2497e92"}]},"branch":"refs/heads/master"},"6859975ff7b93c9345da14cb38be2014ab1c9971":{"kind":"REWORK","_number":11,"created":"2023-11-22 20:04:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/900435/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/11"}}},"commit":{"parents":[{"commit":"070d2765b13ab305a9987a6019907714254dbb08","subject":"Merge \"update constraint for neutron-lib to new release 3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/070d2765b13ab305a9987a6019907714254dbb08"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-11-22 20:03:32.000000000","tz":60},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nI have included the following changes:\n\n- Ensure we stay on SQLAlchemy \u003c 2;\n- Avoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n  * https://github.com/pyca/pyopenssl/pull/1223/commits\n- mypy \u003c 1.5\n  * pep8: commands[0]\u003e bash tools/mypywrap.sh\n    nova/virt/libvirt/utils.py: note: In function \"create_image\":\n    nova/virt/libvirt/utils.py:161:21: error: Argument 1 to \"write\" of \"_TemporaryFileWrapper\" has incompatible type \"Any | None\"; expected \"str\"  [arg-type]\n- alembic \u003c 1.11\n  * manilla:\n    - AttributeError: \u0027BatchOperations\u0027 object has no attribute \u0027rename_table\u0027\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6859975ff7b93c9345da14cb38be2014ab1c9971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6859975ff7b93c9345da14cb38be2014ab1c9971"}]},"branch":"refs/heads/master"},"e4b95986b6fb98f2d76192cfdb560074a2b99ef7":{"kind":"REWORK","_number":12,"created":"2024-01-19 07:52:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/900435/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/12"}}},"commit":{"parents":[{"commit":"047670f3e7080d202af658dca3e99577e0dce4b3","subject":"Bump Django version to 4.2 in upper-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/047670f3e7080d202af658dca3e99577e0dce4b3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-01-19 07:52:02.000000000","tz":60},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nI have included the following changes:\n\n- Ensure we stay on SQLAlchemy \u003c 2;\n- Avoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n  * https://github.com/pyca/pyopenssl/pull/1223/commits\n- alembic \u003c 1.11\n  * manilla:\n    - AttributeError: \u0027BatchOperations\u0027 object has no attribute \u0027rename_table\u0027\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4b95986b6fb98f2d76192cfdb560074a2b99ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4b95986b6fb98f2d76192cfdb560074a2b99ef7"}]},"branch":"refs/heads/master"},"eb2b2f2765933473802150f3da4af386256e02b1":{"kind":"REWORK","_number":13,"created":"2024-01-19 15:35:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/900435/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/900435/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/900435/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/900435/13"}}},"commit":{"parents":[{"commit":"047670f3e7080d202af658dca3e99577e0dce4b3","subject":"Bump Django version to 4.2 in upper-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/047670f3e7080d202af658dca3e99577e0dce4b3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:19:36.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-01-19 15:34:19.000000000","tz":60},"subject":"Manually generated constraints","message":"Manually generated constraints\n\nI have included the following changes:\n\n- Ensure we stay on SQLAlchemy \u003c 2;\n- Avoid pyOpenSSL 23.3.0 as it removes (deprecated) code used in barbican\n  * https://github.com/pyca/pyopenssl/pull/1223/commits\n- alembic \u003c 1.11\n  * manilla:\n    - AttributeError: \u0027BatchOperations\u0027 object has no attribute \u0027rename_table\u0027\n\nChange-Id: I72338abeb617ff554d862fa5e4fd230545145ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb2b2f2765933473802150f3da4af386256e02b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb2b2f2765933473802150f3da4af386256e02b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
