)]}'
{"id":"openstack%2Frequirements~900436","triplet_id":"openstack%2Frequirements~master~Ie622ad6c36214527776eddc17ae1382d621c3c2f","project":"openstack/requirements","branch":"master","topic":"uc-testing","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-18 09:01:54.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ie622ad6c36214527776eddc17ae1382d621c3c2f","subject":"[DNM] Tools and data used for previous commit","status":"ABANDONED","created":"2023-11-08 16:20:51.000000000","updated":"2024-05-03 17:33:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"36e6fb43466df14b743e1b6487266e5db865fe3c","_number":900436,"virtual_id_number":900436,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-16 03:01:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 17:57:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"573963b8c31a6c05a8feeb78a01672f9c4a282af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 16:20:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd965331aef4fd9f30a7334b5b830048a769681d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:29:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48f09fa275a7abb4f1ce77895426cde77dea439b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 17:57:24.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/670216419c124020aa7415b7423f07c6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dacdbea8b5554a1b8c6ad768f6d704b6 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f578562417e48388f90a45628c9308a : FAILURE in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbfbc81540764a6d94f95cec338a2906 : FAILURE in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62a5abb2e7e64433bc44d98b5944ea73 : FAILURE in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/afe836ba829149ac8c4041624c65022a : FAILURE in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f86b270a077c4d6bb29ecfb241bab6c9 : FAILURE in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/121ef1ee4fd94ce39fab0066a67c411a : POST_FAILURE in 6m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/518372dbae184a3794c6220eaff0d633 : FAILURE in 20m 02s","accounts_in_message":[],"_revision_number":2},{"id":"03ac3759e62455f8ea987b76eba1a3455062c0ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 21:06:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"30db7a76de58af30d7b8cb4e72158d9d232b98e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 22:51:27.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1caf7db529d45ce83956874310d8f36\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/39e64e1277da4057be2b2dddfb249bd3 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd927f4b5fff47a0822c7fb9ffed5145 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5195195e4d3f4e7b951d7a233223fc83 : FAILURE in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33187e4111b441e8b88ec6621a82b3e8 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0367cefda4b64cbaa71f4611ec2eef74 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/030d813a329b4a3ebcd99bde15729b70 : SUCCESS in 2m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b22e9d48d2fd4e03af9cf3d40291cb9a : FAILURE in 1h 08m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1de83a440ff640b0abcdb8a93ae9ba75 : SUCCESS in 1h 38m 44s","accounts_in_message":[],"_revision_number":3},{"id":"a2a7f771f921db26487f70540dbdc742724fd22d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 15:41:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ba09ef0f8600c37df90e0a4438c6e554ff85d1b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:38:27.000000000","message":"Patch Set 4: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f226e807cde647298c46dfddc1e7602b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd76163928c14808b70d8a214dba23f6 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63cbb014ea674967aecbf8b064e70126 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db356ab4bcb144db9ea6ad2c2bebbe83 : FAILURE in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/991684ef35ac45568bd00bb1f95d1229 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f41d77d01a64c60aa7f6bcebea670e5 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3df095682df44838e3c8e7c2511943f : SUCCESS in 3m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebc48299bfb745f8b06dae5fc2d22d96 : SUCCESS in 1h 04m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f2a8ae4d6d045eda6d789bcaa5a1c14 : SUCCESS in 1h 48m 34s","accounts_in_message":[],"_revision_number":4},{"id":"6ca0668d9429575d7ccdc4433b99143e47ab8a37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 22:27:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"52120cb55db2ccdbc6b0c9a72dccd9f52d245e73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 00:15:13.000000000","message":"Patch Set 5: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91ecceb639ee4fb4a2d51eaa52028c74\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e41a4c09318347eca8708f27b5779904 : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff182572917843729f168a36a1deb38d : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69abb8eecf1c46129ffed302ebf94796 : FAILURE in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/102b8d73db4c4ef19087c934407c7cd2 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1843a872bb384678ba89e90202505923 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0134910e1a146a7887d39c157820b1d : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d925c7a3754b4da4bf3a6ceb6d8ac5a2 : SUCCESS in 43m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b872cf0ba33e4dca9cde8f67a9d56b9a : SUCCESS in 1h 40m 35s","accounts_in_message":[],"_revision_number":5},{"id":"0207f4f46a72c974b4f970f2e392e3ee3f453649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-13 19:19:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"446e356923674eff7969c7fa5911f9d001fde362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:37:05.000000000","message":"Patch Set 6: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/217275b13244432aa75fe5f3dbc980de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9cd8e778e34434d8112075cbd43e68c : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dabd24b62984b60b4acabfa5025cb55 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9447623897b143cbae41f7cb23846e56 : FAILURE in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/182f9a15efc447c58b1134f1d6ea76f6 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea73548dd56c43418616443be96b3418 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2073ff4651cf45b1a3bcd8474ead9ff5 : SUCCESS in 4m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42c0970b5a254cfc81fa3561c4434f7e : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22656a6b64844c398d84f3a475c77794 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":6},{"id":"c37d9500ed7e24d4a4b817e41231fccd1ac44f48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-14 16:21:36.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"45ea31ed5daef0878f393d3ac470f3cd68c787f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 18:03:38.000000000","message":"Patch Set 7: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af49b2bd2a0c45dc801689b178a15563\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/129a3a30b81a4b4c8cd59444053e73be : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4df8048af1431faeda76e32e14dafd : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540304341f5a4676bc49dec02420a7f2 : FAILURE in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41ccdb0ee9a84a489ae60d5c6430d571 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8ba1c182c4046b6b48f713ee1c04a12 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8cc4110c64ab453d94384ec209f1feec : SUCCESS in 4m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/876e7af2d8b044899811e1fc59f567ce : SUCCESS in 1h 06m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84fd62016895477ea003e2eb66ba271d : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":7},{"id":"57748382ce073bfd29bc0e38ceee455dd3728a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-16 01:02:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fc424b42960924e2cb59b74c52a1b3c9a11e340d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 03:01:38.000000000","message":"Patch Set 8: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb12a3ad00e04719854646b316017d15\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a014b1ce35d6406b88b9d91d6b7d4892 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff41002475f4ab68c490a36e6aeeb87 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13af96549734e3c8830aa0775ca204d : FAILURE in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fe8e0f6108649b099da517342c96948 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae2d0854b68845b8b7e80a12bf6115c2 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e364cb11a4f94bbab8ddcde2ef37dfdc : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad34182e5cd0495288d34c9c9d53a3d5 : SUCCESS in 41m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc2220eaed584da8ba0485524fd145cf : SUCCESS in 1h 50m 56s","accounts_in_message":[],"_revision_number":8},{"id":"9e526e39455e6764579eeffb8930f5ad3fce8220","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 09:01:54.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"36e6fb43466df14b743e1b6487266e5db865fe3c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-03 17:33:42.000000000","message":"Abandoned\n\nhttps://review.opendev.org/q/topic:constraints-tools-updates is a better solution","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a1668de5a6bfb389da615ef66ccb47679f3d6115","revisions":{"29ed39d67fedfb712e88c2ce8fb569f8f63377af":{"kind":"REWORK","_number":1,"created":"2023-11-08 16:20:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/1"}}},"commit":{"parents":[{"commit":"46247cefaa873cd273c52038943a12446c7eea56","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46247cefaa873cd273c52038943a12446c7eea56"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29ed39d67fedfb712e88c2ce8fb569f8f63377af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29ed39d67fedfb712e88c2ce8fb569f8f63377af"}]},"branch":"refs/heads/master"},"f43e73d4e01daefbcdcbca0604a076b713929f1f":{"kind":"REWORK","_number":2,"created":"2023-11-08 17:29:32.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/2"}}},"commit":{"parents":[{"commit":"f2c73a01376d4aae25abaadd82bb8fe0806f765d","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2c73a01376d4aae25abaadd82bb8fe0806f765d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 17:29:10.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f43e73d4e01daefbcdcbca0604a076b713929f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f43e73d4e01daefbcdcbca0604a076b713929f1f"}]},"branch":"refs/heads/master"},"a9a49e9b79444004741386e52c7db45773134efa":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-08 21:06:44.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/3"}}},"commit":{"parents":[{"commit":"c106e2a1c031e47f5cc1a8b55ad6f42bfed35b80","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c106e2a1c031e47f5cc1a8b55ad6f42bfed35b80"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 21:06:30.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9a49e9b79444004741386e52c7db45773134efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9a49e9b79444004741386e52c7db45773134efa"}]},"branch":"refs/heads/master"},"d664496d78b09b1468267d2c36f5b3dcb025a372":{"kind":"REWORK","_number":4,"created":"2023-11-09 15:41:00.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/4"}}},"commit":{"parents":[{"commit":"58890139c408aaeb23e51436abf32967947ba766","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58890139c408aaeb23e51436abf32967947ba766"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-09 15:40:50.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d664496d78b09b1468267d2c36f5b3dcb025a372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d664496d78b09b1468267d2c36f5b3dcb025a372"}]},"branch":"refs/heads/master"},"c42b784f323ef3b87c656159c73a0f3d0a3c6d9b":{"kind":"REWORK","_number":5,"created":"2023-11-09 22:27:52.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/5"}}},"commit":{"parents":[{"commit":"de91c8af45b658db7fbe5d8b4581e2fa0fec962c","subject":"Simplify upper-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de91c8af45b658db7fbe5d8b4581e2fa0fec962c"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-09 22:27:46.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c42b784f323ef3b87c656159c73a0f3d0a3c6d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c42b784f323ef3b87c656159c73a0f3d0a3c6d9b"}]},"branch":"refs/heads/master"},"6ba896ca80c0a1fe79a3875db2b00b390e7356bd":{"kind":"REWORK","_number":6,"created":"2023-11-13 19:19:12.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/6"}}},"commit":{"parents":[{"commit":"550fe583a9f8a68b16f1f6a17c381c3863e3c247","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/550fe583a9f8a68b16f1f6a17c381c3863e3c247"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-13 19:17:39.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ba896ca80c0a1fe79a3875db2b00b390e7356bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ba896ca80c0a1fe79a3875db2b00b390e7356bd"}]},"branch":"refs/heads/master"},"c682a5a4951b5c18f3dca0cfc8fde6f4c4eca4c5":{"kind":"REWORK","_number":7,"created":"2023-11-14 16:21:36.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/7"}}},"commit":{"parents":[{"commit":"53ab4439c06d00b10518584321f6b0bdd29af084","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53ab4439c06d00b10518584321f6b0bdd29af084"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-14 16:08:34.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c682a5a4951b5c18f3dca0cfc8fde6f4c4eca4c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c682a5a4951b5c18f3dca0cfc8fde6f4c4eca4c5"}]},"branch":"refs/heads/master"},"a1668de5a6bfb389da615ef66ccb47679f3d6115":{"kind":"REWORK","_number":8,"created":"2023-11-16 01:02:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/36/900436/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/900436/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/900436/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/900436/8"}}},"commit":{"parents":[{"commit":"22246cdf640de6bb4da9e67c192edaf6449c9dda","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22246cdf640de6bb4da9e67c192edaf6449c9dda"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-08 16:20:31.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-16 01:02:02.000000000","tz":-360},"subject":"[DNM] Tools and data used for previous commit","message":"[DNM] Tools and data used for previous commit\n\nChange-Id: Ie622ad6c36214527776eddc17ae1382d621c3c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1668de5a6bfb389da615ef66ccb47679f3d6115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1668de5a6bfb389da615ef66ccb47679f3d6115"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
