)]}'
{"id":"openstack%2Frequirements~900508","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I09331f5263c41108a5b48b2352a2c83e552d1d66","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I09331f5263c41108a5b48b2352a2c83e552d1d66","subject":"update constraint for glance_store to new release 4.3.2","status":"MERGED","created":"2023-11-09 09:59:33.000000000","updated":"2024-02-14 02:22:53.000000000","submitted":"2024-02-14 02:22:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900508-new-release","meta_rev_id":"9239f856cacd55b8dd237dde97af1fca0a5b48c7","_number":900508,"virtual_id_number":900508,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-14 02:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-25 13:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-25 01:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-01-25 09:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-01-25 09:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 10:04:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-09 11:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 17:20:22.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2023-11-10 17:47:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-01-24 21:32:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-01-25 01:45:25.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-25 09:57:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-25 13:20:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-01-25 14:48:36.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"}],"messages":[{"id":"b0a495c9988957986e2c935ebf324205137c6556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-09 09:59:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58801c2fb578d08a7d8c45ab02c1c9cc8c6494ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 10:04:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9ae4bb3e226482bb5c067550fb966aa\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a1ff94b36dd4480a8c14fde513ed6f3b : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ed271aeacac2a9f18911e7269b4445d9593ff09d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 11:03:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f61f0cd95b443cbb1ea1c38697cbc50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/397bd7cecc864719adda7c7feaac9710 : FAILURE in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b58d29ba8c1d46f49e5e4ac61d6df589 : FAILURE in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c57b197a19924df9bf6b2eba14701e48 : FAILURE in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e1e188afa5849c6b59e41414017ebfe : FAILURE in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e554168d5854740aa8062618f69f94c : FAILURE in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a5ab6a3740664b4fbadcf7d1e34fbb6d : FAILURE in 11m 13s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/80b6c1c6968a41f2978525f47ec92179 : SUCCESS in 7m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bbddcc41f9f24e549ed881daf0783f7e : SUCCESS in 4m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bbc1c8d5a4c24a3fa538f37ce9a5a367 : SUCCESS in 4m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0fef0e4ee3b0419884cb6eb5eb829300 : SUCCESS in 11m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/425904c5102c4ddf9d8bef635e5df086 : SUCCESS in 5m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8126c81fa99a41d792ea33a96891e285 : SUCCESS in 5m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ed2cfb43d7a04b85965289d3a5fea490 : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d480526e972743e1aa9661175ecf0b12 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cf22434156641a2a1736d22a7750a26 : SUCCESS in 7m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e38fd153b9f74ec39d866606579233a4 : SUCCESS in 6m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7277f3289e1942c0a130dd46f90fd3dd : SUCCESS in 11m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1766779f256440f690840b6b23ff8811 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3b827e0cb2f14ff796b2960a4e3d66f3 : SUCCESS in 10m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a7098f249fd84a9bae84100b7a4316d3 : SUCCESS in 4m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9734eacd23d54827a1e7d2948882dcba : SUCCESS in 9m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ff1b71b5e6094805acd811c8fedc505f : SUCCESS in 21m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/90d6d228b6574172aafa86e35fab3866 : SUCCESS in 11m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/df7ed5e60d5442d687d3b79e33379adf : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7316de8052644ecaa47be5177c61a1ca : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5fff4b303501466c8eaecfbd8465d231 : SUCCESS in 22m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e42706610a514728a11325975eea6746 : SUCCESS in 5m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9bd35a5cc59c4055ba756e975034739d : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/257285290ca04768ba0a3b2f2091862e : SUCCESS in 6m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f6176841367f48688644b8303b8fbabf : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8bf334c8c2b4ce9ae387d4110a52123 : SUCCESS in 45m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd3d52c4368646df919ebf64eb36a2c0 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":1},{"id":"04e6d2f054ced48e8e2e08c63a3bd42cc5a6f537","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 17:20:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62ae6e60544b1f2ab0768fcfdf582a8e8300bc1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:27:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32d1b6559c2d4791940759e380890a75\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/87d99f1df8bb4cfebaa047b0aeff3cb4 : FAILURE in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9d87da0e8c7d0c65a7cd48577a11585707330a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:48:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b16b7d8a95a84158bec27be74040d37e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c817f26fd9174310913cd8b0d321ebdb : FAILURE in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/558bdcb178c8466099ebd6e6d252090c : FAILURE in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b0e6eda200c4952be771c402c6d68b2 : FAILURE in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c215ed075f404095bc2ec50921a116b9 : FAILURE in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f234a105ea5d4333ad22d80e08a4e0e5 : FAILURE in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/423ee7d616344b028fbd78692f367e87 : FAILURE in 10m 56s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2feb6a1f4fa2438b9df68d2a8efc856e : SUCCESS in 6m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f58a586eb89c4209a8746580ba75320d : SUCCESS in 5m 28s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/80e7011b7806412d877f9f81fcfbae6b : SUCCESS in 6m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/77ac10babbd74dd6a0de81935b3961d3 : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d5763aa181164287bf95bfb1ef8c8689 : SUCCESS in 5m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9f658ff491b34d3081fb48c2fac5a58c : SUCCESS in 5m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a0563d4fc49c4915b0ed7981c9bd1489 : SUCCESS in 9m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e1cebf1b1a0d417ea6966239ab001b82 : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92eb832e9693448895a4d112894f3c4b : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9d600b88c3424d3bb4490ea4748a1c9f : FAILURE in 8m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/64da6a5f0f6f46eb8683edc084e2c66b : SUCCESS in 11m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4dbdb3624a174668b241603e940b79a1 : SUCCESS in 4m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/76f27a379f8548198b8a4d1023edeb59 : SUCCESS in 8m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/80e6df58dda94f8b8010b1c6fb4f8365 : SUCCESS in 4m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5438310e458c48ee993bf643c27196c3 : SUCCESS in 9m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e9843c32584a4494a92c152dbaf1ec3b : SUCCESS in 21m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dc18b143218b44968f2dd700af2fc41d : SUCCESS in 15m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2f83dbe5badb46598e6d9874c10d03c4 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/942506374d9b4c3183894371ac246083 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f1c7c04cf2b4f628626f731d713fbb2 : SUCCESS in 21m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/08c732c913ee4f5e9593a3bb92dc19dc : SUCCESS in 4m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/05c3c92e64734689a44e6dc270ae5ad7 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9887780de914bbc9b507dc2aa5b6100 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0dc407912e59412f94bfc95d2478f081 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba49e95487de4acb8512adaefba26232 : SUCCESS in 56m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b79d5a816494ffab49b953c4c04ba62 : SUCCESS in 1h 22m 03s","accounts_in_message":[],"_revision_number":1},{"id":"771a40db5b8e22ecc6e084f0c62b6a215a90f541","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-10 17:47:51.000000000","message":"Patch Set 1:\n\nrecheck\none more time","accounts_in_message":[],"_revision_number":1},{"id":"f6cc51f5b2e212a89633ca482496249693d04fc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:54:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc38e67622f467c888d17194b281ec9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7baac9031804b4eba3b22951dacc527 : FAILURE in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fe9bb7290f611cd3a8bd8014e91e0a35d717320b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 18:57:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb20b145f75f4d01a153f7e32e80f937\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71cb876ddaa646b8ab12fc2a4b67c248 : FAILURE in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f22717600ebe4305abb920d263bd245c : FAILURE in 2m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d06b31065c6d4fe4b7ba8c9055168d20 : FAILURE in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3fb1afd20e1a42beb7f3153fb824aee5 : FAILURE in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da56ca4b422d48f59cc92a81cc67207a : FAILURE in 7m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e3c656985b844a549b045beca6b18ef9 : FAILURE in 12m 58s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/27766be7c2284d5b80d63e3924c106a5 : SUCCESS in 7m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/dbc21861115a4cdd8cf650b832ac00bc : SUCCESS in 5m 25s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2e2df28839e446ad84e0f8871efa2009 : SUCCESS in 5m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5dc8a1c823264ba38972f8cbb744671d : SUCCESS in 11m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4f933954131943efbfcb8b3926ce219b : SUCCESS in 4m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b4422aaddfae4310be71bc53a138ccf2 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d2b9f709751444f8b55561eb429705b9 : SUCCESS in 8m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/759e5640a54c4a94a9abb804674a41f9 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc4cffea27a54ca28e1ec44228684abf : SUCCESS in 8m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b674c8e35ebb4bba84338cdbf855424b : SUCCESS in 7m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/01aefc8bad794519b9940e7e08ee7227 : SUCCESS in 12m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b547df1820d14a33af1d6de33fc4097b : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2464f9c54cfd47b6a8ece5e83b159ac7 : SUCCESS in 11m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/20b51c1bbc914f3da1f6a50a42de5bc7 : SUCCESS in 5m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4c8852ec46134f10b6a374fe6ddb793c : SUCCESS in 13m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1479a4fd365c4b6e89ccc2f79e43140e : SUCCESS in 23m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cc9d1d9921ac49378b1cdfd993ec416b : SUCCESS in 11m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bdcc3089a48c4353872f63f6222098bc : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43f45bf755bb404196479bd8a1e4d3b9 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/054785e978c34b33a354bd553181b70b : SUCCESS in 24m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/997e94a1b5394d04beb867df6fcacf31 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5511f6bb53dc46aea325d66055c2dc5c : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14a3f543dd2b452b8b0cbf22aee77110 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cbaae156e27441408a0af02c40f1e3e6 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1453f369bc6f4b23991bf5af6365b079 : SUCCESS in 1h 01m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3ae12798cff4153ab7ca86b2a045e48 : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fe887afc49d69f3b7fa66e821009c7eca0120f8a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 21:32:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfc4df985e7e7ef03bf1d994c71eb8a21440c837","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:38:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/444703ed626a4b3b97ea298109151e6d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/438c174a83634ca299c0f01d1a6d784a : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bbc71017d5d3bafd6d73113bf07cfb7289efe53c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:00:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fa7262a785448eb8494ad9a7f1344d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07ac5714dae47f697916582138b67cc : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40243a11681d4ef98792f8cec2ab4059 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6461748c47441cf8423fdf007960e1f : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c7d69ed225c489f906c207b4e5f0db3 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46786cb6330c403986ebcf5494d46bbf : SUCCESS in 9m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d6c93df9c5b640e4bfff911c9ddec552 : SUCCESS in 12m 49s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/33886879c6ca4bbba47012a538498118 : SUCCESS in 8m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/09be8e82b4e84939a0ec7b1cc0eb36a9 : SUCCESS in 5m 46s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/83b30e701e524e0abd7358c2c4dbdbb3 : SUCCESS in 7m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bcf380e43e724e9e8caeb1bf34d2a08f : SUCCESS in 12m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/624c371bfdd543f59dbff7cbce5ee6dc : SUCCESS in 5m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6223613ff2e342dcac407c79396a7cea : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/71550aabba4c4570851c7e2462828bc8 : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/47fdd3f3a67e43b392d0adbca20c13c8 : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ea037ffbefa4d999790287276eff3c8 : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9234fbe4e3174382b209cd068c87cd75 : SUCCESS in 9m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/169302c7285b434da5c49f0639a40c40 : SUCCESS in 9m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b6c507f394fe4602be82ba1dee167682 : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/16fb1386b44d461dac4ff71f9671dd05 : SUCCESS in 11m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a6f317486a9840f5ba437d174376643f : SUCCESS in 4m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/075236c8e7d1481d839f900e03301735 : SUCCESS in 10m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/febc52615ee54f578ade128252bc1cfc : SUCCESS in 23m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/179ffbe39e94471b968f3da37e26be67 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/076cd7f9a6f148f2a3d1ac056ddecb4f : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c13b85a8a7a147fbac5f6f380348a882 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b05ff87593924c1e87e585bf1f2a1c31 : SUCCESS in 24m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1e14abc3730e4241a392877692479db5 : SUCCESS in 5m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/019ff50202164113b602b0e2bcff80d9 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c3be3ba647c4571a2b2f327f033d2e8 : SUCCESS in 8m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f938139e8d754338a5948b3fb4c2a7cd : TIMED_OUT in 40m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7be6617386de4bcba873f673dcae5601 : SUCCESS in 52m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0a46514fd69489a86b74fa1d91ee04c : SUCCESS in 1h 20m 58s","accounts_in_message":[],"_revision_number":1},{"id":"97e8d60e41c95c427a341a49f3cd95eb75bbeda0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-25 01:45:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9758fc3c68374a55b3bae6bd31afd17241f89cfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:51:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4ec87892d2b4839be42e94d842b207b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af540b768f854260a29b2e354f637613 : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"614a66cb8e824ee3d4043d6c4b8b1d0abff87f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 03:19:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd3d34615a4414092d4e87db43683c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8266e7327424a7a9f39a95993f52b3d : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdbfe526fbf44695883872ec8f6ba1b1 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6417bbd5fd3a484d9406313b09807fc2 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e918bcd16b04e13bda5b24801b40da7 : SUCCESS in 4m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc2eb79731db47cbb95ed58718e06c5b : SUCCESS in 10m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f400adef64e147d19a73b50be24b86b8 : SUCCESS in 13m 37s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6629271e5e2a44358141545069e80de4 : SUCCESS in 9m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b7772e9d5c984977b470c025ed8bf0df : SUCCESS in 6m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6f46489d4e5c4b348b18089b042b7999 : SUCCESS in 6m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/66f90583c1f147e89ac2104c3c6d6724 : SUCCESS in 10m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7ac05900ca4847d7bff92feb5c80c2b3 : SUCCESS in 5m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2b413a1ade1647089db5876256405e73 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9d974e88e82d460ea24ef7b9b9cf5aaf : SUCCESS in 8m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/90d350b8713c4325a9cd684bf8bec73e : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd29f96e2ac04d1790550a4dbe1c6c9d : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bdd0f6562940402db696f920e8099c00 : SUCCESS in 8m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b25bfa4888074d97a8a42b30d2749eb9 : SUCCESS in 14m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/51eda1430fa4409091579976a413743f : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4b643b9143b64842abc59bd410f0ab2f : SUCCESS in 8m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/360d15eefaf74d79a178279d17b2f99a : SUCCESS in 6m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6b8a3d0480f34cf1800d6c8e06c9eefe : SUCCESS in 10m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/22b6d1a5ade74defba7412c4dcef3ec8 : SUCCESS in 24m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2e23a6c6d9d441a287c0db35252702d7 : SUCCESS in 15m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/37286097ad134f0784287b0415d3ff0d : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9bd34d9cea1c4035947f37d90a3b46ad : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9ee334f5a344b078237831802f60d4c : SUCCESS in 25m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/eac02a6d5d93407ca34cbf18f4e0f58a : SUCCESS in 6m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/19bd13abdd5f4e999d8441d4880c722a : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d15a9f98a0a4107890468e94ce8ece2 : SUCCESS in 6m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b14549a02c2a49d392a74479302c676c : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9918c4d718274575aa726b9c0ce58644 : SUCCESS in 1h 04m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e913ae07d8154132a94d8abe48d1506a : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b546a1744fecdd97c8e10fb3d0fd30ab31205cda","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-25 09:57:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79a02bf252519b6d7cd6d64ddfee98dc216be73f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 09:58:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db59b095f2008410802d1a143f016bb367a50d7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 11:53:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d119183ff13481d8aa5b48e572c2155\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183b758e6f7b4b3893f709e882a12c42 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc7ab236416c46b092d61785e582b6a8 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45b73ece9d1943ada43d7633c3939fb6 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8f7a7cc8a714bdeb7ca0c68d2e835c9 : SUCCESS in 4m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/626a0f5b9c644d128df97c091e87cbbf : SUCCESS in 10m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/16b975fa2bfd4b1a81fbfd81bb51db5f : SUCCESS in 14m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b912b2fd24e64515b8e6d20fb79c6109 : SUCCESS in 11m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b7139a62d1ad4057ab2b6e4e926a97f7 : SUCCESS in 5m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/602d437c796c4b75b44db73e92bb3929 : SUCCESS in 6m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9591fd03a3604cf5bb9f12036d0f21f9 : SUCCESS in 10m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c21f9536f7944897920c441ec4c7354c : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3776978a8b54475db732a41cfe9a0702 : SUCCESS in 6m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c2cf3bf78e794f19bb07ae153faa8fd7 : SUCCESS in 10m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2957438c37734fed8739557719132e62 : SUCCESS in 12m 39s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/417dd80f97cb46039836ef06a54dd1e5 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d1c2aff3387146a5a51dc4a9672c1599 : SUCCESS in 10m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3a4729a58cbc4b4e9a1c7ea9472950cd : SUCCESS in 4m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6563ff7ba2104f62840526d3c5723751 : SUCCESS in 23m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/340f805a9f6d4da9b7185e29a874a956 : SUCCESS in 13m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/452fafd8e4174950a18c595a2b97ef69 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f67fdc1a6de4c98a7ab171dbb83c046 : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6276c8b7dcc43c9aa8c3b37a2a8719f : TIMED_OUT in 40m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7b25b97db2cd4b9d8cb05a34ffc6c3b6 : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/540a31e38dcb4058abe6c2de3b0cef8e : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24d8b5ffaace40ff90515048b92094e1 : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b532f21480b743368f2bc534202b9315 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53991e3b86b84f7aa9411fca4570b3c8 : SUCCESS in 1h 05m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e46e261dc03b4361a9d6a66a83e194bb : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8b6abdba3426fc564fd7343a4ecc13d90ee89739","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-25 12:30:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c2f08c241419ae1e3ba2449272dc8d3f524b9f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:39:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6897ee580e04870a71f7364700f126e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4a6a53c3ec224531a24cb4991e5d3c20 : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cbfe111720752df6714c3e9fdceb59aefe024400","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-25 13:20:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"08c9334f9638d0b2963d6f956e564e5530927139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 14:04:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc0bf2197fd74a3ebb956dec70a8724d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a92941ec9e046529375a399f7b40acd : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71d341b288cd44b3acf4139da841f364 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5696cf1aae04fe79bb4e79257c567ec : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aabdf72b0e5a4a25a32a6d2fb8fec58c : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6cf8876022a5477db6b309fcd0ff06f8 : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/beef487ae92c43f79dee42e14bdd6da8 : SUCCESS in 14m 00s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ff4d4824c1864a778e4c29f9e29e158e : SUCCESS in 8m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7d8eae143b714676acf40f25cce95257 : SUCCESS in 5m 41s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/337aaa1f40314215b9a2699c3329fb07 : SUCCESS in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/24bbf76aa2aa4d8e8129fdb149f13933 : SUCCESS in 10m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7add12d73cdc4614a7a8c9e31d44a7a4 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fc616629ac8a4eb29c5f0039f4e3f6c8 : SUCCESS in 6m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0c0e0a827a7e44c183c2ee5dd1d3be15 : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9f87984ef8824084a14e25a326f6c693 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be78c7f0d4344938ae50f39a30fb309d : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b08fc7b1264d4592b45f7fbbebe78577 : SUCCESS in 8m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7f0200bc5dd94892a0c40fd636f71012 : SUCCESS in 10m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3e6177ffe50c41ab95b9847fe9a07f08 : SUCCESS in 4m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b28de94d98c3453b813a3647b66d0167 : SUCCESS in 9m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7e14f4aabe9e43ce9da2cb10627a5db8 : SUCCESS in 4m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f2dc22b331b141b9b8189a48297e1a1f : SUCCESS in 9m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4b7abc5f8e2c4bccb49d0dbd1df2cc27 : SUCCESS in 23m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/11c413a6a6b24d2d841b514eaf702cd2 : SUCCESS in 13m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/af6035c0247c40fb87fea1ec74a26999 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57308d50ca054494856ba960b6b05538 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e485baf66457414fa8b79420e50441ad : FAILURE in 24m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/784317636e5b4087be730e3269201107 : SUCCESS in 5m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b73162791bb449f38b1fb7a2b5b0875d : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a0bde5dba214ad3a19bcbedbcea97df : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/14bd8e1058ec4f76a36c3ea9bc05a883 : SUCCESS in 9m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5214fa4215c44ec863e9319e76c13c5 : SUCCESS in 1h 00m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b97a350c90a8483299113c79fd81313c : SUCCESS in 1h 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9fcdbf80cdadbfbe2ce66482b69cc7ae113c68f4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-13 23:49:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17ab6c0b5f1a47d8e89876f57280b1b23786496a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 23:54:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f070a60770744e51ab301661f52b5839\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c0254dd563ec45f0918301a3eb8759e2 : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7e0b3fdd2c6988f993823ac63ca6f565000e724d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 01:16:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1052d11a430d45888dcb32f3cbd651be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d04b0c893504d80ab3532f3a326ee58 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f4496e93bb745459f05c9fa9453bb2c : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11367846a56c4070b2ce98f3411ef92d : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/667711712dd040babc21d9cba0a65058 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/90c6b6bb3c9645759278c26974d375ff : SUCCESS in 9m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/72b8af38908c434d8f14850bf8e57a41 : SUCCESS in 13m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8a94c2f5ca254403b4fb4fecfc3c90df : SUCCESS in 7m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/856869dae12f41b0abaed5e1506b9148 : SUCCESS in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4cd896941c704f17aee32de4489aa264 : SUCCESS in 4m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aae69a98a244467095b2e656d81c23d2 : SUCCESS in 12m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/60ee3e0173a44696bfcc5eb162b44477 : SUCCESS in 6m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f7cbd4455a6a43188893275a5ecb04af : SUCCESS in 4m 35s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/50ce1010fee84df9bfd244c573068ddd : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/925f6c002ecc4849985a4fe82d5efdb4 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b4413c5433644b3a2cae1ea11d1cdca : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bd8ced7f6b324ea2b0342ac3354044cb : SUCCESS in 10m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/717b2cd7967740199b4e12e5566a2e26 : SUCCESS in 11m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bc95d2794a8e43ef993b8cb492c69463 : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ec71224661584ab0a1fcf904b4fd422f : SUCCESS in 8m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6547e90a9ea14ac28800e62e51304173 : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bdeaf3ffc86c401e8c3a5f3f0d25a1f7 : SUCCESS in 12m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1b27b737012541b6b3e726b86d34f41c : SUCCESS in 23m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ef2ed75d4b004cdcae126d3d573f234b : SUCCESS in 12m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9524d4d5656541f49ca9f1e54a0b8f35 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c7c0bfb4c6f47b7b30ac93d8cf6061f : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ee0830ee0e149ff8385cba61ce54f86 : SUCCESS in 26m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/53f4030ad2ed43dc8ac4bfe719bfd138 : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d303f42c0b424272813dcebefcd53fd2 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc3891cf22d740bf941d6ed35c54b183 : SUCCESS in 5m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e2158466d7074b3ba40c7654e6abf11c : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1399bb951e6f449c92a38fa9ff157297 : SUCCESS in 59m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24d8176b078e466b9bed7a03d99bfbca : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":1},{"id":"bdbe2a5c4372b45fcc703bdfc141cdebae43e484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 01:17:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97c44baabb93a4ee344af517c58e8b646e16d8e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 02:22:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d750c558ff7d4f5791c50bdb506405f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d230a6d4229d436688c2990680138705 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5e7adf8243243ccb5f60617f01ef5fe : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3256822bb63b4e0b815df032bebb41ae : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ae59129d0c64498aa8d4633156931b7 : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/186641865c554f75a8935833edc5c3c8 : SUCCESS in 8m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/89a7ae74250044a5b2ec6973d602fc2d : SUCCESS in 12m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3d4f7dd80906462cb2dbf415739c144e : SUCCESS in 11m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/15f6e7c9575343b492a218124f2cd3b0 : SUCCESS in 3m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/49c18e18fc86461ea819554f53dc5891 : SUCCESS in 4m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b73731b3299044f78c4154b13534bd98 : SUCCESS in 6m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5a02bcf7ec2c4412bc3073b14e8a157f : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c145e8cbb7e49479f575aadb8aeae63 : SUCCESS in 7m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/57034f7595dd413fb59b7683f6cf3fc0 : SUCCESS in 10m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0c82c8362d744b2c880009f44d126ed0 : SUCCESS in 13m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/561476846a3e4ad6a7efc93b85624966 : SUCCESS in 3m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e2bb3c82e790433e8ed46351f56990b5 : SUCCESS in 8m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2ae3f015b9804ff6a974e5541f31faf3 : SUCCESS in 4m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/248708f804374a76b3aa439ad0da50d7 : SUCCESS in 24m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b43f1d76bda34b9d880e09f2afcae20d : SUCCESS in 10m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/59d3b8055177480f897f932e59542832 : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2dd371c2b1b44050b9ba7b904abbcd8f : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9bfa91fa868e48d1907d2ae0641a116b : SUCCESS in 29m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6651c1144c424beea3fdb17fc700422f : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e9cf53597d08469f8cca7247801ce2d8 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7cadef847e9e472f8306a120ef79bf32 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2bb42f300e654434853079405b47eec4 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e189372285834e89b828f4aed212c766 : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cdc157294244b218ea16fb4dfe5fcad : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9239f856cacd55b8dd237dde97af1fca0a5b48c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 02:22:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b79cdc19c6673e9783d3a042dff3b21d184b9138","revisions":{"b79cdc19c6673e9783d3a042dff3b21d184b9138":{"kind":"REWORK","_number":1,"created":"2023-11-09 09:59:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/900508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/900508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/900508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/900508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/900508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/900508/1"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-09 09:59:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-09 09:59:31.000000000","tz":0},"subject":"update constraint for glance_store to new release 4.3.2","message":"update constraint for glance_store to new release 4.3.2\n\nmeta: version: 4.3.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: PranaliD \u003cpdeore@redhat.com\u003e\nmeta: release:Commit: PranaliD \u003cpdeore@redhat.com\u003e\nmeta: release:Change-Id: Ifef92009d01d09f52de28771b44e7f404a145c86\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I09331f5263c41108a5b48b2352a2c83e552d1d66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b79cdc19c6673e9783d3a042dff3b21d184b9138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b79cdc19c6673e9783d3a042dff3b21d184b9138"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
