)]}'
{"id":"openstack%2Frequirements~900534","triplet_id":"openstack%2Frequirements~master~Idd599d80573f4740a4729d27106c329cc13ac25c","project":"openstack/requirements","branch":"master","topic":"add_py312_support","attention_set":{},"removed_from_attention_set":{"9649":{"account":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"last_update":"2023-11-10 19:26:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd599d80573f4740a4729d27106c329cc13ac25c","subject":"update constraint for testtools to 2.7.0","status":"MERGED","created":"2023-11-09 15:10:25.000000000","updated":"2023-11-10 19:27:27.000000000","submitted":"2023-11-10 19:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900534-add_py312_support","meta_rev_id":"706b8b0619d70bd04b40a8a4aca276cbc72cc5c9","_number":900534,"virtual_id_number":900534,"owner":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 19:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-11-10 17:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-11-10 17:47:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 16:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-10 17:47:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3c065c3c5d196c7efb6ecfb8a6bc043d6c566ee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2023-11-09 15:10:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee5abf0ff8294aaad71af81d8c0fe25b43535262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2023-11-09 15:27:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c894d297336466a64da2ae71accafee6d09a6959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 16:46:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efcef35bd2f44f7984a6740ba7118dcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c748116262c24f7ea9b7ac175c3cf7e5 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5b3cc248664396947d5e039bfba231 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f110d36972c747c1a2ee0ccec48a42ed : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b050b802e19c4c14941f933aca2b8450 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c498fae98478438a80d4696c18bb2e8c : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7d6e2ed24684fe4baa9a419ac632181 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f9194511064f4e659ce74ee95cb57a59 : SUCCESS in 11m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ec969712e7e74c7d8a6c3aff6d973912 : SUCCESS in 16m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/883e37c8515f4464922b6e5f63b1b3a8 : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/92834a78e9e04cb993b25f3365a98cf7 : SUCCESS in 16m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ec2be5cea66d4181a4fa581061ba82d0 : SUCCESS in 7m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e66fdf3018d6420db478a43e879f6bbe : SUCCESS in 4m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f77e255e5bf9472dbf1edfdd0144d671 : SUCCESS in 4m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2b65c0c0a8c049e6bc59e60a2f6cebdb : SUCCESS in 10m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bf95b39567354a3b9c0f2231049b3f6b : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2035523060d543a09eed5b8af7905cdf : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cf38f062a3be4bcca915c5fab5dc3f37 : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3e008a96f6b045e995a7093ea81642a2 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58faaae895cd4fee9d3eb263ac67dc98 : SUCCESS in 9m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/904859a20ec24e299ced066cad8d635d : SUCCESS in 7m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/997fca10f2f04bb182353f7d5bc51eec : SUCCESS in 11m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fb10ebc1a7ef493993bd0c614497fb06 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7e09d601cded4bc0bd3d4a26d9cdc5ca : SUCCESS in 10m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cdc2d2218b814b4888a49636d7354c2f : FAILURE in 4m 38s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b9a952f026ed4814bd25fd409745dd0e : SUCCESS in 13m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e8a01b025eec457494f28ab93edf9b65 : SUCCESS in 26m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/af17588eee6d4fa89ef2833ba51ebb10 : SUCCESS in 12m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/555758b187e3490aa6359ff1f05ef60d : SUCCESS in 12m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e1574766efdc47fc8e00fb7b889bb261 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/068dafa85fe046068ac0e1eba8b803f4 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9912d07bb7e4641a956288a6a637ef8 : FAILURE in 25m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e0c32408a8674127b1cfcda973b7e9c6 : SUCCESS in 5m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/55ee24099a9c41858f4200a3ade23086 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9d1901888734473a19c09d76854019e : SUCCESS in 6m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/046863d426664051945470b575296503 : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83b16bdecd5344d1bf70c7a59ad05ec9 : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05f62dd718554974970dc798eda1c959 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ede74f7cd589cf49fb96096e55fa3cd0479ea2ab","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2023-11-10 13:04:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f108f332ac8f01869d5519691a1f87fd5d53303","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2023-11-10 13:04:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bceb826fcfe32836326af1d04e6ec137352b8873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:56:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/705458f22d0d4097a00f16d1fb8e02e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8230997e07f44540af85a6210039a24b : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/478f09334bf34bcab0d6374ac346e9d7 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/109e9d77c1784c3ba0d5c16f51738ab6 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52d6e9d6cd3e4c4ab5ca958ca361d2d1 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe0581f4c55d4b0fbe30686fde8a0639 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60ddacec02e448849a44cdc150722814 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fe207c277a84e258e773916cc8fb0b4 : SUCCESS in 12m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aa4d42b2db35415e9858f1ca8575b244 : SUCCESS in 15m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/281d2292eabc46bbbd9b416fe62483ba : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/990c7038bbd24e4799d7a4a724cdb14a : SUCCESS in 16m 34s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d4c0bb82ae32469ead1f8b7a5810ec0d : SUCCESS in 8m 03s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/26a7cf2e339b44a7994f6145b21c0eeb : SUCCESS in 5m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e6744c42217f4bcdaddbd1d1bf9e658f : SUCCESS in 5m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ec91829a23b244a494b982d80431ba91 : SUCCESS in 10m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8063f71f18f748e8ae697d408863a63c : SUCCESS in 5m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8a08987d61db45399e7ea7479d8571e5 : SUCCESS in 5m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/74b7e72d83154efe82bac031aebe919d : SUCCESS in 8m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4b6c47702708441293cafa52a7dd7370 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14db04a72dc847699238cc8ee94c58d1 : SUCCESS in 7m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2604853f155c4b55bfa38ff8069b5eac : SUCCESS in 7m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2678ff148fdb4155a8bb563fc7697361 : SUCCESS in 12m 39s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/21550c9c42184b4694fd6c5483b34003 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/737a37e086f2452392885e43850e88c0 : SUCCESS in 9m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/84f87cd2db494f1d8bc6fd52f8744095 : FAILURE in 4m 18s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/359bbc63df3448b68ae795e7daa875db : SUCCESS in 10m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6a4df8ac210e47539d130c0e46319361 : SUCCESS in 19m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2da0b7af711a491facd70eadab48bbc1 : SUCCESS in 13m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9046fbe1e13349458e224228d198b4f8 : SUCCESS in 13m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b810f928fcb042ad98b99638980dd4df : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99d42d2a08ad4b07a92921d295737d2e : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c3075d870434795b5dd3c4567fe808c : SUCCESS in 21m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2ee67206b23f494bbbbfcb8d12e1a2b5 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6b9099267f4f49e0bba80bd5dc59f3df : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab5f5f574ae34bf48f388885c469d3d8 : SUCCESS in 7m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7631102714644c7f8ec954264f079e93 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac0439ce594946d7a0ed7ab6009b166e : SUCCESS in 1h 00m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4152297e9530452d9d4bd694433bdce8 : SUCCESS in 1h 43m 22s","accounts_in_message":[],"_revision_number":2},{"id":"625d2b3ab65989448e26a142f6bf20c870451ee0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-10 17:47:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb89022a84dbcb8735da06dee346b3ad60303f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:48:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"717f4854b1f327056ad0becb75e95d2887099d89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 19:26:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aa70f81ff81422fb56cd1bd15b73c94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d37e9a87de3405ab74f367b7812b28c : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73476534291d47ad9050bb3526e38017 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b65e5b5e0442a1b6ba27bcacbcada1 : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa499eb9de9f4641a5a55d73aebd9fd1 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db74b597abc7452c9a4fae83a8c6e2c3 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f80ff76651de4f7086dd637b81744619 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a0f56dddffe4c13913e2e6a50eba55c : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/faf24e8d0fe04e6cbdf323e77c92b618 : SUCCESS in 14m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c89c07fbf7914116ae5543ee594e8f97 : SUCCESS in 11m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f4a997716a0f46b8898caa8616b56ce9 : SUCCESS in 11m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aeda55c5d84e4431a846d57277c6bbd8 : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/36728fe1cb574857971ed18fe38695c0 : SUCCESS in 6m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8c84042d51c742748bbde1f7f4bfa1c0 : SUCCESS in 8m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c51d9a4f5e33441f927671bcdd5f7b19 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d63808e2c9b04ee58748e7b6724541e6 : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c1fabff714924dd1ba92bfcb2da0b96c : SUCCESS in 9m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/66e04dcc661f421bb37a0079b6a0fdfe : SUCCESS in 12m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/98dbb98ca0024cdb9fb287ccaafa7895 : SUCCESS in 5m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9ec01db3b486492bba268a3ce460dc6e : SUCCESS in 10m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/99c97b133bda4e36b0b27bb538ae93ff : SUCCESS in 24m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/60d349f4b4784df4bb53d85d7fea656c : SUCCESS in 12m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/245859894284454e9d8738725e69f998 : SUCCESS in 14m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a430b15008e24c7f8d1fdb95901f2a60 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f813b11251db4763978d521ec76ba68c : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0433d38102b4411694e6a9a2f33dfbc1 : SUCCESS in 20m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/66bc978f58a6449bac04d2f03c38eddb : SUCCESS in 7m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d2aa35557828497b94912bdc925ba8c5 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1107a0e2b2a46cf9a6b8c6267b785f1 : SUCCESS in 6m 59s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/87db9403148e43ae9ec4468018a57f56 : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44566b053d344edd9dff5d58db374479 : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1afb637eb024ec0aa64f0d509340afa : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a3a02f43deab6c1c9966a9ce09e75b860d8f67df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 19:26:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"706b8b0619d70bd04b40a8a4aca276cbc72cc5c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 19:27:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c57e13bf22426e939fc53f6cb8587a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c064067aba47c4ba48e8850ba29943 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73ff9c82be03cec9b68e33dcb9f52038607d5494","revisions":{"df8cf5afd20f187a195c264ec95a690d3d587598":{"kind":"REWORK","_number":1,"created":"2023-11-09 15:10:25.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/34/900534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/900534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/900534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/900534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/900534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/900534/1"}}},"commit":{"parents":[{"commit":"1a2d1998e655e5b5b0afb96018f1ee04f289642f","subject":"update constraint for pbr to new release 6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2d1998e655e5b5b0afb96018f1ee04f289642f"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2023-11-09 15:08:51.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2023-11-09 15:10:23.000000000","tz":120},"subject":"update constraint for testtools to 2.7.0","message":"update constraint for testtools to 2.7.0\n\nFor Python 3.12, stestr does not work unless testtools \u003e\u003d 2.7.0\n\nSee:\n\nhttps: //github.com/testing-cabal/testtools/pull/353/commits/18e1a7157562b00655561ee25ce0b41d383b2ed1\nChange-Id: Idd599d80573f4740a4729d27106c329cc13ac25c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8cf5afd20f187a195c264ec95a690d3d587598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8cf5afd20f187a195c264ec95a690d3d587598"}]},"branch":"refs/heads/master"},"73ff9c82be03cec9b68e33dcb9f52038607d5494":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-09 15:27:00.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/34/900534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/900534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/900534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/900534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/900534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/900534/2"}}},"commit":{"parents":[{"commit":"1a2d1998e655e5b5b0afb96018f1ee04f289642f","subject":"update constraint for pbr to new release 6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2d1998e655e5b5b0afb96018f1ee04f289642f"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2023-11-09 15:08:51.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2023-11-09 15:26:30.000000000","tz":120},"subject":"update constraint for testtools to 2.7.0","message":"update constraint for testtools to 2.7.0\n\nFor Python 3.12, stestr does not work unless testtools \u003e\u003d 2.7.0.\n\nSee:\nhttps://github.com/testing-cabal/testtools/pull/353/commits/18e1a7157562b00655561ee25ce0b41d383b2ed1\n\nChange-Id: Idd599d80573f4740a4729d27106c329cc13ac25c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73ff9c82be03cec9b68e33dcb9f52038607d5494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73ff9c82be03cec9b68e33dcb9f52038607d5494"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
