)]}'
{"id":"openstack%2Frequirements~900564","triplet_id":"openstack%2Frequirements~master~I3c4966861cf767940b82d9138e51dc0bb4eaa5c8","project":"openstack/requirements","branch":"master","topic":"uc-testing","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-18 01:41:30.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I3c4966861cf767940b82d9138e51dc0bb4eaa5c8","subject":"Simplify upper-constraints.txt","status":"ABANDONED","created":"2023-11-09 22:27:52.000000000","updated":"2024-05-03 17:34:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"0e969f1e98d530f19b857dd37c30efdb40a617a6","_number":900564,"virtual_id_number":900564,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-10 00:12:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 00:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"caf5263a574de0317a215c407c64fa007c9a13af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 22:27:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86cf5f024709d07b9969de3fb557e44aa3e0c8c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 00:12:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5aec3599a65499588d8e8ef6150feda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf932cd23404f43b98a61457308b05d : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819958f54a954d19941a13bb2774c931 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/994d0559d63f4508b16e36d6a5855b99 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e3da90341c54bd2944aecaa6ea76351 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2cd8859f3c14f528f8bf455961e4514 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81adf0d407dd4fbb8a6d221f7d8269b0 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b9d1453154c141d382d52a39d2ea8e5b : SUCCESS in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e7dee30ae04345129f07c8db5dfbdeac : SUCCESS in 15m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5adfcdc14d5248d3bfc96a35e71fc90f : SUCCESS in 7m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/33df21220f1d4c309fc4432fca3397cc : SUCCESS in 18m 12s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2c0a2e031dc244d8bc88f5275452504c : SUCCESS in 8m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/dadcb73394754100b7edac8b143cf2b1 : SUCCESS in 4m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4b943775e2624d0c9e3966dd9774a614 : SUCCESS in 3m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/67bcf851b1484cf2a1e1ef42e631dfdc : SUCCESS in 11m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/09d5153df61447f9960c5d8b511a563c : SUCCESS in 3m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8f6f0f05319a4117813c969c6f161384 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f677a36916444bdca446f163b22a0e50 : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/94e808277c114d3fa8f2d343bb889503 : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f7610f028344c59b570596388565788 : SUCCESS in 6m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a2cc270216984f0f8b88f93aec17f272 : SUCCESS in 8m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a8902e23daab41638d8d282bdf9d817b : SUCCESS in 11m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4eca2deb76c547f39b29c3f3c347ac9b : SUCCESS in 4m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6e118bf7190d442c903eba53006a0503 : FAILURE in 8m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/520d72f0292d4632be882fc50a3a6aa9 : FAILURE in 4m 52s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8d7dbd5c4ba84dd38125b79bcad8ea39 : SUCCESS in 11m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9675a696a4cc456089534ac03725ebdb : SUCCESS in 20m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dd0a30c8233f410e9fba42cc8f07c3aa : FAILURE in 5m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e2c20ae4200047f5951a53172bab5ba8 : FAILURE in 12m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0eedd18656d74694b579774b5a13450a : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/207b1d61466d47129c219c3c6cfb3d63 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b90de14ae7f04a87974145c36e3199ff : SUCCESS in 34m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8b2e6f202aea44ffaf462c1dccc48e66 : SUCCESS in 5m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3f2db93643ed4b08a6dd016b91b1761f : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73f4dde93ec24a04ab3f12f32c6dd509 : SUCCESS in 5m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1423dfb96b664c2abc7ce641ad56c954 : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30943bdfa4b94c8f80595e92d6570a11 : SUCCESS in 1h 02m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebab26a7668d4c298e0add03cd0e64b6 : SUCCESS in 1h 36m 38s\n\nWarning:\n  Job cross-nova-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/virt/libvirt/utils.py in the commit\u0027\n  Comments left for invalid file nova/virt/libvirt/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"51d0bb7fc3ca847bc1eed195eb871125e7cca444","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 01:41:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"0e969f1e98d530f19b857dd37c30efdb40a617a6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-03 17:34:49.000000000","message":"Abandoned\n\nhttps://review.opendev.org/q/topic:constraints-tools-updates is a better solution","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de91c8af45b658db7fbe5d8b4581e2fa0fec962c","revisions":{"de91c8af45b658db7fbe5d8b4581e2fa0fec962c":{"kind":"REWORK","_number":1,"created":"2023-11-09 22:27:52.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/64/900564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/900564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/900564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/900564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/900564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/900564/1"}}},"commit":{"parents":[{"commit":"58890139c408aaeb23e51436abf32967947ba766","subject":"Manually generated constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58890139c408aaeb23e51436abf32967947ba766"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-09 22:26:39.000000000","tz":-360},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-11-09 22:27:46.000000000","tz":-360},"subject":"Simplify upper-constraints.txt","message":"Simplify upper-constraints.txt\n\n.... to be more human readable\n\nChange-Id: I3c4966861cf767940b82d9138e51dc0bb4eaa5c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de91c8af45b658db7fbe5d8b4581e2fa0fec962c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de91c8af45b658db7fbe5d8b4581e2fa0fec962c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
