)]}'
{"id":"openstack%2Frequirements~900863","triplet_id":"openstack%2Frequirements~master~I21451769a56318d07e7af16f3898e7a54124d81d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I21451769a56318d07e7af16f3898e7a54124d81d","subject":"update constraint for oslo.i18n to new release 6.2.0","status":"MERGED","created":"2023-11-14 09:52:35.000000000","updated":"2023-12-14 18:47:50.000000000","submitted":"2023-12-14 18:46:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900863-new-release","meta_rev_id":"dd906c7f8d7ae963fb23d1af0054eb316fe20466","_number":900863,"virtual_id_number":900863,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-14 18:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-14 13:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-14 17:24:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-14 17:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 11:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 13:42:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-14 17:24:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-12-14 15:26:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"62911411ae5bb93841c8e4dcadaf826050579e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-14 09:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c081f819b40e9c2e37ae1ee05cac85dd7e797c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/016f41f53c9b44b386465fb8ad286c1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e4e86568c34b9185359110576aa919 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/620f15467d4f41acbcfb7037fb83c467 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25cefe39b4cd48539f408f3f7917c3b1 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3c563d4a3ab44aebeefe2b6a5dda127 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3ac8bacbe9d4d19bbd96c10abd732bb : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2295ed3b1b67497d8df8be40cb5566f2 : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d7a217d7bbac4404b3ed39ad30477e95 : SUCCESS in 7m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3dcf54747cba48c8aa286c870107e64d : SUCCESS in 11m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c417ebe906e844b0ad66388b972ee411 : SUCCESS in 7m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6c72df94967143ff80156938f2841499 : SUCCESS in 13m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ebcdfa59a794455284fda2fe503f4b02 : SUCCESS in 6m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/716e363da57f4652be701f32231b348a : SUCCESS in 4m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e9d461c817494b679e2dc01e44046b9f : SUCCESS in 4m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cf1fac70c2d84663bab373c11540e476 : SUCCESS in 12m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/52ecb1d583b7471b963fcb7a97a01961 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e426e790d5054222b31bc94dd8c8ce7e : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6045120fbcdd4514a4b667780ec298b1 : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ed5eb921a1c448999863292f39c6e13e : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7318c29cc8314f12b6f7bf42229747a6 : SUCCESS in 8m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b14574710b4543b4acbd6ac069148b61 : SUCCESS in 8m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6a0aa00eda854c30b1d46a6939a1c529 : SUCCESS in 12m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/45e59653160e4d95ac41de992fa77ea9 : SUCCESS in 4m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f65e69b9bff44beb989e928c10e170c9 : SUCCESS in 9m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/35f3f90963614ffda154405164a7e479 : FAILURE in 4m 07s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/13740000dfe14735891e2593675281f5 : SUCCESS in 11m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ddc96eb567d74dba8e894569399fe888 : SUCCESS in 28m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0294afefe88d4e47856a5464fbc9221e : SUCCESS in 13m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/23a463aa2b764b2a93af14d7f76e2005 : SUCCESS in 15m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d0bcb310e27143949046c7b2cc7095df : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c8e8affbe6241f3bdaccb083d971c63 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eabf2d1c38a14b1aab33399d6935f33c : SUCCESS in 23m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4d678dfd1cc6495c8106bde7f6a1b5b5 : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d779298235b842e6aa34d8a3b37cf391 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e10491739ed0483b93ee1c23478391c8 : SUCCESS in 7m 22s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/81ad43f7d899426fab65240d04ad08a0 : SUCCESS in 14m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a52628fbea4240fbaa6e61c29867cd9f : SUCCESS in 44m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ff8e6fe82ce417aacfdf84f7167bded : SUCCESS in 1h 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b72b41036b8ca4f1d8f19e79ce416e896c4925b2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-14 13:42:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46e0b59f45c8b9bb5b2b983f6a167c25eddb4a4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-14 17:24:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"477a7f267e16bae25824cd207c92fa5d215f6b17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:28:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0160bd13b66d7d15f7c4544d75acb8bae79a6c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:41:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf196617ef5481aa19ab389327223be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8758fb56da7943b185a06a2beaaf362d : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8981c647c501407da26152775d5c236f : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12727173b9c44663aaf84087f34eb896 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d252c04ef6446bd937edd0b9018471c : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a8151c29fe043cd87b1f91684934e22 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9719eb0a9cc04f0aa16a86e2b03bbdd8 : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c67df333f0a44acb030ff7a896cf352 : SUCCESS in 9m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c8fb54e612684040bede945a284bb35f : SUCCESS in 15m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/718021bfcc8c493789b4b1732b96743a : SUCCESS in 11m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0f9f32ae226f41eaa9377134c8b89668 : SUCCESS in 12m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/38047a0f0c3a44deba3e9190b53d0d82 : SUCCESS in 5m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5fbbf5e60c744eb0b90f9300e31188e4 : SUCCESS in 6m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e64e070bea9742748f88beba7279eda5 : SUCCESS in 8m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8875a40f92af4a8fa200f01a85acd413 : SUCCESS in 13m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cfd1e4b8fe9743579a43dfea419c165d : SUCCESS in 8m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4b3d5731f80d42cc9e1698ee8b0ba5c6 : SUCCESS in 9m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b8905ee2fd2d4034a610768b8f53a3f9 : SUCCESS in 10m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e2081df222254eac8fb2d484c7863db5 : SUCCESS in 4m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b90e6058149945df930aadebc6481f97 : SUCCESS in 10m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/76e7637569ae498490526e0b13907ee2 : SUCCESS in 25m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e7fb97fc3144f20a7d80b8fdaa2c845 : SUCCESS in 12m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/de108533af6e413eab40b73afb9115bd : SUCCESS in 13m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dd3cb6b3889d40a5a129482fe252c806 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a0f2b67653ed4ca58ea588d68614341a : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a91c04416ba7433d8af5d70df0949c89 : SUCCESS in 27m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1aa987f9bead4206b3a3f4caaaafa72a : SUCCESS in 5m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/495ed1ccd14147de8161c26725aa978b : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b19a6a3714504a8d867b37dad93508ec : SUCCESS in 7m 18s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/222af97e91b84fd69764aeb2703f0a7a : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f1076e835f4465b9ed41a600f62342d : SUCCESS in 1h 06m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05579f496d9a4b41a8d04c94dc11d2de : TIMED_OUT in 2h 04m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b456bf68f3da98c613547530e08007953bfd1572","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-14 15:26:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfc700b576441faf99cdafbcf733cc1ca6cafd18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 17:04:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc678a57759486da53b924ea6382a42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04b1c8d8e3d4c4281cb5222375ecfa5 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bdb2a691da941cea66b1245558cb209 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3498dfa5e79432fb8858e83af9cb86d : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/292ecd68455049a189e28877f5f27089 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efac72dd9dc54670b5c2157d14a4863c : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d31da109e6fb4ad59078903d9f36cace : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb95be97374d489ab82725150528a392 : SUCCESS in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/819ebf86eb6c45aebc863a7029194e75 : SUCCESS in 16m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b285c931718143e09ebcd9a3dec20720 : SUCCESS in 9m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b1d6407c3099439f975f8a812568a19e : SUCCESS in 18m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/612627205eff4711a4c7cf508c0b0cc8 : SUCCESS in 5m 13s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/003b56a3a74644f1849c2638b424f94a : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c171aa5efd6449e9a388a9b5060eda9c : SUCCESS in 4m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8e13511d729d4864a8992cada0a690e5 : SUCCESS in 12m 42s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/3faeaa24d125422e89b3f0b636edf07f : SUCCESS in 5m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2e4d3866b03846d288bd786f7e9d7c7f : SUCCESS in 5m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5b31a8c5bfe24edba6c861e540977ca8 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ae6eacdbde5c4758b66bca16622c170c : SUCCESS in 6m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3bd7bf629a5c4dabb45bc3d9f17783c7 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae083159de4348b49fac8feb675ae55d : SUCCESS in 9m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6044db0041804181b42ad0c13a867a92 : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6a2421f04e654b298b84ca5d3600ea1d : SUCCESS in 16m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d120b68b09314424b328a0b60a115cac : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/26bd153a01d94c1f89124ee4daf484a9 : SUCCESS in 10m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5d5b37482dd447aaa553cfa0b682d846 : FAILURE in 4m 24s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f022047b90094769b8313c2708a35f0c : SUCCESS in 13m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/355bb1afb81946588a78327a83ffcbf8 : SUCCESS in 29m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/397c1042937b46b89eb3819f36196514 : SUCCESS in 11m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fd5916be9d3f4da1a5c643b6fd69adc2 : SUCCESS in 15m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e0eac4a999e14082b4cffcf4a1747973 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b06d74ecd7b48788901c986fe3ffec6 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c41930863ded4be883abea3981a5a65b : SUCCESS in 21m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/afaf6862f7db422eb1f19276bca6f4b5 : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d3747213e1d44e4b89c5e86ece2014d9 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c86768564bb4ace92449ae6665bdf30 : SUCCESS in 8m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/67ffcfa11be348358c0da23d1e3ded78 : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/748ac89f60e9433f853d041b442abcfa : SUCCESS in 1h 09m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63588a946829412f884bb87cee32671d : SUCCESS in 1h 36m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ae49473d206702c592c1c176fe04a4502ec5ad0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 17:05:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6929b654a11998e937a8b7b19d516c320bd2772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 18:46:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96a2e9df8eb941caa003c46d0887ee6c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee36c5d1b6ec4fd695f2f11c5e6658e0 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de9b9a626f84fed81485d5944205fe4 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f84f0d4c3244476994bcc9a8ff8f60ab : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/611b37a352c94eb1a735f29e7319cd05 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/778cf1ddafc84625996e9f350f4e456b : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d040298ea73f469a8555663195bd73e1 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0f70f02c2d249d6af96ba26f5f88379 : SUCCESS in 8m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/decc826ff7ad4db99cf6a894b4812d59 : SUCCESS in 14m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ae31207fc0034a6ea9e98ade329a5a6a : SUCCESS in 9m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3e69c6bb126a4fc387583f3f959150a0 : SUCCESS in 12m 34s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/9ea0e09c50b9436381c2dee1f112201b : SUCCESS in 5m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a1f406bf44514f989a4979ecb0640bcd : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0e6f646564eb40ba85f08d1e8572e895 : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/143ee783a1b6430f86547cc559cfd8bc : SUCCESS in 8m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6ab6e142e0c44c4db7c25d61c7e474ae : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/641a0806e4bb4b96ad101d366f84277b : SUCCESS in 6m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b6a9e09d424e45f7b823b6ce6119d28d : SUCCESS in 9m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a3f203f0cf0b4fdca429458e95591075 : SUCCESS in 12m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7da78fdb12ce4e3c8c614fbd239158f3 : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1f86946e6dbf414e9035d317f644fefd : SUCCESS in 9m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/03f12870120f4908aa98c218bd7299a1 : SUCCESS in 23m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/90cb468522984c389f405d20b0a04f6e : SUCCESS in 10m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cee1f9e0f60a4a809e2185c517c58997 : SUCCESS in 13m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8aee61e3e4c74774a58adab7827ff23b : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b96339ad2fd249bd93dc31d72c44843b : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a346b146aac847518b7b250889f6c1f9 : SUCCESS in 20m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ef7cb325bc314306ac1b23aa7e0c407f : SUCCESS in 6m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/40f402de03084b9db478a9cfc5e40934 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a4fba185100640009f8ec435fdfca68c : SUCCESS in 8m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/28e147ce24644639a21925d7e2d506b1 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bfda0e2ef9b4e0aa7be5511751d69df : SUCCESS in 1h 10m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/782c1ff83ccd4b3a8fef7c20e1ee34d3 : SUCCESS in 1h 34m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a810a4e3a59afb3cce560552dd6a1e495d0e4c06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 18:46:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd906c7f8d7ae963fb23d1af0054eb316fe20466","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 18:47:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b84fe32de3ad4e39a6eb603d1c75214a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbfb53741a04657bc31a6d7c66fc40f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0598a713a4ce70c7ebc94ed7526f88d1a2cbcce1","revisions":{"0598a713a4ce70c7ebc94ed7526f88d1a2cbcce1":{"kind":"REWORK","_number":1,"created":"2023-11-14 09:52:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/900863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/900863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/900863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/900863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/900863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/900863/1"}}},"commit":{"parents":[{"commit":"82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66","subject":"Merge \"update constraint for testtools to 2.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-14 09:52:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-14 09:52:34.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 6.2.0","message":"update constraint for oslo.i18n to new release 6.2.0\n\nmeta: version: 6.2.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I83182970614ab32a93dbcb4babec5ffde7fdbcea\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I21451769a56318d07e7af16f3898e7a54124d81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0598a713a4ce70c7ebc94ed7526f88d1a2cbcce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0598a713a4ce70c7ebc94ed7526f88d1a2cbcce1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
