)]}'
{"id":"openstack%2Frequirements~900867","triplet_id":"openstack%2Frequirements~master~I0e7e712d5320603c927a3c12ef20c8d26491eb62","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0e7e712d5320603c927a3c12ef20c8d26491eb62","subject":"update constraint for osc-lib to new release 2.9.0","status":"ABANDONED","created":"2023-11-14 10:12:05.000000000","updated":"2024-01-18 00:20:17.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4be3c886d6008060a64814f0e10cbafda77e55f","_number":900867,"virtual_id_number":900867,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-14 11:54:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 11:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 13:41:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"}],"messages":[{"id":"69c4207aae5bd77a46aa42b9ef8fd5c7a445d83f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-14 10:12:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac072474bf49a1d047c19ebacee95fc49efd0d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:54:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45aa032bacb24029861aedaffde0cc51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0086068e12054863aa9ccbf64babb205 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140382c400dc448c99bc5d2dd7d138a8 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c82e779f695478592d970e310735727 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70a7ed59cc42412a9355bff9d5cfa12e : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb759369d90147a9b7b855958e1e2451 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e83d3dcc952e4619b84fc51324c2e143 : SUCCESS in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cd8fd88c46464382842321045c7e9ef3 : SUCCESS in 7m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6291a83885864d9392756cb7fc22e93e : SUCCESS in 16m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f164f025cbeb4b89b7f8c6d4b58457e5 : SUCCESS in 7m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fdfe8a8791144aa6ba440399b874ab5e : SUCCESS in 16m 42s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4586695bb26c415c8529f6a4caa1029b : SUCCESS in 7m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/febf327a87e141ef9cf8fee312039fe8 : SUCCESS in 4m 41s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d8a7fe560e804e9381bb1719f2ae13d1 : FAILURE in 3m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6a811038c9294688b07964923c66c58a : SUCCESS in 11m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ba21f33202df4d73b98f4f8e1f6b0cc8 : SUCCESS in 5m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1fbb4880853141bc93a504b598f1b2fc : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c38af23f85c34ed89df6913e352f959e : SUCCESS in 8m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f85f8d1666614583ac9866adb274cf6b : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d25919162b144c8184e8dde081b0c64a : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/319187660125469cba7f0603d6ff689d : SUCCESS in 9m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f89a10d7f17141898ae1cab57815c85c : SUCCESS in 13m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e92f464b61c041df87e139c3e0925479 : SUCCESS in 4m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4d6dd5908fa84d69b4fc885684b2fa80 : SUCCESS in 10m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/38859ba4531d4d94970d8d58e02e8363 : FAILURE in 4m 12s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6fcbddecfe084a78b1d7365485216693 : SUCCESS in 10m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cc4a1249728d4a72af25dd8fb8e32124 : SUCCESS in 22m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b4bb1878c0254941954d2153e1f444aa : SUCCESS in 11m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e67d2a5f20c5464f9dab69432c4eb921 : SUCCESS in 15m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dfed688d3f9a483dad7136f0e9eb37e6 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e33b5493edc486e9d45328f4bcbe3b1 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd820a973b764e4bafc9adf1aee9b0dd : SUCCESS in 22m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/19df82b453a7457cbae70756526f65c4 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/93d2f844794e4971975621b30826a4e1 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b68dc4b2f43243b6a3c8cabe5d28e461 : SUCCESS in 6m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/47b1750ef18b46529861178f143cda6a : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a59e12000c6c4178b1613d212fd40f60 : SUCCESS in 1h 00m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75d989fc03b84e33977e1b47bcd5ff37 : SUCCESS in 1h 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1332faa01c0e155ce3bddf04abd6dc5edb16913a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-14 13:41:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4be3c886d6008060a64814f0e10cbafda77e55f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 00:20:17.000000000","message":"Abandoned\n\nNo longer needed: https://opendev.org/openstack/requirements/src/branch/master/upper-constraints.txt#L403","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83988ca67d0b395a83c2959e1439165a18da883a","revisions":{"83988ca67d0b395a83c2959e1439165a18da883a":{"kind":"REWORK","_number":1,"created":"2023-11-14 10:12:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/900867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/900867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/900867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/900867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/900867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/900867/1"}}},"commit":{"parents":[{"commit":"82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66","subject":"Merge \"update constraint for testtools to 2.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82b8bd3c25fc0ddea8d92e7cf99763bdad49cc66"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-14 10:12:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-14 10:12:04.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.9.0","message":"update constraint for osc-lib to new release 2.9.0\n\nmeta: version: 2.9.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib09d6178a13e66fe0cb78454b6afedf1184ec9ba\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: likui \u003clikui@yovole.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I0e7e712d5320603c927a3c12ef20c8d26491eb62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83988ca67d0b395a83c2959e1439165a18da883a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83988ca67d0b395a83c2959e1439165a18da883a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
