)]}'
{"id":"openstack%2Frequirements~901013","triplet_id":"openstack%2Frequirements~master~If86b98f6249d8aa8c17bdf91e496ac5981d8a499","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2023-11-15 20:58:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If86b98f6249d8aa8c17bdf91e496ac5981d8a499","subject":"update constraint for python-masakariclient to new release 8.3.0","status":"MERGED","created":"2023-11-15 11:09:53.000000000","updated":"2023-11-15 20:59:53.000000000","submitted":"2023-11-15 20:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901013-new-release","meta_rev_id":"0862ae5aadfa1a1513e398c7a0d5a75d201e71f4","_number":901013,"virtual_id_number":901013,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 20:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-15 14:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-15 19:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-15 19:22:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 12:59:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-15 13:54:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-15 13:54:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"state":"REVIEWER"},{"updated":"2023-11-15 14:08:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-15 19:22:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e6c85f33aa52659e3a6515b56b2df7a75853bd53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-15 11:09:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ecd69c339a1e1a8e6d07a21d3c5aa6a94bb19b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 12:59:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ce9a412d884bca919cc59b81329001\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b1e8bbd07b0499ea8da827ec9fa3dd7 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d0c6de1a5e84c2eaf676da2a2d47457 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b5d0cb1327947188538f01556aa5ad0 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77155658aee44d4dbd3edaef8f061457 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07538e26f1bd4d92b7cc41798d0a6b76 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/657299edf2794849a425bc24f7622d8b : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4528ea9639345a98d64e3398255d42a : SUCCESS in 10m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5d1b1406dec3481295fe59263da774a0 : SUCCESS in 11m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/358937a9674f4305be3e29b87b2e50f6 : SUCCESS in 9m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/af5ab61a296a47ddace32239469da621 : SUCCESS in 14m 07s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bf3dc04441ee41b0a9c28621390b20bb : SUCCESS in 7m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5fa19f8d0e554d179e2aab6a9f479b4b : SUCCESS in 5m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f17b93c6208c4d28a78d34899a791b28 : SUCCESS in 4m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/05d51c8d1c5d44c99d77e53603c278cf : SUCCESS in 11m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/af1ee09662cb494c95d05a7208d162b5 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/56ef276cc73b40e58a53b5618bcea180 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/14bef54f53094e3e96b6e79e0f41d9c1 : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/30049acabdd14f67ad0a40c6e1f6e601 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5ef5c287ba74f81b425cc42db160044 : SUCCESS in 8m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d8235c2822084b8eb1f8783513689393 : SUCCESS in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a622d27547184a65be2f207eca8e0687 : SUCCESS in 11m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/760423a2ea67411c8929b0a0433b2689 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/eca381cf34b2465282d8b55c7f466cab : SUCCESS in 12m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8aef49bbf7b24ef8a8f00cfbe4483d8d : FAILURE in 4m 16s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b4a9cfac63fb4708915f2d1446ded7dc : SUCCESS in 13m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0f8a021e7d1344baba6ae99458632ca9 : SUCCESS in 24m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b591bbd45154d73a12ff4c369da7791 : SUCCESS in 13m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3e537010fd514e9096d0ed1c4de0d7b8 : SUCCESS in 13m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/39481e7653e045a3818aeb17d1286406 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5edb472b3ef648d98728e1103d246c61 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db46d3532c8040d3aad52def8cb0171b : SUCCESS in 23m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f9018cfbe1084a82a33dbb650f13dd80 : SUCCESS in 6m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4787c63e46fb4a68858291e1d12746a4 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb23e7660c8249489b2652177c136b2e : SUCCESS in 7m 10s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a067c63b5d6c4f979ec3ab668f7604c1 : SUCCESS in 14m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cac7b1da956d44f2a9654e8ca85232f2 : SUCCESS in 1h 09m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db832c8d59c84b86b364d673fc18fa6a : SUCCESS in 1h 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b7b38f7d3a44fef32621069402ae3c5ff71d97ce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-15 13:54:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28ebe9810b8604faf52afb1ee9d32b4dd9deaf0a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-15 14:08:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9afb288622fde69fff78eecd33f9860a84bd6330","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-15 14:09:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b72583afb9cb6b0e058b8b9f628769d1e0cf986","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-15 19:22:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e33954c6b8ebb7c95d2540f61c8e4df20edf8fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 19:23:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e023f02533cf3012642c75aa62d04d36387376b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 20:58:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a92b565a789f47a886bf4919ed756d0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66846fbb09248cb8e073508fb23cb52 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81090b89f6e459a9bbe82ffdca74ed9 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33ec7314d47b4f4cbb072202482484aa : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7939b4ed2c284a199c6362baa012ce5e : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/075c5a38872849e6afadc496343fe23c : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4825c33af4bc429f9d543fdf2a0fec1a : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6736338d98e646f0b6c758e65c448dc0 : SUCCESS in 8m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4654666086c14468b59123705b7cc913 : SUCCESS in 16m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c8283e74aa344a9db513544f534e0d61 : SUCCESS in 9m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/955bacafea934ef1a44471813605e0ba : SUCCESS in 11m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f05e4a9b34754215bf8df7eaee8b015b : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c8bbb76923a74ce2bddce3e7f632f5a1 : SUCCESS in 6m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/117b74ebb1274545acaadaaf289b4adf : SUCCESS in 9m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9d7e5d5141a0457985d7e6e085dd5566 : SUCCESS in 13m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/89945bd07c994ed7a661a7ab4a04905f : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b891523c67234d5ab0a20258e2676af2 : SUCCESS in 8m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4c78787c8fc542f983b18f9620167a77 : SUCCESS in 13m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a18534b3c13046ddba5eb420a7aa3d75 : SUCCESS in 3m 54s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/af869d260fb7411c9de185ebc1870894 : SUCCESS in 16m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/35acaa170e294c20aaed0806020ecf58 : SUCCESS in 29m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/909da87b91b6420cb49317a99b93a943 : SUCCESS in 11m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/931882debca74efa9a92d5db79f2045f : SUCCESS in 14m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1077c77b591546e4880febb1a8b0ae6a : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c952fb9ddb5741d6bac5fd7657e413ce : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96332e426de24b6f8893b9036896e796 : SUCCESS in 23m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1502da2348eb4dc78ca51e1fee02bc3b : SUCCESS in 6m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6bef98fc44644ca39854200196b316f9 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/619a8114ee2f453eae134f664d1eb18f : SUCCESS in 7m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d1440d8265ea4979a0a835e194ae94fc : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31725afcf0ed4e13a6810294124871b3 : SUCCESS in 57m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78f445884f894dffa59efcb6f5d5dc9e : SUCCESS in 1h 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4fd0bdbd25b80fdadf167d023c779220a1a4e53a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 20:58:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0862ae5aadfa1a1513e398c7a0d5a75d201e71f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 20:59:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26700721c1be4b02b6e08565ed23cf55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268ce96761b04a8b9291afddcf6af4cf : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86a97fd8f555e734f6114929fe37ed4d9fe60e31","revisions":{"86a97fd8f555e734f6114929fe37ed4d9fe60e31":{"kind":"REWORK","_number":1,"created":"2023-11-15 11:09:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/901013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/901013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/901013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/901013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/901013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/901013/1"}}},"commit":{"parents":[{"commit":"6470ff2f49451569126e77e5e02a0462ec87d0f5","subject":"Merge \"update constraint for cliff to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6470ff2f49451569126e77e5e02a0462ec87d0f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-15 11:09:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-15 11:09:51.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 8.3.0","message":"update constraint for python-masakariclient to new release 8.3.0\n\nmeta: version: 8.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I194769ce10436d2eca63e01b7d1ebfd3eece99b0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: suzhengwei \u003csuzhengwei@inspur.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If86b98f6249d8aa8c17bdf91e496ac5981d8a499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86a97fd8f555e734f6114929fe37ed4d9fe60e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86a97fd8f555e734f6114929fe37ed4d9fe60e31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
