)]}'
{"id":"openstack%2Frequirements~901119","triplet_id":"openstack%2Frequirements~master~I2bd5671962b3ca1abac91c2afba6e2c566616390","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-26 18:06:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2bd5671962b3ca1abac91c2afba6e2c566616390","subject":"Allow locally requiring a package for specific python versions","status":"MERGED","created":"2023-11-16 08:25:27.000000000","updated":"2024-01-26 18:07:56.000000000","submitted":"2024-01-26 18:06:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901119","meta_rev_id":"a7012b93bc74620b8047398b18939c86cee44bd7","_number":901119,"virtual_id_number":901119,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-26 18:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-26 16:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-22 03:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-26 16:39:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 10:18:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 03:06:33.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-26 16:39:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"eb368d9e0a01c711c366023b06bed2cf59df67e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 08:25:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac3cabb0f3f64408f3de222989333fc38c6aabb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 10:18:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/674c90ba0f1b4d7c8d6027761792ef60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591f815a12e0450a97187576d9508df5 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d96d7a52317045b6a0f417c688cccc95 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8d1faa042c1408c82f62dd70d8443a9 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c7684b4d03549ceb56281034c7bfefe : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c1a430bd8f142afbfcebc43c876b4c1 : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/841a6d7aad3c4993a50340ae661b702b : SUCCESS in 1h 00m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49a01b61548f48cd950ac6c34b59e03f : SUCCESS in 1h 45m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4306e581dea3b9e02a9ca9bcefbed7f9dc03a7fb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-22 03:06:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d80a27b3871dcd142af5886efb6f25caa2e7eb02","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-26 16:39:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f661acd2b84e17f3d27ad5a97f6716b522f25c99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 16:40:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca0245de26380d56aff78c930b52ec263d5de668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 18:06:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08483303cd874c0f873ac94aa21c9f29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd24135dd8340f3b7bb0d5dbe5f37f0 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2fdab09b074a5abe24640ce6257f72 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddcd5156205b41549809ecfd0b0e9d27 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbffeb62ec7f45cdb53f4f65ad5b6a38 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d2debb61d574382879213fdff63b379 : SUCCESS in 3m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa52574997dc49bbb0e941dd0a0ca229 : SUCCESS in 29m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/871dc6715a25481d8f6084f20ba397d2 : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8e3ad070dbda981a28bd7afeb3eb2cd1a455feeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 18:06:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7012b93bc74620b8047398b18939c86cee44bd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 18:07:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59a26a44cc774da5898a5d7819fce925\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb925eb815bc4b748e9edca641240ea2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28af2ea15706bd7aee7e7dc810d70248eb65c09f","revisions":{"28af2ea15706bd7aee7e7dc810d70248eb65c09f":{"kind":"REWORK","_number":1,"created":"2023-11-16 08:25:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/901119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/901119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/901119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/901119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/901119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/901119/1"}}},"commit":{"parents":[{"commit":"6470ff2f49451569126e77e5e02a0462ec87d0f5","subject":"Merge \"update constraint for cliff to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6470ff2f49451569126e77e5e02a0462ec87d0f5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 08:19:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 08:19:25.000000000","tz":540},"subject":"Allow locally requiring a package for specific python versions","message":"Allow locally requiring a package for specific python versions\n\nSometimes we have to require a specific library in certain python\nversions only.\n\nIn oslo and a few other projects, we are now replacing pytz by built-in\nzoneinfo[1] to reduce 3rd party dependency, but zoneinfo is not\navailable in python \u003c 3.9 and pytz is still required in python \u003c 3.9.\nIn addition we need tzdata when zoneinfo is used, which means we need\ntzdata only in python \u003e\u003d 3.9.\n\nHowever these updates are currently prohibited by requirement check\nbecause pytz in the global requirement does not contain python version\nmarker.\n\nThis change allows having a local requirement with python marker along\nwith the corresponding global requirement without python marker to\nsupport such usage.\n\n[1] https://review.opendev.org/c/openstack/oslo.utils/+/897623\n\nChange-Id: I2bd5671962b3ca1abac91c2afba6e2c566616390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28af2ea15706bd7aee7e7dc810d70248eb65c09f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28af2ea15706bd7aee7e7dc810d70248eb65c09f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
