)]}'
{"id":"openstack%2Frequirements~901128","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I127e76dea281c3207f97d56b67156d68354b7eca","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I127e76dea281c3207f97d56b67156d68354b7eca","subject":"update constraint for oslo.messaging to new release 14.2.3","status":"ABANDONED","created":"2023-11-16 10:58:38.000000000","updated":"2024-02-13 05:54:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7eae7b807c0fd8aea59ad79243d3902899a76ac2","_number":901128,"virtual_id_number":901128,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-16 12:35:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 11:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-16 12:35:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 21:32:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"202dc779f994ac01207139d9c333584dc5640707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-16 10:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84eff02968eb279794029a84bfa617c7c7d651a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 11:04:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d659c72825794d5496c16e9c24c70749\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/954a724ea09b4790a02466028e78c968 : FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"54561626d72c11eefd3d69e385927df8e88aafb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 12:35:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04455d6093094747b3a0c170a884cf81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c40ce5e6f748f191517d10b0d2785f : FAILURE in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa5c9fb1c4434475a9f2265d55639db9 : FAILURE in 2m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9695476b9204b2c85e26c460eea8a4d : FAILURE in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/700142066d084b3dbb92b12c75ee5d45 : FAILURE in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d46f787f7f41425aab11eb6abe4d8634 : FAILURE in 16m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8a69b14f04ed435d9453aa543ef338f4 : FAILURE in 12m 18s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4506d17708e549208f894b7ba6d74630 : SUCCESS in 8m 13s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8abdf071bb3743f3920f86c387c33b0a : SUCCESS in 5m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e00890f9f5324d67991e7dd3c17a1cb3 : SUCCESS in 6m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a7e59df814764c32bfaccdddc69b6269 : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c936ed653c9e4ce0bbe6660c2ceecbee : SUCCESS in 5m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/888db4677e564fd8bb03c1ea6f460500 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/47d207b3344944479ec7580f4f18f245 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/765b31a07d33447bbe8704b175c214ed : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f8156b4bc8914ab8bff665b0a74d7bd5 : SUCCESS in 7m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/64257b7a66be4c2aa369697e0f0feb15 : SUCCESS in 9m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/78955bc96ba94c35942d2a086a6a7f6b : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f30e39b0fe9a4f89be40728a25d3e153 : SUCCESS in 3m 55s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/301ca09b53804e3f95ef1fb1dd85af49 : SUCCESS in 8m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bf30bf886c3b454f9e94e7a35d8381ca : SUCCESS in 5m 03s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c33e6acfaa0d4a4e915eee3e500f7525 : SUCCESS in 9m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f386c1d2cbf44a579b9d2d58f02fcf5c : SUCCESS in 20m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2972b6e43fa14a9d85a7eefc56e3277d : SUCCESS in 14m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c36ce8a95aca425cb4e9d7ee86320211 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d927e49c0aa84a84a4a1d8b288ab59ae : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df27c335a52e49c6ab2443dd24d982bc : SUCCESS in 21m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/997cd9d93b0e47118856dfed846617ee : SUCCESS in 6m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4f5b66e5a50645c3b4b2ac3e778f4752 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/707e21d323fc4d428f192eb52d275af9 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/25e4451384484c63a8624ccc9ecc32b3 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c56402349374489ea2e2b260725284e1 : SUCCESS in 59m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/966cc98726244aa9b802421c53294739 : SUCCESS in 1h 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e7f20ab268af13a8fdb57e7eb80401815d47cf93","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 21:32:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff422def9b0d150f61530246034f8e7b0ae1494e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:45:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c7b7fd10d0458c89606fab17eddb07\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9a69b4115e4a4fb2a95c4f745b5683a2 : FAILURE in 11m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b8a2bcc311a313cca30146f115f710608878eb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:34:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/006d46d271794a2d8f25b50c9698a241\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4128c72066419695103a3085d4d60e : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c49c6720b4e4325b36ea9e992e0b0e2 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f97d043e6764ba5a6d82b7b89e09a88 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4a1e452422548288dfb0dc733c4ea8e : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02ee86c7fbdf403190ec0fb6bcf3272f : SUCCESS in 9m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9b9eb15f1ff147008511051ea5f62794 : SUCCESS in 13m 46s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ddcf12862797427fba7b226c0b20450f : SUCCESS in 7m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8445849b979c4cf08f97bb449538db83 : SUCCESS in 5m 23s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/98c721da7c4d42de8769ebabbacc09c5 : SUCCESS in 5m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bc8222be60f74263b5b544171076a1f8 : SUCCESS in 10m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9911265e6c9642f08d7ac9983a8d5523 : SUCCESS in 5m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/37419e6738dc46738f4bd44115ecb106 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/212c71f87bb74ab89cffae3ffa8e7e72 : SUCCESS in 8m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c8cd7b8019f84c44be1431b604264846 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f333355a7a8946c9bd7857544c917a54 : SUCCESS in 7m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6ec937b44e674225864ab6f4cfcb2d4b : FAILURE in 9m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1a1c4707c3bb496396a32bc53cf25c89 : SUCCESS in 12m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/caad0b23892e4db7927a7aa86c865343 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0d3923edf1f1454e9df4cf5ce25e884e : SUCCESS in 9m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bd82df48e7654ca08b8c6661009c1c44 : SUCCESS in 4m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ac977bfa142546eab33c9de3fd1f8713 : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/36a77a844dbd4441bee028440ff9e08d : SUCCESS in 22m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/580621d11f934889a460df0346e06a85 : SUCCESS in 13m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ec42a1ff6ce842cd82295279fd3bca6c : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e6a2e65ee614b22bfd91af851364165 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/843f6da7744748f39b3ec11038172abc : SUCCESS in 21m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/08bc5c553aac4ceda03ceb4ee0db1269 : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/14d4bf2b26594556a0a8ae8275015c15 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da7788d497074b1d8ece5211313313a8 : SUCCESS in 5m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d7a6e12f127e40aeb7adf563a89568f1 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb31c352b9f4465e922191b120c16d65 : SUCCESS in 1h 04m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/414a87c610bb4a1aa4e6926a34fa319a : SUCCESS in 1h 53m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7eae7b807c0fd8aea59ad79243d3902899a76ac2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-13 05:54:24.000000000","message":"Abandoned\n\nNo longer needed: https://review.opendev.org/c/openstack/requirements/+/907414","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f406e28bff78d9e5e920724909bac63d04ef070","revisions":{"4f406e28bff78d9e5e920724909bac63d04ef070":{"kind":"REWORK","_number":1,"created":"2023-11-16 10:58:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/901128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/901128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/901128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/901128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/901128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/901128/1"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 10:58:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 10:58:37.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.2.3","message":"update constraint for oslo.messaging to new release 14.2.3\n\nmeta: version: 14.2.3\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I56591b172dc3518dd4de6a794ef28fc41bed3013\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I127e76dea281c3207f97d56b67156d68354b7eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f406e28bff78d9e5e920724909bac63d04ef070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f406e28bff78d9e5e920724909bac63d04ef070"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
