)]}'
{"id":"openstack%2Frequirements~901129","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I42f282cc77ce064092c22f4c3c9316653c318fa2","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I42f282cc77ce064092c22f4c3c9316653c318fa2","subject":"update constraint for oslo.messaging to new release 12.13.3","status":"MERGED","created":"2023-11-16 10:59:36.000000000","updated":"2023-11-16 21:03:42.000000000","submitted":"2023-11-16 21:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901129-new-release","meta_rev_id":"b9a17cfe1e24647773be2bfc728b96d6651dd6ee","_number":901129,"virtual_id_number":901129,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 21:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-16 13:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-16 19:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-16 19:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 11:17:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-16 12:29:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 13:11:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-16 19:32:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c686e381d09d53c985e305bc2d2c37cbde727b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-16 10:59:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f930849fbd40b05492056e5068dae98ac26b4a9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 11:17:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f8c2b46d3c44140bc89f8cfecc34838\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4e9a95083dbe4ac4a34181d0063527ce : FAILURE in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f30f4956e1f8dae3e0439fcce1839dbe36debb04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 12:29:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b64a44c6a804cca91a38b56474db3e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a6fbf07307484281aaa5b3dee0be1f : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f121971c7a442879b5e669877f7cacd : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b2aba3ff623431c8667924b8f282784 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c15afab5e10642458f06fd6f303ad6b6 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f1155da7e854a259d7485b76906eacc : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dfaaa96b62bc4c4fab0a748981dc6b78 : SUCCESS in 19m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d646bdfc814c4f57a71f725b33f5a05e : SUCCESS in 18m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9e4eb6551e0549038950970c2e3c8be4 : SUCCESS in 6m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/05d8193b9bef401e92bb4a5d05f66e6d : SUCCESS in 10m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e85b081741fd4892b8697cece46221af : SUCCESS in 3m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/18a00a8c83704076910e56e7c5ee22e6 : SUCCESS in 5m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7d72dd960d0545f7b5c604c836e85e6a : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6433fc8dbe884bbd988c99a55e1e9def : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/474cae1dc7c14899a7f7dc3b6affcb59 : SUCCESS in 8m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7c809671ed384c45b446233e14fb5dfd : SUCCESS in 15m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c3ecb89757214ab3ab06f62bdfdec973 : SUCCESS in 3m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/019a16ef41ce4c3dbd03d30abe12f082 : SUCCESS in 9m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1ab9f12b0b084055a82f2717ec8d1a0e : SUCCESS in 4m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/403d17424d454408bd5b947cfeeebdaa : SUCCESS in 27m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/87c33739a73740388e0aeb6ae83e61d2 : SUCCESS in 11m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e51d8f341d6c4095a940f2e11b20b363 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ef1be4feaf6427fbcedae3feecf172a : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/261cfaf82a824b4aac8bf8cdb658b6cf : SUCCESS in 22m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7c8ed2f979e04b0b944b3d178d9bff95 : SUCCESS in 5m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/81739175bebd474282437098e595adcd : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be108be146344c1497095f4239505c70 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e47aa309cda047ceacbbcf03a49605f3 : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa1f17c94c6349fb9b182bf9ac8161e5 : SUCCESS in 53m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6dfc03650bb4771afde11696476c61f : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":1},{"id":"74df9a05b671d36b98d7234a806b8d627d9d680c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-16 13:11:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5afb6495efecea07734c334f92bffec16e9ecb1a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-16 19:32:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63b2bb511b232d6dcff15a4026a1a4685cf8a90a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:35:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7cf3f4db476f59caa818fc09f0e3565c9e119272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:03:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62f562d5d0eb4a7a80fa9feea504c368\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5fbcf2ab5964a10a9e6a8bb7ca6fe28 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8437dcec8e4668b0dddb5fdb5ca07c : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe3f48dd1962499bacf03346b4116ef1 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5473f4dae4c4560b8c3ce5f6e9c86f3 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c77ed8aa5d7f4992bd21a9ba955fb32d : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00f61683bacf4b7085366b4c9492f028 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c420d5d4b49d476ba28d27892f217c81 : SUCCESS in 10m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6eb420dda17a4a7e85eadf1e159b1303 : SUCCESS in 12m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b41605408bbd4764bf95539a9dbaa81c : SUCCESS in 5m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9ccc3d0aaded4db39e95b0d77499cbc2 : SUCCESS in 5m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/898e39030b714422ae9273fa104e9ab8 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08f2c572663a462bb7a075a5251a1534 : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ce1093edc8cc4ca6b6fa8349a2095826 : SUCCESS in 8m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/eedde843a50c4de09f611e14f998308e : SUCCESS in 12m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/96c871be788f4add869b3956017493fa : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7d9911675b894ddbb3612f3f08edc3f7 : SUCCESS in 10m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/86f09a4e08974bf981d31a90eba48c62 : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f283d7dcf4d4458d815128b6a3f9364b : SUCCESS in 27m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/de5e75983fcd4d3da0bd940af61218f0 : SUCCESS in 12m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d906eb8bdb0e48fd99e095f86f49b371 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a646224f0ef942b5a220aa53f634657f : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96bfe0085ee446a5aca9861209f0d10f : SUCCESS in 20m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8bc19e51c733497a875420799b6f5c8e : SUCCESS in 5m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6a4620bc298c444f94aeec3c97aa9c47 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/731d502ce460426e9e61aa2ff9ac4c20 : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c0f04595454744f8b637d6e08dd7a9aa : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5fbff0306274ae68b7db51be73e3ee0 : SUCCESS in 42m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d62746430d8f47c994d9899ef8ec191e : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b9a17cfe1e24647773be2bfc728b96d6651dd6ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:03:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a08cef69adf7250f8408ecf3d06058b31667018f","revisions":{"a08cef69adf7250f8408ecf3d06058b31667018f":{"kind":"REWORK","_number":1,"created":"2023-11-16 10:59:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/901129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/901129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/901129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/901129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/901129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/901129/1"}}},"commit":{"parents":[{"commit":"65d13a079dbadb4b8cff91264b2e779495799bad","subject":"Merge \"update constraint for oslo.messaging to new release 12.13.2\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65d13a079dbadb4b8cff91264b2e779495799bad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 10:59:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 10:59:35.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.13.3","message":"update constraint for oslo.messaging to new release 12.13.3\n\nmeta: version: 12.13.3\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib6849783af656add6370b8ca736be1723139488a\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I42f282cc77ce064092c22f4c3c9316653c318fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a08cef69adf7250f8408ecf3d06058b31667018f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a08cef69adf7250f8408ecf3d06058b31667018f"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
