)]}'
{"id":"openstack%2Frequirements~901130","triplet_id":"openstack%2Frequirements~master~I04663b9c05306118bbef1d5d6c49257d4fd33030","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I04663b9c05306118bbef1d5d6c49257d4fd33030","subject":"update constraint for etcd3gw to new release 2.2.0","status":"MERGED","created":"2023-11-16 11:02:58.000000000","updated":"2023-11-16 21:03:45.000000000","submitted":"2023-11-16 21:03:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901130-new-release","meta_rev_id":"ab94d228f40ee5e41f94926ed22a70c58457cb0b","_number":901130,"virtual_id_number":901130,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 21:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-16 13:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-16 19:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-16 19:32:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 12:39:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 13:10:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-16 19:32:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e6fec9950c9a878a6aafa111bd973441c15cf4d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-16 11:02:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99e508d43dbbccbb76e5eb2cab8733de559d1409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 12:39:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70d664948c9347949205813b4534d9ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11acc5920c43414cb7778c66b35e25e0 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16357ccb8214a568f5d4301fd834663 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94f32b4eb9a245f7a77d845f8d305285 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b78b896bb844c7dbe4343b6664649ad : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9258d4b050c4a9a8773b74c882cda60 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dda730fce2764f89a4daac21a0417d4d : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4049cff205dd4440a1ed3587b931335b : SUCCESS in 8m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bfe1b75859f54cdb88ba841486535480 : SUCCESS in 28m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a2215a24a23c471187331b51a826c521 : FAILURE in 4m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/acab71deb2a14dc182e7f007421b6fdd : FAILURE in 4m 15s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f54e0ab3b0494ab8a4844c02b1dd3d40 : SUCCESS in 8m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2a757e801c7040dab5d35ffd87e2dc9b : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/31d77650846a4995821e109de449fbd7 : SUCCESS in 4m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8199d531ab9b461d83c97079607e546c : SUCCESS in 12m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/36ea9ff16be84c68b63c44ce28e81ff3 : SUCCESS in 5m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/53b49c4effe64627b5e3da102d2d5635 : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7ae9d8b4aa194c5d812f42f665c07a5f : SUCCESS in 8m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9edd32dfed37436489811d5737c609b0 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51af5f37875b4213bf48b882a8d8376e : SUCCESS in 8m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2b130f90ef4848478afa3eef1eb88331 : SUCCESS in 8m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1c16e38ff03643128844930ef738822e : SUCCESS in 12m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7366975ab3dd4882bfda297a3a57ac54 : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d057c1fc1156432bb33cc9d793aab9ef : SUCCESS in 9m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bb5e5b1489e24d45aa09449f44f2ef84 : FAILURE in 4m 15s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6bb0e8ddf10b463592939164fb4c8904 : SUCCESS in 11m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/485830cd85f84ab1a7de1f14554e9ec5 : SUCCESS in 31m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/920dea8f10184d04acacc63c7f2b1678 : SUCCESS in 10m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4728e6320eb4416a916414c26c9c320f : SUCCESS in 13m 30s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7cf950a9746340ce8a55b2d036877917 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5883ef110b024d0099e1718a0ae92147 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ed900d1056f4905b8feea7ab9b8bf38 : SUCCESS in 25m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3dcd5257bb6742ae8a12fb0468263f7c : SUCCESS in 5m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2a5ff56fd6a04fb6a47b9ec630f34363 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e603fd044bb747ceb05752966e0d63cb : SUCCESS in 6m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2063da29a8084c61b7be1ade80e00236 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/330728bdfa864815bd3bdfc6b72eb4d2 : SUCCESS in 1h 05m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2ae7dfb2f1346758ae482aac8409873 : SUCCESS in 1h 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1c3ef9d49796d66a6ab1a268ef6739f3aecf33d3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-16 13:10:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d116d0afc0fe304bbb85694ccb3d05ed3d51682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 14:49:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c41bff68ee44e669c95109d0ab85d62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b181b2ae34c467d8f5ba708396379f9 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb043a49548b419d8eac5562ed4f92bd : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afd038352e9f4a87a023e18c4c38d59e : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f6c8a8ca1174fe69a1d383a81663ac0 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/556edf1824a04980b9a7820692aebde3 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1df8914e65b140058ec8c509cbb4089c : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b548107b76ad4877a96945e2361fdda2 : SUCCESS in 14m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ad836a3191124f16a89341a673ff284c : SUCCESS in 15m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e0893762e7fe4cbd9d2483744900d736 : SUCCESS in 8m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b1e372826bf64cc4bd884a59ea0e161b : SUCCESS in 16m 09s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5e9bccace43b46349711170ba6de64ab : SUCCESS in 6m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6fc47f6581a84ab9958dcc0833466f42 : SUCCESS in 4m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f8a5488d9f284eb3bda039e2cc522c4f : SUCCESS in 4m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2cf931b9d6cf4dd38da313ea9edcaae1 : SUCCESS in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/045dfdd85169443da7c21852ca4ae4c7 : SUCCESS in 4m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9458b8104ef742f5a022dcd7db30cb08 : SUCCESS in 4m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/45712b1db8084937b76b585b1892c206 : SUCCESS in 6m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ae75e8de6d524a48b3b3c0ca326ad072 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1de392029e841d3ac10d6f3ea642ba1 : SUCCESS in 7m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/89ab8040dcda4158803dbf434e131a01 : SUCCESS in 7m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6d09288973064f16a4176d5e8c3eba4e : SUCCESS in 11m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f762f3b278b146c18f92d7c0a8e0aec6 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8dcf9ff031fd41b195ae9126187c3450 : SUCCESS in 12m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/195ab03862544957838551d33be8a570 : FAILURE in 3m 43s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7aa4cebd21a74dac91648f2bdbef6bfc : SUCCESS in 13m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5bab5657fd8548829d9acef7b14f5639 : SUCCESS in 25m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e162fc1bff34d14b7c938ca8dcf3161 : SUCCESS in 10m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a27b7528ad764264910590e6d413db0b : SUCCESS in 12m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a3d4926d7fe64f5dac23fb8c0ba97d98 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/524752e5d74c49a48387f33caa48dcbb : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59fa17505c5f446c82f5f88c21f2d857 : SUCCESS in 23m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3f67a927108944a9867f43e7b289d69b : SUCCESS in 4m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cbb0185e5fbe448e8c011126eadf5173 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23c72b4bfbe84aac8430417c29a1ca24 : SUCCESS in 4m 55s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7a04ad1fccb046b18cdcfa69d7a698b1 : SUCCESS in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8743d9046d1041d9a438f7da3063664c : SUCCESS in 1h 04m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55d15d2709f2402b95f999061e598f9f : SUCCESS in 1h 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8da48ebdd4331f320692bb40411d2248f5703af1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-16 19:32:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8150fc4d2da26ab12978af6c86c90519395e543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:35:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbc8f4a38f49a99f250763d9fb70ae1a3f7c98c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:03:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5578861f65234b669f289063ebd218f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7030d55e8b8f4181bad5cdbe8d1c8c7b : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac8ff74cc2e4444b0ef431238918651 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d842510db670419b9c9c50458907d82d : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d78c787a1dcc42ca8e1a204dcfb67cbd : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e156f385d3b24a398c56208fc71b96b3 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4f99bb7ea9648dfb59ef279cca5c54f : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fde0c782225b48edbf89ac7c52117193 : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17e28ad867ec4c0bae8e72360b3dbfa6 : SUCCESS in 12m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/af022eb2a48b4a358034ce47063c42fc : SUCCESS in 8m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/013402cbd0b1417c80e206643ee67d5b : SUCCESS in 13m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/91516c4586774f0abf79f22add533205 : SUCCESS in 5m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3489f5bdde094234971bc7a1c5fb489d : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1ac3e784572847cb9cdf41ed88ee99bd : SUCCESS in 9m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/37e7e41662fc4f499789bda4f571254a : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83ea7c7f4cad47d2a10507aa12b35a21 : SUCCESS in 7m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/24fcbae4ec8e4bebb5a187379b7934cd : SUCCESS in 7m 47s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1c3b4f61516d4db288198c0dcf905ba8 : SUCCESS in 12m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1928238832724cec9634c216fc75e244 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a01d7471b28b4874808bcf2841a4fe03 : SUCCESS in 10m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1fcc438863be4df180b3f053a138c310 : SUCCESS in 25m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3ca52216de4b496f944a0ebc2c52ee97 : SUCCESS in 11m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/75e9bf5e72384ee1a17408af98747c43 : SUCCESS in 15m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/153ddc30484748bea43c45ed32e8250f : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac3a0c71f442495382852da34382d142 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/829e12fc09f74ceebd573ffb7d20f49c : SUCCESS in 23m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/75f2307b8d2446c58e5d4a9678799f59 : SUCCESS in 7m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/55334155a3394fe4b07f8a2707e36a5a : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0aed432249414e648337a2e3428029bc : SUCCESS in 7m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6cd78e7237d1490fb7b950ecedd43448 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3037c1225e04eab87ea851a2acfe0dd : SUCCESS in 1h 03m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d429fec6e9fc4fd19be2474e7d1e2545 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ab94d228f40ee5e41f94926ed22a70c58457cb0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:03:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58f96c7f10e149091fb8df8500469062a9bd63ce","revisions":{"58f96c7f10e149091fb8df8500469062a9bd63ce":{"kind":"REWORK","_number":1,"created":"2023-11-16 11:02:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/901130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/901130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/901130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/901130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/901130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/901130/1"}}},"commit":{"parents":[{"commit":"799ba44062027b90a88849d1afa3c549ae396044","subject":"Merge \"update constraint for metalsmith to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/799ba44062027b90a88849d1afa3c549ae396044"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 11:02:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 11:02:57.000000000","tz":0},"subject":"update constraint for etcd3gw to new release 2.2.0","message":"update constraint for etcd3gw to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I492a934613c5aac56f497eba37801ac5a5277b88\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I04663b9c05306118bbef1d5d6c49257d4fd33030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58f96c7f10e149091fb8df8500469062a9bd63ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58f96c7f10e149091fb8df8500469062a9bd63ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
