)]}'
{"id":"openstack%2Frequirements~901131","triplet_id":"openstack%2Frequirements~master~Ie9613a78f73a5ebb701339da3781c8a0fa832155","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie9613a78f73a5ebb701339da3781c8a0fa832155","subject":"update constraint for tooz to new release 4.3.0","status":"MERGED","created":"2023-11-16 11:06:00.000000000","updated":"2023-11-17 10:39:50.000000000","submitted":"2023-11-16 21:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901131-new-release","meta_rev_id":"c78a2902ae08f1a7df197fd1fe8e341f9e75a174","_number":901131,"virtual_id_number":901131,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 21:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-16 13:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-16 19:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-16 19:32:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 12:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 13:10:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-16 19:32:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a340dfc17cc2bf8c06ade744aeb417b12e7dcc99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-16 11:06:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dc89ec2121e395fd3423115eb276d18b7c3d57f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 12:37:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b37092c028f94727a4ce3ccbfa1d4088\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8b583663ab49e7bd4ce525861ac4ae : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae4011b0b2804b6fa5c79f9c65d52d6c : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe41875261b4efb8f42a7539142bddc : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/deebc7c578e04203b2a56dc3bc4b637f : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98e312a7f0714dc7b39497204f20c8b0 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a59abeaab16d4697923590b380f0ec78 : SUCCESS in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b64077b9da342a1ae22870de07a6403 : SUCCESS in 16m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17be723a32e54deca01f9b89f28255eb : SUCCESS in 11m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/da2f9079a515497086b91e48186704ae : SUCCESS in 8m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cbd9ac938a27425f81f2f3a7e2626cee : SUCCESS in 18m 44s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/530f04e3975e4f25b57626694f440e45 : SUCCESS in 7m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e080041a257444f4b036a267ac958558 : SUCCESS in 5m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3890114ff3cc4a92bf348c4cb620c5c9 : SUCCESS in 4m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9c23b60f4ca147f49d0b5e365bb38f89 : TIMED_OUT in 40m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c7d299751afa46bb8b89f08dfad2fade : SUCCESS in 4m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2284b1063384461fafaf0682b69cd151 : SUCCESS in 5m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/596803d82f154c2faf25ee3a1d03cea7 : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c1db3c6474b744689ea1ec639442efb3 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eae926f0777141c1ba3f44c5b8cde4d8 : SUCCESS in 7m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3fd9ce0b510c40d7a5c22654102ca7d8 : SUCCESS in 8m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/00945b1a98594de0b3a4f0f94b8b75ed : SUCCESS in 12m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2cafc815a3174e66bc8db6c97d0c12c4 : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e867652c64024dffa13b7dd1f10d2624 : SUCCESS in 8m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/035a562768314f3994a452e189b686b3 : FAILURE in 4m 32s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/00432abf56db4f6bb8a14dcf68215e2b : SUCCESS in 14m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ef2319f71e87467aa3c3df6d3c43b6ad : SUCCESS in 23m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ea9b1aacd367423e9826509fb34ea38c : SUCCESS in 9m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c959f808a97e4171bb93a709d9af84ca : SUCCESS in 12m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/77594fad361f40caa4a02329a6976328 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7b79eabc950a41af8b3aafb46c398a53 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb7147e73afb44ba9815e5224ce679ec : SUCCESS in 25m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c55dee0a048d4750b5ea5bd80279159f : SUCCESS in 4m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2db7a15d0b2e451387e4960aa7c43db4 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/16210cd6dfdc496a844b3ea13e46d17d : SUCCESS in 6m 12s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/baaa3e7c7c6a4f3fae824353928364b2 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9089db5329e2453b8e5ea80c1778eb3a : SUCCESS in 50m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a44441afe4464cf0b356e0727c764b4a : SUCCESS in 1h 25m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0f5a81a129f0ee8e06c31ca1c33714d8e58baf31","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-16 13:10:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b86b95dd031115fdec10dfecbbe42269518ed3f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:02:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54740ff76ba8426aaafbbfb943a28126\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe034a64c224c30b839ae982bb53a32 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c148539d8c48f8980027af5d194b37 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21678746a37a48059971bec68617077f : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07fecffdee8a4f3396a1d02467613fab : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7db8969456ef4223ab743f1a7eeb80c9 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbcb91e597814827adea74b678e43e0d : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ad1f106a6244e2daf2836a408b2f2a7 : SUCCESS in 11m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/abed347fa94a40249d40a536efb08509 : SUCCESS in 12m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4ac9be3ec35147ca8f47f5944347108c : SUCCESS in 10m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0c449f072c6d4c16bc453d62f4c0b321 : SUCCESS in 16m 42s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/35ff660a39c548b1951083fa9c398dad : SUCCESS in 6m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/66463593c6214496a65487ceaa44386d : SUCCESS in 6m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/664f9ccefb634a478ce24bf15d12dfcb : SUCCESS in 4m 09s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2747c973177a4f798bbcd1f587e73b28 : SUCCESS in 10m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8616f1c7aae34faa8deb2d2533b30870 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/97c560e0297c4539a65f117203f231f9 : SUCCESS in 4m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9e96494629174b57ab5d9a68238635e4 : SUCCESS in 8m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/db3d60266fd8449da30204436809d7b2 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16f6ff3c00254e328e74cec7b5ef2f17 : SUCCESS in 8m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0d5f64c8675f4b99bc41ccca9ac5eaac : SUCCESS in 8m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ce5e6743173d4877938f9e29e8bad7d4 : SUCCESS in 11m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/355880867f8a4787b76929b85e099a44 : SUCCESS in 3m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8df4c928445f441f9993d7757c71894c : SUCCESS in 11m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bc7cdf980b5c482f8f211adf93c14604 : FAILURE in 4m 13s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/49ea20d998e348b68a7f6886e390c040 : SUCCESS in 11m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b6f1fe09a2504f99bcf9378984d73344 : SUCCESS in 27m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/829b311be1fa4e7989a00f337c122012 : SUCCESS in 11m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bdafd8419f3c4627bf9643970774ccd1 : SUCCESS in 13m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5bb7441c73b14d0ab3abef7a389751a0 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80b27655cd6647bc9b05b1ba261bd863 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4d15ed1f3634533949f91f682a038e4 : SUCCESS in 19m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f2770e7a929446bea33b010ed755c6f6 : SUCCESS in 5m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/07564e5fa8a946f49df86da17ef6ab37 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ffa9f817c704e60bd87ced16ce11dbd : SUCCESS in 6m 10s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6e17a03ae7564e269fba0e4515328a66 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63413c11d14b4b8487f3051e121121a9 : SUCCESS in 1h 03m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d01fd2076bc4d8c9a8f9dadd8ef71cf : SUCCESS in 1h 44m 24s","accounts_in_message":[],"_revision_number":1},{"id":"37e0e3e04132f542fef09a9ff65f614d9f88fc34","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-16 19:32:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b26689aae8cc2f8e0410c1316419d1f4ea580808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:35:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c136c84e405963a95e9cc7fb7e86481147dac28b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:03:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/417c1499f3b4433094665c9949f21692\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a8d1e2c75b482aa10f7460ecc46530 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aefbadcc60184359a175ca5730731a4a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32ffb1202d474be1875948cb585abaeb : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57e34e9db447454eafc739f76e78047e : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0013f79928824de8891a633f891ac251 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0fca86eaea04c648121d41d45dbb069 : SUCCESS in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b0102bb52573448f8b2e9018e8ddc07e : SUCCESS in 10m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/93eb8cf9829f429491667527170bca28 : SUCCESS in 12m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2c3c773918334142baa4346a508922e6 : SUCCESS in 9m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4a86b394368e4565b3c42b00bbaaf246 : SUCCESS in 14m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b6bc214d6bfc4056a982226740bc4eb3 : SUCCESS in 4m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5bfacc7c58b245e7bc6d7c8ce3058c83 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7d7bbffcb44847ca9140b1afe6eac977 : SUCCESS in 9m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c4a348d4d4d74c679cef718352314e8d : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/935db4d912024ee5b751cbe25e2af791 : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/28f15033866540e882377c98620e60eb : SUCCESS in 9m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3b2ace877a7048d6a608397cd699082e : SUCCESS in 12m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1e4c934e7b59490d9e0b6adaa9a6b4a6 : SUCCESS in 6m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/06a158b217f6486686df31d41699cba2 : SUCCESS in 9m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/eb6bbbac46e64086aa7fddd27c1f7335 : SUCCESS in 26m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2aaa2d747cb84252ab65238954387d0f : SUCCESS in 11m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c5e8ca594f2a46e8a710f214732d9b78 : SUCCESS in 13m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d95572de36734032bd80a2867797b778 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4d4602e6d2649909200df5efa11b45e : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/419812d71d564ded8052d1fbcbfb45d8 : SUCCESS in 24m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fe373bbabf884cdf9e793197ed3281b5 : SUCCESS in 6m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a3c93292ed5240f0b0e73166487fefe0 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f7e14b8e0904b5685dcad65fed754af : SUCCESS in 7m 22s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/08d1aef2d08a499894f6f3fbd802b662 : SUCCESS in 14m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1736dca010ac41ce80a960376fabe77a : SUCCESS in 54m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/680abde6b3df4551b63c381498b9ec24 : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9fee2ca4f265cb892541eeace0c0072aec381006","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:03:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"237f637c4d4277be00f3460f2c7229717b653586","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:05:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0a305334058447eba33f4c138b72183\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6beec7d65c485b971a7683dc1a40ce : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"c78a2902ae08f1a7df197fd1fe8e341f9e75a174","tag":"autogenerated:gerrit:revert","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-17 10:39:50.000000000","message":"Created a revert of this change as Id6b01597e02c8b19e237a8bc7b2636706e534536","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"102288da7d335b28aae1e55482e9f4bb11ed1a34","revisions":{"102288da7d335b28aae1e55482e9f4bb11ed1a34":{"kind":"REWORK","_number":1,"created":"2023-11-16 11:06:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/901131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/901131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/901131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/901131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/901131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/901131/1"}}},"commit":{"parents":[{"commit":"799ba44062027b90a88849d1afa3c549ae396044","subject":"Merge \"update constraint for metalsmith to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/799ba44062027b90a88849d1afa3c549ae396044"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 11:05:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-16 11:05:59.000000000","tz":0},"subject":"update constraint for tooz to new release 4.3.0","message":"update constraint for tooz to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I58c92e4bd3976b2c7621cd53637be34d6b0c69a2\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie9613a78f73a5ebb701339da3781c8a0fa832155\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/102288da7d335b28aae1e55482e9f4bb11ed1a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/102288da7d335b28aae1e55482e9f4bb11ed1a34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
