)]}'
{"id":"openstack%2Frequirements~901231","triplet_id":"openstack%2Frequirements~master~I53d95bfdc6e638b5117659b7dfe5f25f024c04b6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I53d95bfdc6e638b5117659b7dfe5f25f024c04b6","subject":"update constraint for neutron-lib to new release 3.9.0","status":"MERGED","created":"2023-11-17 10:23:53.000000000","updated":"2023-11-21 22:03:04.000000000","submitted":"2023-11-21 22:01:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901231-new-release","meta_rev_id":"51d8beab99371afa89c8d56037b25bc249f702a6","_number":901231,"virtual_id_number":901231,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-21 22:01:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-11-21 20:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-11-20 05:41:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-11-21 20:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 11:59:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-20 05:41:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2023-11-20 05:41:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-11-20 20:12:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-21 20:09:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c786723ac29d479e4279a874c2f36f66acfa645e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-17 10:23:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b92c5d58056701ed3540eb267448ff7cb36b2ef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 11:59:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbb1b3a234da462c86dca169d1c59467\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cde7189b03947c7ae49e74c7ecc6397 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d34c964a579e4fada074fb22788586f1 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90ae5822badd4cd49bbc9c44abae04ba : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0678673755394768a4274ac89c14f7a7 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97b2d9fc24fe4294b339225dec2caeba : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a0f54e624c84560885d575d66c98e88 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3358a44ff1604ca4b8651c7e988f7024 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/94ce11c648764fb3bf159777c3f4d1c5 : SUCCESS in 13m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5b8f4d6b6e4c4b6c87414013f140041c : SUCCESS in 7m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/98b2202271de4b90ab72c12a8fc59430 : SUCCESS in 12m 16s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1293b9420b5840438da25772891491bb : SUCCESS in 7m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2b8792e5357944e29c3963e3a45e6536 : SUCCESS in 4m 42s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4ffc1618c7c74f40bfb2e78b8d435a41 : SUCCESS in 3m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d121f9d8b07c4cadbf709753e3e88d86 : SUCCESS in 11m 49s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/0504e024745841a78c440ca37cf6caca : SUCCESS in 5m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f68aebb6ab274203b95bbaaac992db36 : SUCCESS in 5m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/baa32e3a004348cbab8319bd69bac96d : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0e03b1dfff764a93bc5b3bf1b77fba27 : SUCCESS in 8m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e2a423b8e8294f4a9720cab2ad989680 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a6cf7652cb4b49cea52012932379c4c4 : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/464fefba3af8485aa383877024347abf : SUCCESS in 8m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/97f3f5d841504a309a42e5655c52ae23 : SUCCESS in 11m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/00e7fe1cfd3448fe8e9e60c55cf34bea : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/14cecef034c345788e4b96cf3f077c21 : SUCCESS in 10m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2d9ea4da28c84e029d368859e04ba752 : FAILURE in 4m 38s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a310f0f4b1f44a4d8843fe1c68bd4c35 : SUCCESS in 13m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/239d67bd12744480b4f0f3222dd4b4eb : FAILURE in 3m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cd02ffc19b6d461cae0d4ac1685b1039 : SUCCESS in 12m 35s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d0da4c49fd8a4c8d813a822f05e01683 : SUCCESS in 12m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cbae1e6e48574c70bbf2a54c36edffab : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/446eb54d61aa4777af1b8edc513928f3 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b399fa9dcbe5404ea34f3d20c71f0f42 : SUCCESS in 23m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3a099fa2585e4f949e809c45a8550a9e : SUCCESS in 4m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d60aace234cb4f4e84113cfa063711d9 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/179dd5ff1b4640d9a16b990af71ddd44 : SUCCESS in 6m 10s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/dbae287bcf3a4fdfbf00bbf98e87278c : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faea4e3b3e6f4b3386554e14eef18319 : SUCCESS in 55m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95573076b2014298949c071f0c5c28ff : SUCCESS in 1h 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e5baf0cf25bc3e2630e65ca1fba61dc2e280991b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-20 05:41:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe10d3e4287b9a23a0883572b5b2aa19cee33805","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-20 05:41:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ffe1e819ffe2a3f9c904a3f2d125fd27d262c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 06:50:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15a776691b8b4c78908e3122bf35c5aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6369d667c20444ce850b5fda1c6c0485 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be36eb24a4e84d128b2e1a5095dca2e7 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a44e454bb53a493f9d961c8ec5912302 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/adc62ab9450149788b7ad49a0277a6f5 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10a8080d3d0342c38d49ddb4ab00160a : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/951be4845abf488f84d7aab145a0e36f : SUCCESS in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7bb32939e7474d578d3c6bacd9b6ded1 : SUCCESS in 7m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/af81d88123a04ad7a19ee0c5d52ef321 : SUCCESS in 14m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/639dd7dc25ee4846a886a1502b2d818c : SUCCESS in 7m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9c1878a27b15422c9a9d0c8dc8a649e1 : SUCCESS in 11m 12s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7c59e10fe1f24e419474cf4dda0ba9ad : SUCCESS in 5m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0e03901585dd44f2be36b6192d21744b : SUCCESS in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/04eab2740d7b4ac28f3137d3aed67bbd : SUCCESS in 3m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/01b66ebfd5e8414281be68075275fa87 : SUCCESS in 9m 28s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/c8452afa0afa41e1a007b75f5a45f448 : SUCCESS in 4m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aae3b12ba1cf4af8b09c048856cb0d38 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/01890bf7f5c14e7e8bdde11b49f80ad4 : SUCCESS in 5m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f1fbf470aeef479f944bb29b768575a5 : SUCCESS in 6m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/028cfe0186934b98b4b6e7e03a73d36f : SUCCESS in 8m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff51b90b831f48c39767a72e8b43c37a : SUCCESS in 7m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/360790d34e6f4843adfc555a4d397878 : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/32368d73df6049cfbafe069b4d8230fb : SUCCESS in 10m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4a6ff61361fa4a3cb27c9082d67feddf : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0dc430d0e61d4ea98e6f754ceb803f5d : SUCCESS in 9m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/069169eca73b45489124b5e37892653f : FAILURE in 4m 26s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c00cab90c59b4ba18db4fc3eab3bc4b4 : SUCCESS in 11m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/aca8935ca0a64bf1a818467fcfc73de0 : FAILURE in 25m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9b75dd9bf6ab4cb8bbe9c69eafecbacb : SUCCESS in 11m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2fc8e02a5f9e450eb8bb1b83753f1747 : SUCCESS in 14m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/47f8e2bc07744e118ed8db00a2c898bd : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90145ae882fd43969f89ee411bdf0c07 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e07a85cf69f345be8c4092f315a9f2f6 : SUCCESS in 23m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9aacdf8d9fb44540ae325d47621f8a6f : SUCCESS in 5m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0c50b176f68947a58120c1ba63818930 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4a3919db6f248869b39bcca99066b67 : SUCCESS in 6m 18s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f1f69954d88f47ec801cd8fe94935c89 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac191c997b074dab9b26c7581a3de1d2 : SUCCESS in 1h 02m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8056ad5905844e0aa280f5aae41a1a62 : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":1},{"id":"48e789c61d529d0e029655f8c01412f8db7c2d3a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-20 10:59:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ab91a389b0437c60c9af2d705c91dfeaf021055","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-20 20:12:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc318fe9d30ca64b77eb0f1324df1c6ce445738a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 21:43:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4ed560fc70e48efb0aba0c5756d3eab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9a5250983d416780213c0af120c0f6 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c783c213ec254218a2a5d9c4ca22d801 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45d1dfc260414874b3cb0e2b2c7a91ac : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/682a87ac9a85459b82177590d9aae185 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0265592a1814aff9170a1ff3128e605 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b510d5776f014cd98cb059bcd822fc09 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb7b17c967af4cda959a54ecfedeaa77 : SUCCESS in 8m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f173e2c59f9440b5b775ab6729946b38 : SUCCESS in 12m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5c35231fb5a244dca477d7850b7e02d9 : SUCCESS in 7m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c8bb087a58864226bdf8608bc331516a : SUCCESS in 12m 36s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/535cfcb763514b17a8f31219fe54220a : SUCCESS in 8m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7118f8b369ff4443a5835dda4b6fe167 : SUCCESS in 4m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a8a4bda448fb44c39f417ac364a47a76 : SUCCESS in 3m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/190cb3c761c64db184d6fb1a6ac44ff0 : SUCCESS in 11m 19s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/600a8a5df6c14c47a903a4db00bf0c64 : SUCCESS in 4m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7ab1bd425e1470793836504f5bf6e26 : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5a98ac9c2ba54b17b26c073b15ba27e5 : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ede93bc7633e445786fd412250b8b072 : SUCCESS in 6m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e12fe11624ab4176a480f817d99b2340 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3972a152a78440bda616a4ccc822646c : SUCCESS in 7m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/755e54af83d54ffa81e8278762489e25 : SUCCESS in 7m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c3c80beed7434a51b77b281085c54501 : SUCCESS in 16m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d41aab8b6a65496bb76fc9a656dbb193 : SUCCESS in 4m 20s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3a673fba23ff44bb9dca5cef767ad0a8 : SUCCESS in 15m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/139bdf62a33b438e855546739351f757 : FAILURE in 4m 57s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/861bc817159e4b069a5c70a492cc7e98 : SUCCESS in 9m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/239f063eb6c9457c98ddd8a49d8fc790 : SUCCESS in 24m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ba27baec05fd4439a818002b7b143f1c : SUCCESS in 11m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ff892b3197f544b0bff41e02eed69a28 : SUCCESS in 14m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/60f75e9add9b42448e1a46f5e989cff1 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a07bb8238ab94ce7b42b19b81c5be9c0 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/339feb51dc9047de802e5d8adde9f934 : SUCCESS in 23m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3c223a8f309c4c71bf301cbb633accd5 : SUCCESS in 6m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0658b9013f1f4c80a95b8a74c094c0e0 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d9882c6fd15946429f7b03ae4d14cb28 : SUCCESS in 5m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/870226a7e2d544d4905489b7bca46ad2 : SUCCESS in 14m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40a7fb655aac4d86b0ada0d1141c9158 : SUCCESS in 58m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84acb9b35a3644b3a2039d006ec00742 : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"52dc9a3e33d10c475f3a977ede7893d19e56a5d3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-21 20:09:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0abe67e6e9900d167eae2d72fcde5960ba5e6e1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 20:10:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"149c4d47f0204f46cb95d0ecb46c5d73a219d6d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 22:01:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/688022deffe14d948d7a4ac6368edee9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ad597cc94d455c90b1a6dc07f8f68e : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278729e005af4607b9c6b547b677f34e : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/911514ec954c4e069b98c86f358ec047 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e870d85f0ca4a2a8929d82abaac1262 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3e36f1fb10f46f5853851cd08ac54f7 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f48337c2fa344cac871be020c82b6399 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9d370489600443094cf2f96365bcbd8 : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/41919fb7415e416dbd1c82bbccc14cf7 : SUCCESS in 15m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6e19eece607c47b9b6b18cbc5e3e1572 : SUCCESS in 7m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7452be0bbfaa4c218fb843b869438465 : SUCCESS in 10m 53s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/76be34e70730462f82290caaa45a742a : SUCCESS in 6m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3367e9fd9752419a926c37dc56f9e89c : SUCCESS in 5m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3eae30dbc28d47c7bd70e838b8d7292c : SUCCESS in 5m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/958f570cf3bb458b85bdc9c90ad8a949 : SUCCESS in 7m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/acf34aaeb57449cf89bd69a3d75c10da : SUCCESS in 8m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2669c08845e48578fccc2a9f49b056f : SUCCESS in 6m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/85c544fa936c4189a715ef2496b277c0 : SUCCESS in 6m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7c3f06b4e6ab459d81b4219166e1b785 : SUCCESS in 9m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cef90483bc2e459bb9d466510d9ccbf0 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6d70a8111f834daa8f2227c9587719d4 : SUCCESS in 10m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4dcd348f1a8c4e43b7941468dcf7496d : SUCCESS in 27m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ab4627809a6f4344b212b67105755724 : SUCCESS in 8m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7e64a562df0a49ca9204e23e064013ca : SUCCESS in 10m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f9b4c13cddba4210b59b1b7c482315a6 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd4184cc16e847e6ba64318d5e435b65 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6eee7c3f345f4037b81d928b4b82149c : SUCCESS in 26m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f4719715210a4534b950875ce73fcf00 : SUCCESS in 5m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c2220b62de9a41908432ee7add7c6bd8 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/284acff2c9ca4bd58104e29fe80250be : SUCCESS in 5m 54s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e0e39a93908a454ba0c3bcac83f1f46f : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62cc499a580640f292f42cbf650feca3 : SUCCESS in 1h 06m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/320273e8b9f744e28d1333076f202587 : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b43a8686a841f6d80acdd9baa6dabd08d0d0f4c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 22:01:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"51d8beab99371afa89c8d56037b25bc249f702a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 22:03:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9276c21fa694e3e9fd5aaac2214e880\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71009719bfe048a091371e7a475437c2 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b48281bd91dd521ce67cc24c3ce1e3cd269e53f","revisions":{"6b48281bd91dd521ce67cc24c3ce1e3cd269e53f":{"kind":"REWORK","_number":1,"created":"2023-11-17 10:23:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/901231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/901231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/901231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/901231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/901231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/901231/1"}}},"commit":{"parents":[{"commit":"8570e45e11d296863621efd7c1ef4aa398dddf57","subject":"Merge \"Add cross-project job for mypy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8570e45e11d296863621efd7c1ef4aa398dddf57"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-17 10:23:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-17 10:23:52.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.9.0","message":"update constraint for neutron-lib to new release 3.9.0\n\nmeta: version: 3.9.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Change-Id: Ied01bd54f8ed847e06c06d8062f6bb7e751d5bf5\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I53d95bfdc6e638b5117659b7dfe5f25f024c04b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b48281bd91dd521ce67cc24c3ce1e3cd269e53f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b48281bd91dd521ce67cc24c3ce1e3cd269e53f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
