)]}'
{"id":"openstack%2Frequirements~901433","triplet_id":"openstack%2Frequirements~master~I086b7a56e06ef4d5f1363fd52785c1b393e68018","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I086b7a56e06ef4d5f1363fd52785c1b393e68018","subject":"update constraint for python-magnumclient to new release 4.3.0","status":"MERGED","created":"2023-11-20 07:16:16.000000000","updated":"2023-11-20 20:33:58.000000000","submitted":"2023-11-20 20:33:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901433-new-release","meta_rev_id":"e79c297eed8f6344c48d012373689bb9f09ed120","_number":901433,"virtual_id_number":901433,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-20 20:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-20 14:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-20 17:56:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-20 17:56:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 08:52:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-20 14:25:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-20 17:56:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"134f006fdebc47e7ce1320f8d00d0dca8561e3e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-20 07:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50696cb1e0002e4cd60e5f586d0a766c8f489cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 08:52:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e33bb90114246b2a4228d8f0a5f42e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf101ff73bc4b788dd57fb87146dff4 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05dfff24b50d4111a65c72455273fac0 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bcba8c065d147c9bf0f852fd6533dea : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e52f6e2b0b5746119a51840f3c7790f7 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3ca9637a7024b1aaad3321190658fb0 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d86baeb6db66437690fa0c4c6826b88e : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c6001df61e7f49319f7e51afe71986c9 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a72b40f2d3f74f89b5c7bb308e59dad3 : SUCCESS in 13m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f868be9525e144738607fa380fe3cea2 : SUCCESS in 9m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cf9faa520fc046e1b59d8195b37d9891 : SUCCESS in 15m 35s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/60a7c50106ea42e2975e1b6024bb9430 : SUCCESS in 7m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/038c58bde0c54b9980832c1155cf29a1 : SUCCESS in 5m 25s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b96c3698377a432d9e51575404dd8b36 : SUCCESS in 4m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a789535a8fa7447e87d4a100ba481912 : SUCCESS in 11m 37s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/82d1d64290dc48efa3722fbd748ff971 : SUCCESS in 5m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d6f4e59957134da5a24341de04cd6ada : SUCCESS in 5m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/07eac15ee30544bbbc9bf403daf3b66b : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ddfb630a0be04198a3794c79c53b9ebf : SUCCESS in 8m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f546abde9d6f4cbca2d9848ab5e9b264 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/279027e1279b47b989ff57eeb95f21e9 : SUCCESS in 7m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a413f3e6731d4473b83cd2356f04ccb2 : SUCCESS in 8m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3e2169b5cbe447b0b2e9c1378b3c15be : SUCCESS in 14m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d7100fe70cc54231a9a9db41ae802a5a : SUCCESS in 3m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1d38459074cf484cb0d09ea41c558727 : SUCCESS in 12m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1db091f4886746c1a2d016c7a089539f : FAILURE in 3m 48s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0652db2517ae425a9eb897d5e59eba59 : SUCCESS in 14m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/28a7862957f2468db453d140bcaf1231 : SUCCESS in 28m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8b2f38f4f43f4853bcde83ec0ddab388 : SUCCESS in 10m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ac65303f9ef141e69fa2009ba9284999 : SUCCESS in 13m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1e3787de9ffc462dbe14ca4fc2aa19b6 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6c7c6b643184707a204eb6dc9e4f62c : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80603b9810a641efb74a93d4999ca497 : SUCCESS in 22m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/05b8b8ea4afa447fbe8605acc8b989f5 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/062de621146642d88f94e2d1f05d7d05 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d802c51bc4124c8a9380d05aed12b821 : SUCCESS in 6m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6e498e2978e243dd841e6e4a63f5420a : SUCCESS in 9m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5d1119c630e445db04da30d7ca319a8 : SUCCESS in 48m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48647aeb5fa3411dacc45580107323a5 : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7196ef1b3adec2b96d03d1da7bc8a29ed4088d36","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-20 14:25:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ab1c190a55cc70958822b7af7fe8342a8de90a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-20 17:56:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d95e809bd4b98b9cdc3b7d4f016704736e9755be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 17:56:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0326a644e8a406c100148d993a94d39456873acd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:33:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20cb70a64c743c89fa443915f6e2491\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ba7a68961744d5994fce8111cdcfbb : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02626469be4243868e1acf0b971db6c6 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0e251c13b8e4c8cb20987758cab2b84 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3fdd67132d3495da73e942399dbcd19 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3896db5654f4f49ab781e6966a0a3f5 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d662d23afa34a3bb5921a7ce6acf73a : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/052d1a60e95c4eb683bcac236302a1a6 : SUCCESS in 7m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a57489185c284081b1c9414b0ec17c58 : SUCCESS in 14m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aefc3896d11b4f7481c88486b25e7cfb : SUCCESS in 8m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/513fb0e02dae47d2996998994b3fa235 : SUCCESS in 11m 53s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/683dfa88c8de47e3bf3aeb2389d85594 : SUCCESS in 5m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fc7febb028654c178db0e4310fc8078f : SUCCESS in 6m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/85cf953d3b4b4064a4ca425533503ac8 : SUCCESS in 5m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ab47c06edbf74e20a0f26c68a810a305 : SUCCESS in 9m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/79e0a3d45a36418bb612572cbb157935 : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/729484e688c9497ca0037803f917f852 : SUCCESS in 8m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2f50a67c81954721b94a454d774250e2 : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9046c895d1654c2786d73f69a9939f9b : SUCCESS in 11m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d79c1376e0d145ecb8de9da548d05f5a : SUCCESS in 4m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6a714a7e4cc341928dc1f78d03ae214f : SUCCESS in 13m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3ab11551536d46d0b13ade31daa68048 : SUCCESS in 29m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b689e8ff82ce423f9f4712a0bfc7d871 : SUCCESS in 10m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ce487785915141daa3cc39897d07e4c6 : SUCCESS in 13m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4f598cd2c4fc4e719cf0ce548ba7e042 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6c691a6562fa4ff79775c3b3cd2c00ee : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2877fe7104fd4133ba1155796faedcaa : SUCCESS in 25m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bc230c6ab7a249ff88d695430ffe3101 : SUCCESS in 5m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/955f52003d8847f99cd5ac3125541ad8 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2587bb943f942388022292693bdff89 : SUCCESS in 7m 36s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/02e1f9aa0752449882fb807262384b6a : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7d4f86b64d24e158736fb557047c4f5 : SUCCESS in 59m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/427dc43438da43f693d97f8bb84ff243 : SUCCESS in 1h 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0596fc707c7ebb8699b3dec6e0a4243f3f315b0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:33:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e79c297eed8f6344c48d012373689bb9f09ed120","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:33:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e71a91894561499d8c808bfbb3351eee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69fd40502ac4f7eaa32e7d52977455d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e509f36990110eca57f80bbdbcbad41846d075a","revisions":{"8e509f36990110eca57f80bbdbcbad41846d075a":{"kind":"REWORK","_number":1,"created":"2023-11-20 07:16:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/901433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/901433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/901433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/901433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/901433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/901433/1"}}},"commit":{"parents":[{"commit":"3a122203dd3aecd85436469587d8048083d8fcca","subject":"Merge \"Revert \"update constraint for tooz to new release 4.3.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a122203dd3aecd85436469587d8048083d8fcca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-20 07:16:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-20 07:16:14.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.3.0","message":"update constraint for python-magnumclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie02ecf629db05ca7b15aa70a34a556024341d186\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: I086b7a56e06ef4d5f1363fd52785c1b393e68018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e509f36990110eca57f80bbdbcbad41846d075a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e509f36990110eca57f80bbdbcbad41846d075a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
