)]}'
{"id":"openstack%2Frequirements~901447","triplet_id":"openstack%2Frequirements~master~I4dfed4d414aef32fe011054c68df2b24ec8ba248","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4dfed4d414aef32fe011054c68df2b24ec8ba248","subject":"update constraint for os_vif to new release 3.3.0","status":"MERGED","created":"2023-11-20 08:54:20.000000000","updated":"2023-11-27 17:00:13.000000000","submitted":"2023-11-27 16:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901447-new-release","meta_rev_id":"843cab57b953d0c0d6e03ba08cb4de6faf8a9881","_number":901447,"virtual_id_number":901447,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-27 16:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-20 14:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-20 17:56:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-20 17:56:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 10:07:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-20 14:21:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-20 17:56:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-11-27 13:39:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"8cad064ff45d2007e6709566aa50d4e6abc4ba27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-20 08:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e509c3677702a89ee750ae44320b560cebe56e2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 10:07:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5340f33dcb413e901e240cf730e77e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b8861853b24191999cf979844b097f : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38cad89480c849639b281563af12ed63 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40513087feca47c199c182a56e0d47eb : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/379c686686ff42e1bbfcf78d7789e351 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ab877792c2240f197a57d2b405f7d0d : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de99e73cdf97438ab53883261180b370 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e623606575564805880b3024a4fa62f4 : SUCCESS in 7m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8a9c199bd74d4f64be05621be577b031 : SUCCESS in 12m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/526561fd009b41ea9b4375c73c65eaa4 : SUCCESS in 6m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/10af603f33b94513b91a282649064dd5 : SUCCESS in 12m 02s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0dbaedcd53974eba91a26bc1ceb6da02 : SUCCESS in 7m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cb06c173c05c459fb4ea2825e5a05a52 : SUCCESS in 4m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fb559f0e8e104fab8e023e588a5e4429 : SUCCESS in 3m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f2727a70e3c84871b3e678a210327654 : SUCCESS in 10m 04s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/6d6ee0f00ff843f4986f07cdac26ba36 : SUCCESS in 5m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/234ee484211c45629be7165b9289788b : SUCCESS in 4m 35s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a8586539c3b54c348f8c744752f68528 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2821c54c5c104ef6b1f0ae1a1a4d9ce1 : SUCCESS in 7m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/506e2ebf2e8b4b8ab2d2a018a81b1957 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/000bea7ea13341cdb4c6f8cd2d89eca5 : SUCCESS in 6m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/06c6d591cc704627a058d86b70af3b2b : SUCCESS in 8m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6f68b03105be4661bc5b6625625e11a7 : SUCCESS in 13m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f5d02f64e62e486cbe247c8a72c5d872 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0b1bcbb510ce47758d34fbc50eb8c9c6 : SUCCESS in 11m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c8a625a726f34c9bb16ed5134d78f429 : FAILURE in 4m 13s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c5293c1fb25444a2bf28850dd3bfc692 : SUCCESS in 14m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/928c2889cba54275b99cff04c688bc50 : SUCCESS in 23m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2c58163ad589465db877ba4fc10f2349 : SUCCESS in 9m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b4bf5e8395a4451f8e52aecac637ce42 : SUCCESS in 13m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d9ef9230f0bb4d9e85548ac6a6a8b818 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54e397c088364ef68a2db7635c1f9572 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dae0c51edb5642a9a679dfbb9945bc34 : SUCCESS in 22m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/234740a88cf44f3b9b85fc62e7858081 : SUCCESS in 5m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0e6bd769e9534c36a81302709e4ff74a : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ba45a75881540b1acc574091eeb1e0b : SUCCESS in 6m 12s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d16f5090806640a5ac9bfaa757408998 : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9fbf37ff3054a9f9dfa0b0fe1a7c988 : SUCCESS in 1h 05m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aca81e52cb114091a67dc62a18ad61c6 : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0eab3be5d31d9c9aef8d869f604ea6c85a3f7f19","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-20 14:21:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba2859b861e729a8cec55acda7dd71b0c45483b3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-20 17:56:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ff2b32b5c42faa4fe340ca9411f069fe9f338ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 17:57:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"998b20a25698c86c12d40c3a361ade904d7b6627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:40:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d8af52b84594cc58ee51485d0d31ccb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86bb39a4d3d2448c9534faf7e400ffa0 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b70f0fb5344457ba3c5cc9b3c3f00de : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d631d31ddd0f4b7ab17da6fc8563f069 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e381108d87f14656baa285ba790b49bd : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44dd4e960769499a99653b8b94b2aeb8 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b366cf216d546f3b5c82a020c75ccdb : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a3d778af6bf14f16b4e830ac31822c37 : SUCCESS in 9m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a751eb217efd435db94b9dba50ba68a5 : SUCCESS in 15m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/69d0894b5a7f4c29bcf6cfec8281c1bd : SUCCESS in 10m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cb391acbd733429aa71ea326566fb02d : SUCCESS in 12m 02s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/2cd60a27621748c399b9f7f8def19f68 : SUCCESS in 5m 58s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/03f52aa1f9994b27bec07fbc58aaaf22 : FAILURE in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bb8a4e30802241fb83e971f50bd94d16 : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b40d6e5b47a44f12a96aec0619e1becc : SUCCESS in 9m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/30728d7f40434946886af9a3d0ed31d8 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e45d8e3ad8cc4e428d8d4878b62bb13c : SUCCESS in 7m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e5d08fecbc98484e9f79f96d8517bc8f : SUCCESS in 7m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/49c77429724e4385a41de3775c38fb12 : SUCCESS in 11m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/38a2c57b67974a5e8477a23c9cd8d36e : SUCCESS in 5m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d5fd4af7e71c486d9fe60cc9ecdd9447 : SUCCESS in 12m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2c520b271e4a481eb84ed8b0dd60e6fe : SUCCESS in 26m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5845d64a1ed0488db6a13df2435bba7b : SUCCESS in 12m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/89d6a4b9ea8c49e59724f0a3ebaac719 : SUCCESS in 13m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1f6596c64976400bb495f171a07884a1 : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/176bb05bd0e64498b257315f192b0cc2 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8e9ea1ac47d47d5bff095752daa2836 : SUCCESS in 25m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6dab7be57a9b4b8cab541a7e5de66118 : SUCCESS in 7m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/303534bd50564b028678ee75009e8457 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02495feca01942d7b34e93acc224a8f1 : SUCCESS in 6m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4871e1438b8d4d79a4f48a52ab2f022d : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a735d688b15b469bbfa138ff35d8260b : SUCCESS in 1h 05m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a263a8df051a4b688629d966ad6948fe : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3a29e3529fc70c5378d2be56a2034a4da719f985","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-20 22:04:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d7ebe3c0ef9e7980d7d82cb79e4e0a7dcef6be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 23:41:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/551a26d95dcb4d988c76a504e2359653\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e6477a9b0143c993388faef8da65d5 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e7b5bde9c2473089f8103857a93f97 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f73f3ce78b1439e868c4fa112dcfb73 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11ba6947599a4b67b4f7e3b53fda86db : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6199b86d35564642a5713daec5e699f4 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d53a7db88c0f4d5eb2f868e3f13d1750 : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aee315ae650d46ed83d5a0502e5b5e0d : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/28ef29589fdf4311ab8ed03d075a92d4 : SUCCESS in 15m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed64c991b9104f459406b649719811f3 : SUCCESS in 9m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/659b51f2bc0047219e4b209139319dd4 : SUCCESS in 13m 54s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6bc865b4d1b746caa251fd338d8ff3c3 : SUCCESS in 8m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fac2f72b64f949f0afb6b2c9101bdc44 : SUCCESS in 4m 17s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8840b899c9234432a2f616f7769f81a5 : SUCCESS in 3m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/31cd9a79b0c746e5a2b656b64e8b6d90 : SUCCESS in 10m 52s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/ed85a5b7cdf7419eb425de3ad82ac43d : SUCCESS in 5m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2f62e6651844424894f49866399f53b6 : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/136ea85ce61d4d3aba876e4a3f897aa6 : SUCCESS in 4m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4f9ec06d7b314cb1b3b89cac8b0843ea : SUCCESS in 8m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a71aa3f689e94e05a2b9c2c358da00aa : POST_FAILURE in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a55bcc3e67af4683a76109205a9dea84 : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f6a633cf17eb4e4892c5f6636cad772f : SUCCESS in 9m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3b75a20c28c24335a85b346db4809b61 : SUCCESS in 11m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0c5b1fe2b581481bb6121a4568434b6d : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5578cf28a0184fbd98d0130254df635c : SUCCESS in 7m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5ea91fe9457043e8b8772904d292f8b6 : FAILURE in 3m 19s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b5bde183c81b4227abdb80783899cc08 : SUCCESS in 11m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fe8e7f26434f4f289654e2b8ca337f0c : SUCCESS in 20m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/db0069a52cdf4549bab816e53b71f80b : SUCCESS in 11m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cf46a6a46b994f61a7d03f4c0566e6f8 : SUCCESS in 14m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7eb66edf5258461e82f5a803bd3af659 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d43c496dcaa2440d952a4ee8afb83eca : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9c6ef9f14684518983d1c128e549b91 : SUCCESS in 23m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8fa6dbdffcde45b6a2ce0d0342f6a62d : SUCCESS in 6m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5575eff3130b43f2ac3f44aea2d5b007 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1517c70c563749a3b943a084c49096f5 : SUCCESS in 6m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a040ef3de3b8451ea897c1fb7d80a843 : SUCCESS in 18m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5965b3cc19fd41d69b356b64bddd70c1 : SUCCESS in 1h 08m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/585eae8111ba4b918ed85103831143de : SUCCESS in 1h 35m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f875234abc44fc56ce1bce8815d7b61dce14a825","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-27 13:39:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0785d92e1e4cf929728af164a9576caee98c462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 15:36:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70460b3b548944018ce170a091f2793c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86dc369bc0e64ca8bd04545fff5a9e82 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89aac50f2ee54397bc093e634f7d3fc5 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95a7908b356b44e5a420587774a5962e : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8294692254b44210b703bcfbeeaa98d8 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/524c581d39234122a68c1a2747f32c03 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88604cffc4ed4699886aadb88db08ac5 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b2b69c487ec442f28797694985f88937 : SUCCESS in 7m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/16fef222addb4e279802f5c9a3e99f46 : SUCCESS in 16m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b62a4782fb124706a62f196cb5e1ab8a : SUCCESS in 8m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2a241bfdbbd64d61a18a759c97940f52 : SUCCESS in 17m 45s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7930014c019649409eb216a9ffad5cb2 : SUCCESS in 8m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a9f2c91488e54d099628d23275828840 : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/085607167c764abdb97dec621683dde8 : SUCCESS in 4m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/be5fa96377784d0d9d191f32757f841c : SUCCESS in 11m 43s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/7ba9e68bc8c8423fa73a28d4e57a2483 : SUCCESS in 5m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/95fb6543396342ffbf391e7a0d19428b : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0e3793c6a7754c3eb246c4c8885b3204 : SUCCESS in 6m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/334b71eb9b584eaa9f73cedc9bd08949 : SUCCESS in 7m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/995cd04e623a454ba30522fd6909e295 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ffb6e22e10254928afc551286d83ad7c : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fbbd80ee97f647788de45e3b68e545cf : SUCCESS in 7m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/557c2778ff5342418c0748054f476aa0 : SUCCESS in 12m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/293a5aeda43849b7bfb0cf7cdacf7b2a : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bf7976adf8504b7d989d98c502ee6954 : SUCCESS in 10m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/42c1464235e741c2ac55738e3485e8b3 : FAILURE in 4m 45s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d03fa20cf37e48998ab0e9b427b237d7 : SUCCESS in 12m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/aa034c56e7924734be07312505f18229 : SUCCESS in 20m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/aba7f048b0a343c38fd10a28e069933d : SUCCESS in 11m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5c55dca183e94b459fcbcef6f71bd7b9 : SUCCESS in 14m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d54b607417214646910e3a9e014124a0 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e29fb72d706a4706a849e2295de93b9c : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e89c90e2d97b4f839c2f343822aa94a2 : SUCCESS in 26m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/402835dc1cd54f319fd1b311c37013af : SUCCESS in 5m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/692cf26d4142495e968a8874b97c2d4f : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9dfce964aaf44de7aa94a68217d8e18b : SUCCESS in 6m 05s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bef1fcf584724156929fdda5beed1eef : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2dddc149333042a4984bc03170615859 : SUCCESS in 51m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bec1e2278fc467baab33b5f4deae29c : SUCCESS in 1h 47m 54s","accounts_in_message":[],"_revision_number":1},{"id":"36358a2cd167276e7b6479fb35f45f89104b28dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 15:37:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41b4f51779cd8775b0cbc940c3c3d2865ded340a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:59:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bdaf217a71f4fadbb92a43c383b2872\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005d9239b983493a8c4cdfaf12a67f68 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8111c1e9cadc4deea2562cd7325fcc29 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/495edfce9af54cc58336e3fc480b738b : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87f65b67a5614d6e91c4fea4c1987b3f : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7def2a9fbc4346229d0333be77d1784b : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73b55d2d22f242a4adb38e4dafa8aac1 : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f88b6971e37a44f98d283fa43b822333 : SUCCESS in 9m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b1de81f84a9b4015b1fc386e8024822e : SUCCESS in 14m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fe6563d8636d4a3b9181c14af9bd7f41 : SUCCESS in 7m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/89070952aa514e92a3ad4022db38c11b : SUCCESS in 10m 09s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/f9085c2531ba4556b75849796ee86051 : SUCCESS in 5m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4ca8866f202f455392b3ad85bd4b1d41 : SUCCESS in 4m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/07605a9509c143a1b8ca66c793736f2e : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1e4c03b51ea84addaf4646aeaceb893d : SUCCESS in 9m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/da1d1e50dc6e455587a3913b550c56b1 : SUCCESS in 13m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eccada02a4a54fa6b5cd5496ad2ca4e9 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bc3640e041044df486a81ae1944574d6 : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4dcbdaf36acc4fa5ad17e15371721cf7 : SUCCESS in 11m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0355857236a440379df7471190ea9fd8 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a5adf4a54caf4ceca88c660a7e0dafeb : SUCCESS in 9m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ed9d5030598e4d8fa9a76ee13d4c725f : SUCCESS in 23m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/69176314a3eb45c29586dfe611f2ac22 : SUCCESS in 11m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/42f34628ec4143a3957daad118a8978b : SUCCESS in 13m 49s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/19e6e19162eb4e5cb82b7baa1d319eac : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a944b1eefa4f4433beeff97a0c35d0f4 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2bdd68d04674f779585e4b13af80337 : SUCCESS in 22m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/00d96b2706d9481fa2e5d2af3027d669 : SUCCESS in 6m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a402951043d94178ab6900c3ce37e3f3 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/914178dcae0a48f98c2c7f6f516a4851 : SUCCESS in 7m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/69b88fcde61f466983d84457b883e124 : SUCCESS in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2c4ae7cd02c45eab4874ff3a1e8c150 : SUCCESS in 1h 14m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/744a491b112d44b9855b0ccc4624b0ba : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e875387ae603ea3a8e03ebc0d63df269bf58140a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:59:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"843cab57b953d0c0d6e03ba08cb4de6faf8a9881","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:00:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1210b1c58904c9fb30e6bcf9a6a8286\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946e3cd3d2864183a3b4cca86ff8c535 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1731536f1721413b4f4b6bc14c705fcde736c0d3","revisions":{"1731536f1721413b4f4b6bc14c705fcde736c0d3":{"kind":"REWORK","_number":1,"created":"2023-11-20 08:54:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/901447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/901447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/901447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/901447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/901447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/901447/1"}}},"commit":{"parents":[{"commit":"3a122203dd3aecd85436469587d8048083d8fcca","subject":"Merge \"Revert \"update constraint for tooz to new release 4.3.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a122203dd3aecd85436469587d8048083d8fcca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-20 08:54:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-20 08:54:19.000000000","tz":0},"subject":"update constraint for os_vif to new release 3.3.0","message":"update constraint for os_vif to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic57072c43661f89486bbb8dbcee24938594fc70c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Shanjiayi \u003cshanjiayi@yovole.com\u003e\nChange-Id: I4dfed4d414aef32fe011054c68df2b24ec8ba248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1731536f1721413b4f4b6bc14c705fcde736c0d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1731536f1721413b4f4b6bc14c705fcde736c0d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
