)]}'
{"id":"openstack%2Frequirements~901528","triplet_id":"openstack%2Frequirements~master~Ic936617a2a70351625ff6137f5a2f7de78d48ad8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic936617a2a70351625ff6137f5a2f7de78d48ad8","subject":"update constraint for ovsdbapp to new release 2.5.0","status":"MERGED","created":"2023-11-21 09:30:17.000000000","updated":"2023-11-21 22:02:20.000000000","submitted":"2023-11-21 22:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901528-new-release","meta_rev_id":"4567dc1c9d67077d45bbb595bb7457afc90cabd3","_number":901528,"virtual_id_number":901528,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-21 22:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-21 13:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-21 20:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-21 20:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 11:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 13:13:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-21 20:09:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c5d5d657314af66287443e9b0192cd671a30f4a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-21 09:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fb982db469e7837847e0213d3c59d5d92f8a422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 11:10:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2c5f3a6b08c4c288822abb6b99ed867\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8cb7d83a4149f589415033c016ad0d : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de05f037ae414c18a527f5417598850e : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/113df7d3953a4387a42287aa3797e04d : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a0b365d556a44d7bae91f784e8c0bff : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7c6bce0ef8c4c0cb54f9545f7a84b2e : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/792f382a0bae4aebb8143f1fdfd67fce : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6430f120b08a4900964d69ceed256be2 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/19fa48c11a674790a262072d33abff17 : SUCCESS in 12m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1c69c38a630044958f2890a8c5276b93 : SUCCESS in 9m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/78d17fe400d44ababdf33aad605bdc88 : SUCCESS in 21m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/228fc0eb724b4116af6ca9339b7c5f1c : SUCCESS in 6m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ff584874365740cc83b8e4c11c4e9369 : SUCCESS in 5m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6e9862a0d92049b882f374bdd277075e : SUCCESS in 4m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7f748d6207f447208dc4be7b345693d3 : SUCCESS in 11m 13s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/d9592dce8f6448a790261fb2c1192787 : SUCCESS in 5m 58s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5ed0b4025d1841a0938e37ddb9f867ef : SUCCESS in 5m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/204a594d2dec41b48ffc651563bc316d : SUCCESS in 4m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b26a571339ac438f9788d921a5d59040 : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e1d8fc076fe940ce89bb6d47e96461cc : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/317999d445074c0ca2308bf27b8df34b : SUCCESS in 7m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8dcbbe2dc6ac46128c5fc0ff5c07a8fe : SUCCESS in 8m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/14a9015454c74e1c8289e1184ae85686 : SUCCESS in 13m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/42ac6f3583f045d49306b1be717bad4e : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/889325ddafdc452a82221599b25f0efa : SUCCESS in 11m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/33347686c18643a1a9a17d0c75151328 : FAILURE in 5m 08s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b67f7c5ef81e410ba070c23207504104 : SUCCESS in 13m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6d013923938344f28868df3867f1aff6 : SUCCESS in 27m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6ae079e46dd246e98a01729f85bfc6f6 : SUCCESS in 11m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c91e63ac7d7f4c64bee8b438e7eee46e : SUCCESS in 14m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cc54d63e18184a9d8e37a11d3a4bff8b : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2e0a0a167b9243efbc416315c1f9db85 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22f4b112461a4e91a3837cac1a550c53 : SUCCESS in 25m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3216dca4c0d74a1d8125da0cf98a7667 : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0ab8feb021944bb4bbd767fa69311847 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccf0646519054fc8bacf4dd6c8c0df9c : SUCCESS in 6m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a39fded3978e430ba57cebae28f96d9c : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a0594c38a2745a69c30ff84a913c820 : SUCCESS in 55m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c4f85775c1d425ebd62195aa5d116d5 : FAILURE in 1h 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b52637dcb0426c98f8d0212538f13c5e6a3eda30","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-21 13:13:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d65bf2558330007b68b8ab9409c36ad90a9375b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 14:43:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5464763ed3b749eb8caa05e844e580e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcee58d7532d4ed88038b513b94586f8 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e425e3ae970f45dca17926a7e14df9c9 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/715b2e3942e941ddb17bc7f06d2135da : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e346e69c359452a971d8bd2f8b89693 : SUCCESS in 3m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56acae895ddb4e238f0dc25a088a3ab0 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fb0b61779404d1a9c21213f7c1de4e7 : SUCCESS in 4m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73d351b37843475989c66e429a232ef2 : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8c9ef56ccc3247d6bc177a7e0a8e70dd : SUCCESS in 14m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/95113531b0a9404cb20eeeae5d870b06 : SUCCESS in 9m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eef7c6289cd4463b95f3a7e6322529ea : SUCCESS in 14m 04s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2e72917b5c8048b4840795319119230d : SUCCESS in 8m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/32636a16ae9d45d49442d351fd3de75c : SUCCESS in 5m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2d35e790a0014f079a9e0368860f1813 : SUCCESS in 4m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1be6e93db13a4bfb9de72cecb0b24499 : SUCCESS in 12m 21s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/f4dda14cc261420c9873c84e96657cf9 : SUCCESS in 5m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/397c3078ef2a44c2b103a238e40aa60e : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bded114b55b843f6a5994c2a52175c06 : SUCCESS in 5m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ca97d375459e4274b7183f3ac0d88489 : SUCCESS in 7m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/582ed6cd9cc34fc0965e548f0d73b6e6 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b184127f68c415bb6d642dcbffb6196 : SUCCESS in 6m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0423a0cd252e4d5c9c7b73702e2c41e5 : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9778cdc7a1804f4b91763d1fd2e6cf30 : SUCCESS in 11m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f4e2534cfef8459c891fe2b41927eeac : SUCCESS in 4m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fc49067cd4f743acbc18948b50dc26cc : SUCCESS in 10m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/217a4413767b4e9e86ed728daf2283b7 : FAILURE in 4m 21s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/904623f0172f45d386c70c20fe21429d : SUCCESS in 11m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8b9fe6ab011a4dc19fdced019276874d : SUCCESS in 25m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/08aa4989de2e4eb59f715230941ee896 : SUCCESS in 9m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e265bf2423414f3d891479c8caa13d16 : SUCCESS in 13m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b96d0687ad6d40ef9363f8195fbf04b1 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cbb122f0351484089b9556db1a4a7d7 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/991dbd532b004b26926097a85e74ecde : SUCCESS in 19m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/487b132b6425405890f58a204af6eb32 : SUCCESS in 6m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3594d19d233b4326a8309f43599cea78 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d303c6d471ec4a47be32b9cfc7fcd8f6 : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ff57a0ed8c094017b566142b26f0df0b : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/963d0a6b9e0543d8b7fb16391cbba350 : SUCCESS in 54m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/310ddc8fbf424cf9ab81945915599b79 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f0ba8fd918435f498651a89a36de270645a9fa02","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-21 20:09:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58aa108c94ef2cc3794216f579529a6cb2c1815e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 20:10:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a77069ffd51a7d52bdc46333a5f7601499b9e4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 22:01:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab06c99da6534af3b9fb0b38b21d9de6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa0b606f8774e16935f99bf1650b8c2 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f44374187c45caa9b3e57ec58c64d0 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13af45e406324af3a8c2a04cf838653f : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d480859bdf5441c68102827de5751213 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b23ccf506beb4ecc99660325dc3c6e44 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/350bd26cee4e43e38cb4fec2e2bd251e : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/61125cd1926e4f3e97bbaee3ecc6ada4 : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c53e0c5873504ee0b636671893d43592 : SUCCESS in 15m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0c1f80f2cd9b46e1bc8a00bca7cbe3f9 : SUCCESS in 10m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2674d1bdeeb44bd39413331338e9e5b3 : SUCCESS in 11m 12s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/ebe2fedc377c41d5af53c72d3b1c86d7 : SUCCESS in 4m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a8aca7a1057344fe8d04536dce08f50e : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8dee706b7c774707ad668fd77c4fc6fa : SUCCESS in 4m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1b910f1151884152ac0e69de6796d577 : SUCCESS in 6m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a0bdde3864c54bbf92fae759e96d86ae : SUCCESS in 7m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85091fa537ae49a4bd3c62d5200b66f2 : SUCCESS in 6m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/15d95a43317f4bd9bcaacacbdf66e222 : SUCCESS in 6m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f47451e55cd64fe88a1d2648b790b891 : SUCCESS in 8m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4036488a64664ed08ead60facc81329f : SUCCESS in 4m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/edcb23cc191a49a5927ced5a1e19908b : SUCCESS in 11m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/96314b6a22e44540887ce941ef84b74b : SUCCESS in 19m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/59092a35940d46db90f753dc1c0df8b5 : SUCCESS in 13m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2c9e4e24a60c45a69d2edcd21ec66b81 : SUCCESS in 13m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b111c08716894e5aaddcd45abec168df : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/331cb7a80e40430e8d1b4bad71d89951 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9269ef99cc7a436093f57642f403e3b3 : SUCCESS in 24m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/21d45c1a51024e24b00917c572663ffc : SUCCESS in 5m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d56a48b4937f4bea89e34becd744f9fc : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e61aa8192fb46a5ba93e283286a387f : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7ab1723e34a64dc79969db5a9bb6b201 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8503adce5e3542fdb165811e2832ad33 : SUCCESS in 41m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e99773e3c584e35940cebc7c2d25d4f : SUCCESS in 1h 44m 51s","accounts_in_message":[],"_revision_number":1},{"id":"987e07959ab8ad29195d236ab766911022403e64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 22:01:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4567dc1c9d67077d45bbb595bb7457afc90cabd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 22:02:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab1bad2dfe784e7588bcb60deccd34d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a09ac18bc4487bad8f6aad34317938 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26fe36f720518e62ff884e88ec7967fc8f37d96b","revisions":{"26fe36f720518e62ff884e88ec7967fc8f37d96b":{"kind":"REWORK","_number":1,"created":"2023-11-21 09:30:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/901528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/901528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/901528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/901528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/901528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/901528/1"}}},"commit":{"parents":[{"commit":"4adb587deb67e42bc19fd82649c2183d8429d67c","subject":"Merge \"update constraint for keystoneauth1 to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4adb587deb67e42bc19fd82649c2183d8429d67c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-21 09:30:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-21 09:30:16.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.5.0","message":"update constraint for ovsdbapp to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Change-Id: I411ad6c47b3e5c63ef2762934d19d00483eca83e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic936617a2a70351625ff6137f5a2f7de78d48ad8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26fe36f720518e62ff884e88ec7967fc8f37d96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26fe36f720518e62ff884e88ec7967fc8f37d96b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
