)]}'
{"id":"openstack%2Frequirements~901723","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I0c67abc37f09f7dea8ecfcfdd88b7bc4a01c8e68","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I0c67abc37f09f7dea8ecfcfdd88b7bc4a01c8e68","subject":"update constraint for neutron to new release 20.5.0","status":"MERGED","created":"2023-11-23 09:56:19.000000000","updated":"2023-11-24 00:42:55.000000000","submitted":"2023-11-24 00:42:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901723-new-release","meta_rev_id":"781eadb2d329523136c3d027fe80cd7cc7467235","_number":901723,"virtual_id_number":901723,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-24 00:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-23 14:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-23 22:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-23 22:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-23 10:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-23 11:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-23 14:08:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-23 22:54:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dd776817cbfadef84c47f5215ee5f7cc102f012c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-23 09:56:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"014c3022f536db1f47fb7408362900a1ac8c52cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 10:04:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/789cda12b95f4cb8a427a136303505ea\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/157d5cf2c4ef4f82896494821216cbf2 : FAILURE in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a5d4d81373d44765541316540fe00a5b2663fee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 11:04:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29ebfc9b8b034f66a03b9d004ceffc29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654a4343295747d89c40678f74e17a76 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78b057d81c544c678c6852d15b40ecea : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f766b1629f274798aae3a7227878dd8e : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dae772cfa6274e53b1de862bf50e4937 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67a1dddbef43422d95e2c5d0fc1b15af : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/06419b26bd9c4a26a3df3199dbab5178 : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ecc909b14b3c4423af13b1116dd16c9b : SUCCESS in 8m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e37f4d43b70043b6a7de344f083ad1e6 : SUCCESS in 4m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cd1946bf08f34973b3d23346ed4e72ad : SUCCESS in 9m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8ef42e2bc4ca4f8e995457b2ecaa078d : SUCCESS in 4m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dac08283e3ac488ea9ab7b4aa99cb13e : SUCCESS in 4m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/69ce6319851a45c4aa1a280ea9a4fad8 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/991e0096834c4ed9870a6f6db3257e5d : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/64212bebb9ab49908c919a423bc06b8c : SUCCESS in 6m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/18c9ee9a33804ef0998b8e56dbe55ce9 : SUCCESS in 11m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/97a42392205e40b0a5b835746b0046cd : SUCCESS in 3m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3a8a5d56c88d4c818f2ee04260240ea8 : SUCCESS in 8m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6b4e84ce9d844069aea048b25b1cfa9a : SUCCESS in 4m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9972bb4a2d2e4fb0b6601c43090984fd : SUCCESS in 19m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/859d5475052d43339134db19584a5a2a : SUCCESS in 11m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/40c79c11d75f4da3a63d16bf89496785 : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/caab5803593f49c285acd38b11ba2be9 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d85fa63e804544b8af0a7a2059c758c4 : SUCCESS in 17m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6daf3c6083684cb886b74888df187049 : SUCCESS in 5m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/711fc1c0efe4487ab2e9720e8e44e999 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13c6bedc48db48f097bf478880b7397b : SUCCESS in 5m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ad1b237c28cd41c8add37bc86b2cbf3c : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e48f5e50fd1243e0a4c028c9069fffb8 : SUCCESS in 33m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e7d388cb4d14586af8378c5f82071a8 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a1c2debc330e6acf957d52c8aa31a1fbc741a748","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-23 14:08:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bbe4b2b5e676d47bbe792824342b0986e3abedad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-23 22:54:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4f0c44bbf94c45753ae2ef09c649f01ffa95aee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 22:55:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d63987ea519e9c70465b4fac97eb6d187854810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 00:42:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd6597fb7b049788447f2ee3f4a474f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b912d45cca496d9d1bf3d0de1de8f4 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d204bdb969e474d874bd1155708b140 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5984feec7e14e6f8c8eb4014749a424 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0cc37ce494b4fc5a9c3c9ed2b4ea117 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5bda645917ab40a8a4d60f4cb310d79c : SUCCESS in 2m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fdc7eb686e5b4c53952da560bb9773cd : SUCCESS in 9m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/342bfdfad6d84bb9bc9a5d288f24416d : SUCCESS in 9m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0dc672598a094b79996cb82c0b6ab880 : SUCCESS in 13m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e9f90da35add47a89b599cef7b4cefcd : SUCCESS in 8m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ca51df5edcd04f51b25527604f97e0f5 : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bc21ac6ba0f44a0b9c415d2cb36d9432 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e91ca8b4165f4070bbcea042d778f2a0 : SUCCESS in 13m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c379579cefce4ca5a57a313d8984baaf : SUCCESS in 9m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cd9912f71af046e4abff2bbd01bd31fc : SUCCESS in 12m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d9103531808841b7ba022bb3e436f6af : SUCCESS in 7m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a8c97b37889b4825a980314ea3150232 : SUCCESS in 11m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1f7f9e5ad415484592d3caa9318dc593 : SUCCESS in 6m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/63ff8d4611444ed9a41a0b17abcf6ddd : SUCCESS in 16m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/69ade4dd50624c96a744e3cb1d92ef3a : SUCCESS in 10m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/be4c71241e664db5a4aeec5fb278f636 : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5fe286719d34a6d836aebdcf3a8c218 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18aeac0ef08c43b8b170b5d835c98188 : SUCCESS in 15m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dcf3febe78914eb6b1e91fbf3baa3f91 : SUCCESS in 4m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/71d8eebc59b145b0885470272f6d6442 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f053e6cabe9462eace0bde5c7d4a014 : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/49c483dce4334d18b18aae506ebf4b49 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7535207291d46709ee89f7ba1d3cdf6 : SUCCESS in 31m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37b3ff2727e2473995af9c40016d52df : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"781eadb2d329523136c3d027fe80cd7cc7467235","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 00:42:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae90b95340215bbf376a02d9be879e5e0b534637","revisions":{"ae90b95340215bbf376a02d9be879e5e0b534637":{"kind":"REWORK","_number":1,"created":"2023-11-23 09:56:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/901723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/901723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/901723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/901723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/901723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/901723/1"}}},"commit":{"parents":[{"commit":"a08cef69adf7250f8408ecf3d06058b31667018f","subject":"update constraint for oslo.messaging to new release 12.13.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a08cef69adf7250f8408ecf3d06058b31667018f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-23 09:56:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-23 09:56:17.000000000","tz":0},"subject":"update constraint for neutron to new release 20.5.0","message":"update constraint for neutron to new release 20.5.0\n\nmeta: version: 20.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Corey Bryant \u003ccorey.bryant@canonical.com\u003e\nmeta: release:Commit: Corey Bryant \u003ccorey.bryant@canonical.com\u003e\nmeta: release:Change-Id: Iffcc45f4c1a9fe34e4f8918870236b0c1dff2a85\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I0c67abc37f09f7dea8ecfcfdd88b7bc4a01c8e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae90b95340215bbf376a02d9be879e5e0b534637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae90b95340215bbf376a02d9be879e5e0b534637"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
