)]}'
{"id":"openstack%2Frequirements~901905","triplet_id":"openstack%2Frequirements~master~Ia54cad9a7713924de9dc29e91c39044e04a70ba3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia54cad9a7713924de9dc29e91c39044e04a70ba3","subject":"update constraint for os-collect-config to new release 13.2.0","status":"MERGED","created":"2023-11-27 10:58:11.000000000","updated":"2023-11-27 20:09:47.000000000","submitted":"2023-11-27 20:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901905-new-release","meta_rev_id":"541c6cd521ab76dd622c4673c2bc5ff24e782147","_number":901905,"virtual_id_number":901905,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-27 20:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-11-27 14:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-11-27 17:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-11-27 17:23:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 12:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 14:35:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-27 17:23:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7030cac9bc8e0137973e9877ea034aa9bcb917a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-11-27 10:58:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0684e4815590cc05741f54b10346609825568811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 12:16:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/731e8a5f509c4ee6b542dcc56294e4d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee7ae87aa10d40ab9e387c6ae1914a38 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde5544f05514b548a355ab2165e34ab : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f10de88a22c4c5593f8ebe1267d5af0 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb5fcfb8f39e48c2b67da56bbeb221e2 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7db327dd77cc45a3b0e805905d94b3dd : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a6e7b1661a049e6a84c9ba74e1b0fa7 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b1ad7298a134aa784ad1c720131db38 : SUCCESS in 9m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/009b0c85ab4f419290842a5c4bad9a5a : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/761d4419e6594eaabd44a9df94ea4c4e : SUCCESS in 8m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/569efc323e2a47afab91f27cacd0cdd6 : SUCCESS in 16m 31s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/be78a008a73143aba202437cdd3096c3 : SUCCESS in 8m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/47e5445fbd2344288a356dd0031ae4c0 : SUCCESS in 4m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9124fc5b187b43cd8afd4a537c9bb251 : SUCCESS in 5m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8a939c05ec3242c8aa13c5908cb9f653 : SUCCESS in 10m 36s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/805e64f50d934cb1a19bd9ea0535deef : SUCCESS in 5m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/53f3840fd2844d1a90c5dd74bdd0647a : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ab2c8cde3092471aa6d6515dba21b2ca : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cf8526113af44e32985c7362f4973d68 : SUCCESS in 6m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fdf9a465bfe9474db076fb0b53587cd9 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b3727d56fc0409ab66a8788001a3304 : SUCCESS in 8m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/78fa75e425954dcda7ac6de051c78dc0 : SUCCESS in 8m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3801366354924b4b8923efaeb44352ae : SUCCESS in 13m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2ac822276e8d4f85b6c2ddd3c4c4bbef : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b656b3aa93eb41bbb9e8446371501f75 : SUCCESS in 9m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/db8d5ab7b1bf47759ec0f8f1525d43d9 : FAILURE in 4m 04s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/342b151b90244ceabd505dd076a21ba0 : SUCCESS in 11m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/41706ae293254a9da5e2228a756e18a4 : SUCCESS in 22m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2f9d80a4a647460c964e27d7e57f0057 : SUCCESS in 11m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/15736dbece9c4b74b2121cc7dd6e5784 : SUCCESS in 13m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/72e81f229f2e4e76b17dbfc2b0374e04 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7125e7ed06c4f4782409b536672a863 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b7c4ef93f194f01a5f3a574a0b8e626 : SUCCESS in 21m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/47a3df6b9f0e482e8b59e4c141203680 : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/564a0f5401c04a5bb6743ea376e49589 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9694ade6f57544e8bca03bca07222e21 : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0565be31145346ecb924e35559f7a056 : SUCCESS in 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/805c89fafb2240a48861bbcb343440ed : SUCCESS in 49m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c16a0a876b354a828ec60ca0da3b197b : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e425cf59ffe16b60031902cb24f39af281c1e3c1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-11-27 14:35:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26fcf4dfeb76b606e2800d3b3cfc8c41a8b05143","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-11-27 17:23:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f46494b56b4f590344c9c05f835a8324b83b8154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:24:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27a7aafe0f553dd42b482c6a090aaae84275fadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 20:07:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/006f707ca51a45ebb13b907a796e3430\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a12f55289fd4657b3c9b503b1f3e287 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8440ee15044728a0c944557d47bb44 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7414fa0c592a44a6bf66f1a47eefa7e1 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e59e6c4cb64c4a2a87a1af6f99ee4003 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/596505f35e84408bbb37469be6013757 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72cb23ed448648268fe705b2be0a617f : SUCCESS in 4m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/402e5dc663aa4fbfaeacdf20d5bbc70f : SUCCESS in 9m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/88ffb404f4154f35a45fa57b1d85a561 : SUCCESS in 15m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0295074389ac490dbde8bcb59141e825 : SUCCESS in 9m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/581dc127919f490283ed3858a46f89b8 : SUCCESS in 10m 35s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/58ce989360694d75a0d382be157d6c31 : SUCCESS in 6m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ded5ae9e8db342d89b33ce6ab322c78b : SUCCESS in 5m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ba9de6a5f3804826b517a251090171ef : SUCCESS in 5m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0d3e6823dfe841489c9d74d324539a22 : SUCCESS in 7m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d0d32032b1354de988194804c076b9d6 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/beb28830f1c743d3b9a15b3cd65b65b5 : SUCCESS in 7m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f9a00ea9b8744a5696d5f6da3c649b13 : SUCCESS in 11m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/085bd19f5b3b4d76a105ea4a798840b2 : SUCCESS in 12m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/314feef2278a4d26a91045b092ab5fa4 : SUCCESS in 5m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/11572adde2434fd6aba8431a4d324759 : SUCCESS in 9m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ea31992add844035b0498eb5195cbdc1 : SUCCESS in 31m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b2e7f790f9bc47afb0d37004a83bb237 : SUCCESS in 10m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2c701e8ddfe44b899fa3a825e221c287 : SUCCESS in 12m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/010af0abe1234f6fa3b3123b8fb95c24 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2ba681c2f5d8433ab52e39faf0110711 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1881c81769c4b67a0a10b2bd0652ca1 : SUCCESS in 24m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/142cf52206bb44dba34f1d84e5287187 : SUCCESS in 5m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/efc8ce150b1d4ed5b4555889c65857ea : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93c0e020adc143049f11d296180fa339 : SUCCESS in 6m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/52652d656c704124867f13ecd93b9f63 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b256fca45a4493087b7e2a1d3c15661 : SUCCESS in 1h 06m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6ec719dadbd44fe810ec3276d09a6e4 : SUCCESS in 1h 39m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7cb9b8c483573a64c9d2841533f1a099ac08ec2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 20:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"541c6cd521ab76dd622c4673c2bc5ff24e782147","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 20:09:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd2720e197cb450db63ab8d4daebc878\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3adedeca7c2d491da3abd747e5c10eb0 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"860387f2b12f6fbc7595ef16ec171790bfb50ce0","revisions":{"860387f2b12f6fbc7595ef16ec171790bfb50ce0":{"kind":"REWORK","_number":1,"created":"2023-11-27 10:58:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/901905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/901905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/901905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/901905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/901905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/901905/1"}}},"commit":{"parents":[{"commit":"b2f50383a06dde2c4a0bf0313639b50519578b12","subject":"update constraint for sushy to new release 4.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2f50383a06dde2c4a0bf0313639b50519578b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-27 10:58:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-11-27 10:58:09.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 13.2.0","message":"update constraint for os-collect-config to new release 13.2.0\n\nmeta: version: 13.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Id5519196cb8e712363962f6248b653c7709f4cc4\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia54cad9a7713924de9dc29e91c39044e04a70ba3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/860387f2b12f6fbc7595ef16ec171790bfb50ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/860387f2b12f6fbc7595ef16ec171790bfb50ce0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
