)]}'
{"id":"openstack%2Frequirements~903070","triplet_id":"openstack%2Frequirements~master~I58da055feaba988388e795186de1fadc8ba143f3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I58da055feaba988388e795186de1fadc8ba143f3","subject":"update constraint for python-openstackclient to new release 6.4.0","status":"MERGED","created":"2023-12-07 11:21:50.000000000","updated":"2023-12-07 18:53:16.000000000","submitted":"2023-12-07 18:52:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903070-new-release","meta_rev_id":"5477f49fb5bdcf0f5e311e9bbef95ef04aeb5933","_number":903070,"virtual_id_number":903070,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 18:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-12-07 12:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-12-07 15:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-12-07 15:31:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-07 12:31:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-12-07 13:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 15:31:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b3a05314e895af731249026c71a3de0a0693cc5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-12-07 11:21:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad2f68857792b674953d0584eb5d1bce6a365ff6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-12-07 12:31:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6dfc9dc5c1d0dc4e2323b4a75f2cba628cc9e450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 13:18:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a556ec1e698445392dad5c9cbb0d318\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84c3c6f23d54d8e8c27c938dd32428b : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4d20394c8542949ef5bf89b2b22749 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a97a6179c184334b60d2ff68582680b : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0a015895b434d409f12b7d98f675333 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2eb07ecd416b44c0b85ed1e66677709d : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9d156d2244e457399a9b0a599d4f149 : SUCCESS in 4m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2779bc7313d54b77be3a102e07e4965c : SUCCESS in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/597e063edf894d598538a0c606ad15ca : SUCCESS in 14m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fdc79ed1a05948faa43da45825c70340 : SUCCESS in 10m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5b5a2fbda144483dbb8f1e4b78473bfc : SUCCESS in 15m 56s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/df5f6e6639074de49242eb85402b16f7 : SUCCESS in 5m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e343fa03d68245ca979478a9a930f6ff : SUCCESS in 4m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3df1a5a699174d8e8a4c0ae4dbc5444d : SUCCESS in 4m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3d5eece5abe744249788660ddc81bd1c : SUCCESS in 11m 22s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/b33be6f8c26d4d21ab0ef18c20459c36 : SUCCESS in 5m 53s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/97e6f292b0f24a92aa9022621560c9a2 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7ba629f54623428998480602a8ec7f84 : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/14924a3126be44f9bbefcbfe7cd79339 : SUCCESS in 8m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7ba0c5e5a27d456fad10dbb12e078211 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6d67fe52f8048a2ac0c0f8e63ddde82 : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c7fd25377b824174aca9aa79e5c9ea51 : SUCCESS in 7m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d28206eaa5dc47688d7a546c4a256a3a : SUCCESS in 12m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/40c1442db8164c41bae99ae3bd42a28b : SUCCESS in 4m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b83d2b641455486aaf706e3d24391d97 : SUCCESS in 10m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cb48f0ed9b424ae1ac8da4bd8cac6545 : FAILURE in 4m 11s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0de2c9f9b5c04e5fa41260df9efba4ca : SUCCESS in 12m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cdac44c66dcf4205af40b8951b24db77 : SUCCESS in 25m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c7a39f28302448d690c812eb81e8608b : SUCCESS in 10m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/939b23b1c5fa4883abf0e46b4a686440 : SUCCESS in 13m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dc2d60271b0d4182a0f1b80da5e8bfc6 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ce8304f8a314f8cb8b69f4adcdb28a4 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd80650803fe4a82949262dc66a1cbef : SUCCESS in 21m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/51417c3055144bf59b37e37e4d29fce2 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/11360ac003804d2cbd6484fa53cdbfaf : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/860ffb1736cc463aa42314db493e4efc : SUCCESS in 6m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/04d39c154c8b4ee085a2e8ec8262a817 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/832675313d0a4004af3e25b8ada0ccf4 : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2c18a1a6b6e4053add86f3fe34454eb : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2e5a3ab4d5e9d5cbf555c8e5743d019a6df11abe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-07 15:31:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a61233304e5114f826bc161fe1e3935f89e86fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:33:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4d6f27cc396005f714c342e8c9073b3979ebf89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:52:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7267219a9d4f4614bf05ac1319dee1da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c87a9831ea4846b12ae6c1f99d16d9 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb0664f102d46d69c7ca85d7d8fc4cc : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59aa078aaf59469d80b1bab85c4920c5 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee4c3f6beec24d9ebe2259f61f07bb82 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ed425e3d40b46dea94274048c92e10d : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/380cecc3dac4497e9fd6cba29140a6f2 : SUCCESS in 4m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e22009e6f52f49d796a72a2499fde650 : SUCCESS in 21m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/82638958d745479d91232ba17e8244b0 : SUCCESS in 13m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bbad786b00674d688597befd08f6f526 : SUCCESS in 10m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5a7c9cca692d42bcb30ce84b85fdc422 : SUCCESS in 13m 04s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/3ab97a60fceb49a79657d980a406c0cd : SUCCESS in 5m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b0f3cdc996344f5ca73f108e0726447d : SUCCESS in 7m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7c0f4f1b81934aea9ebc3116399df139 : SUCCESS in 6m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7a09dbc1ba80458eb66a9c45a49560b6 : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ae3ccfb0367d47a3a1b2f7eb1a398f21 : SUCCESS in 13m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5efe7f5041c4e47993b84964ce37648 : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/70d7e2d3b8aa4176b04b445b5c243ef1 : SUCCESS in 9m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/96ebf8f6b08b49bc9d79ee72e41832c2 : SUCCESS in 13m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/25b7e310626348159f04009c9c1ce664 : SUCCESS in 4m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c3eef916c56f4279be274905d8a8b006 : SUCCESS in 13m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7b61cdd1b88c4e82ae45826f1bb6b5a8 : SUCCESS in 29m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/86aa1aad89a547758b745eff639e91bd : SUCCESS in 13m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ade3186c66184138b43ca791ee6fdeef : SUCCESS in 13m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3be5140739e746758baf3158b85237c5 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aebad31945ac4440917263e5ec762901 : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d0742b137ad6443dadb034dbf8fd065f : SUCCESS in 24m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/75584b501f674ec4bacf62d178ffd9b3 : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7e92beb8d3894d09a16bcd975283ead2 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/868ef9f6d8ce4109b828d4c774afeebe : SUCCESS in 8m 05s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/651f5253af164fd192923cbc3f9fe72c : SUCCESS in 16m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/826618103cf745f7bf0b2af81da39cad : SUCCESS in 59m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c855322a58a43babb88d787aa56124e : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d981f3eb2dc123e731695efbe49e19f04e826ca6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:52:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5477f49fb5bdcf0f5e311e9bbef95ef04aeb5933","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:53:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be0dfd912da401ab69a218dcc6567c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cde29bfda6745ebaf821dff42c94749 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74ba73e0538b52b2371ecb18621808376ad5d96f","revisions":{"74ba73e0538b52b2371ecb18621808376ad5d96f":{"kind":"REWORK","_number":1,"created":"2023-12-07 11:21:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/903070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/903070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/903070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/903070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/903070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/903070/1"}}},"commit":{"parents":[{"commit":"53181f74f3641a3a48ca0a164147aa63161f06cd","subject":"Merge \"update constraint for os-collect-config to new release 13.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53181f74f3641a3a48ca0a164147aa63161f06cd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-12-07 11:21:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-12-07 11:21:49.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 6.4.0","message":"update constraint for python-openstackclient to new release 6.4.0\n\nmeta: version: 6.4.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I59d49551d2579833b28d5c0b98c9d436ed498cfd\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I58da055feaba988388e795186de1fadc8ba143f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74ba73e0538b52b2371ecb18621808376ad5d96f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74ba73e0538b52b2371ecb18621808376ad5d96f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
