)]}'
{"id":"openstack%2Frequirements~904147","triplet_id":"openstack%2Frequirements~master~Ib2e59a207b86ae90fa391bf1dff7819851dc9c9b","project":"openstack/requirements","branch":"master","topic":"eventlet-update","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-01-12 14:01:07.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-16 11:03:05.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-16 12:42:47.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-01-16 12:42:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib2e59a207b86ae90fa391bf1dff7819851dc9c9b","subject":"Update eventlet to 0.34.3","status":"MERGED","created":"2023-12-20 19:06:14.000000000","updated":"2024-01-17 18:58:41.000000000","submitted":"2024-01-16 12:42:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"904147-eventlet-update","meta_rev_id":"b2bdc84b1fedc6fd543c283291cce621465172bf","_number":904147,"virtual_id_number":904147,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-16 12:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-01-12 14:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-01-16 11:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-01-16 11:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-20 19:14:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-12-20 20:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-20 21:16:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-12-20 21:16:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-12-22 08:48:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-12-22 15:21:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-01-11 07:47:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-01-12 14:08:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a2633378438616a610cdc0c2dfbd23c0009b308a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-12-20 19:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3af24279aab6abf344ba7ec185aa26e44da9f167","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-12-20 19:14:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9087a3bb11fd09c2b9146bd42df6858998177b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 20:27:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ae139e8b4754cacbbce7e861684cd27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee38c38086234ed8ba0290f6fdf3d399 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc6b0ac56ad455f9b8d01a922ceac34 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e658b54e5a84cf98bfc28c94c27c84e : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90f1f3f20b6b4b0e86d3f962356966a4 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34cb0908563f4e1f80af5aac71c19f07 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ea17bb63c7f41b193f900b7c6d2dda3 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8a2eaffa988c4e5fb2d98ef9ca3472d1 : SUCCESS in 10m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/52c85093038b4178ab6fc0901f5314c5 : SUCCESS in 16m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3094b2abe82d442ca64aa388b83242a6 : SUCCESS in 7m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/babc39c32f374d3ebbefa5bf1d54407e : SUCCESS in 14m 49s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/cc99799956f44db68de826000f96b4a2 : SUCCESS in 5m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8d9287d913e7478381464c0acf45be8a : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ba014c6f16794aef8fbaec7e5c591968 : SUCCESS in 4m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/704b225ba50c42818594496694c2765f : SUCCESS in 9m 29s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/982a55521de84293bff84537999060b6 : SUCCESS in 4m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dd6c63c00b1e4288939751761b8a9951 : SUCCESS in 5m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/da4bf817ddd3447a94456f015985ca42 : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2fb03ce3ae024419ad872155bd406f7c : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d8956864110d440da32cbf084df27dc0 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/276722a636b64048bb3d0a3b1b660341 : SUCCESS in 9m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5269e4bac873475186945881dcb703d9 : SUCCESS in 8m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dc6707b9452942af8bccb93f8c1924f1 : SUCCESS in 13m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c101a9e95c9f4e0fa1e237bf1772d776 : SUCCESS in 4m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5d26cf31a7c14f48bcd39dfd97fad362 : SUCCESS in 10m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7a192617271843e599f6ff19672bbaff : FAILURE in 3m 53s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/778f25a1148e44979d949799c18b447e : SUCCESS in 14m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d51429c2bd444cbaa63618a77027582a : SUCCESS in 24m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9a3a3f9f96494a5d983e7780e488633f : SUCCESS in 6m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/96776764f2ce445185096ad6d5db93e1 : SUCCESS in 11m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/21ff3cd094a744aead274d3872c2fd6b : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bb8d597c82f947b7bde322286127e076 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e92ba82182e48e180dcdfa1c7647544 : SUCCESS in 19m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0cbaff845fb54958a844d10f2cf20d98 : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5066babd0eb3401e8027556f4941b734 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7808b7939dfc40ffab3473e146c1800b : SUCCESS in 5m 41s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2c07f5f576134790a89097f0511f3b6b : FAILURE in 8m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82128668bf674219b5834687150109af : SUCCESS in 41m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/223105974602468ab673dd765fa90bb4 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"34ede4c50351653a3b1bf14805c0e141dd2618f5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-20 21:16:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b705c263d1f886fda846866353dad1e2d1eb0b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-12-22 15:01:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b1e5db845fdb257321ff2b9c8fbe6ffc5ec89b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 16:58:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fba968820e4e4612ba31026d3ef06db7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06e84babf3d4db58eef2ddbc2151452 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1781dd14a50a4cd3adda03aeaf523d8c : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94067de080d14b3f90c5b9f62e5368d4 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/331d5e9dffda4e1a8fc5ff043b3f2f9e : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/140c636637f64eafbf226b2dd47fbf7c : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40cc17520de34ec28a8859fb69b749b7 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/059bb097fb694e65b1338774e0666d1c : SUCCESS in 7m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d9a0141ab2554896814ac7cf7920f327 : SUCCESS in 14m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/912cf821580f4b07b7ea69f486b404f9 : SUCCESS in 7m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/683da2c8a4584166bd6c94a22de2e1d3 : SUCCESS in 21m 14s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f7eb6ac5db7243e6b6e9ad5c57b66349 : SUCCESS in 4m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ef2f27bacdda458a80cc52467158feb9 : SUCCESS in 4m 25s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/19aff48f4a854341bcf9566d1caa078f : SUCCESS in 4m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/82bd57363c9042489d5bbdecaeb5a9be : SUCCESS in 10m 42s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/d7b98798e6a94fc4a782b38d12131512 : SUCCESS in 5m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3326313f74f543cf912391db358725b8 : SUCCESS in 5m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c200dc84e4544750908c9d6856095baf : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b9eb17c55ce7443eb820225d18b773da : SUCCESS in 8m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2bfbafb761034d64acb3d7b0ff55b339 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/884bdce5de404f24b7f5be0be90a48bf : SUCCESS in 7m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f4949755a9654302a977e08d9e6ca777 : SUCCESS in 9m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c6d6b05651f14339bebb778d4ac4b5f5 : SUCCESS in 11m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/899ef2df6cc04e4a98634d550be7692d : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/dad1558bed2e468193fdc1a3f0098001 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/60eca983b0c245d5ac38e24a5f3a3df5 : SUCCESS in 4m 45s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/331477c8b2b342bb93b0dcda9851942c : SUCCESS in 10m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/251bf8b7a3d64d7dae19ba231d658fb9 : SUCCESS in 26m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8486d866da144202ba1de283b739b0c8 : SUCCESS in 6m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f232a53d8bb5498d8cc6370ac352dc58 : SUCCESS in 16m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1dc8ffd08daa4caeac6a4ee1417698e2 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d094611986f4a7a9e8106e33bb77236 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f227e009cd24700941dcd28f321739a : SUCCESS in 20m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/df01836ee67547779c0f749d09bbd5eb : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5eb8699df5bb49b5b71b12f6213c5cee : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b14707a96d54a389c659778d5d986ba : SUCCESS in 6m 36s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/80b59c91f7924d1d854a76953aa20a08 : FAILURE in 6m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03192e09c5ae438aa1c5194ff0302cd3 : SUCCESS in 1h 07m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b2f71dbb0ae4def8df811f817e4f4b5 : SUCCESS in 1h 48m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ae3d565a3a3287ef73fef6ebb2603c381dd97092","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-12-29 09:00:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99394f1c93e5d89efcaac6600e7d9810842ba6ef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-02 17:16:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26a6315e91add1846f7b8cc2cc2869a584c12bde","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-03 21:22:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af17c67f5786e4eac36eb0ed7d48b8384f3d7223","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-10 17:03:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83023647f64957f48081b1c0b0ed55d7810a8ee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-11 15:42:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"12f1afc0d47334d9d3009b93f59b8703c4809954","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-11 15:43:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0b597105aaf31dda4c899d47dc3aa823c757ce18","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-11 15:44:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a32249e9b9d897cefb302ee1fa5c1afe2dd1a72","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-11 15:57:21.000000000","message":"Topic eventlet0.34.1 removed","accounts_in_message":[],"_revision_number":3},{"id":"663d27c27a7515cc6888e2edb78e03fcd5bca589","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-11 15:57:37.000000000","message":"Topic set to eventlet-update","accounts_in_message":[],"_revision_number":3},{"id":"38d2503ce239971b0013a1d93619b9a047aff6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:40:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e43341130f534d339440eb8245a7a578\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071b1dca1094422a88b8d70cd8f8d841 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b4ed9143db94402be5523572bb3395a : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd416cefb8f24ecf83b0800eb764ab92 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e35aeaa0db54f319efd523af8b75ffa : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c594cec6db624676890f3cbea848f0da : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2e794f21f82405fb50a469ec26803e5 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0921b58a41e04721b658a304c52ae375 : FAILURE in 2m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e6de235df9b5445a9a7b316bf8acd8ed : SUCCESS in 16m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/14287b2f90504c31b4f0428c2243ffb5 : SUCCESS in 10m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e29a72f726e5410a9ac5b8c0a522329f : SUCCESS in 17m 30s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bac1a7a682f74c70aac9228a8ad49ec5 : SUCCESS in 5m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b6e82a19b2e64fd7910beb904be61d59 : SUCCESS in 4m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c31d6db4cf66421b9917214e6e9db15d : SUCCESS in 5m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ec0fcdc1db124af3958b6d90255fc1a4 : SUCCESS in 11m 50s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/de4f5a35990b42109cae52ce0703e0ac : SUCCESS in 5m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9b296891e5df48db864633c2e9305e9a : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a16811ad36714eafb500068397e54c5f : SUCCESS in 5m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/02be4b753af34e1b9e88dd46f86ae6ec : SUCCESS in 8m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f8523c66f3a4434da58db40e861e2f80 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd475da441524814b585e9c5cf5891ed : SUCCESS in 7m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/44bd361e58f84cd7a25c1bd6e77ae9a9 : SUCCESS in 9m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3b6cc4378fb14bbcadf70c68c1377c02 : SUCCESS in 12m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1f383f49b51a4e7b8f31a7e56133287c : SUCCESS in 3m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/33de2a6e24b74c08a39231cd24e7d65a : SUCCESS in 10m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/43ce669b7ef34eeb87d013811aed63f1 : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5d0efd58573f431490274f142c18d02b : SUCCESS in 10m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fd060ef645714eeb9e808b815d460fa3 : SUCCESS in 24m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fda3476013a0443f88da90e3333ecd78 : SUCCESS in 7m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/89bcf920348743bcafff5284b96fb514 : SUCCESS in 17m 30s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/69e13cf8cd494de48e8f8f21928c4799 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7ea42c9d7a184b05b37173bbecae4b24 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd42ef7b7b99442384f106f99f2b8aa2 : SUCCESS in 26m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/99b964ff483f4917a2fa375b89da2621 : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7d820c2efec644b4b4af6ce7a323d010 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eebdad4a311444dea0d0532a93320d40 : SUCCESS in 6m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5c004a93a57b4517a528d1a99f49c97f : SUCCESS in 9m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08a593100a944414ad771d92872e3da4 : SUCCESS in 42m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac5449d7d7704ab289644432b9eeb4f1 : SUCCESS in 1h 51m 27s","accounts_in_message":[],"_revision_number":3},{"id":"287320094115668148baa162436051d5831f610f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-11 19:49:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0605f837ca0a11fca7d707d22a6c3d00cf29b8ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 21:23:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/460e4455489c48f09b29d6b5d6d6fdcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268466a533554ff19f4981d9acb661a4 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e47554923c47c2b5d6a0651d6e36a6 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c1f63ecfc844ebcb35253c7ceca0867 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/978dbc7f6c90433ba2ed71785a863ef8 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d6555ae77a24d29a54d16c98aac16f5 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b50ecbe7dae474c893db62e18ca590e : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/745d2d456725417f809f366a28bd8af1 : SUCCESS in 9m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b39345528fa544dfa2a4dc126470028d : SUCCESS in 14m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fbb0e6855f364116aad713b819f7da87 : SUCCESS in 9m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/71fbaa7001e649bbb6c008151bcf81b7 : SUCCESS in 14m 38s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f1d7c7641c0749f4abff3ecb928cf504 : SUCCESS in 4m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2c0191e434324cb6bf62ee886aedad69 : SUCCESS in 5m 17s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3b04d966a05a433e993b9437746583dd : SUCCESS in 4m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/66359624868c41bf84b6c7581462705b : SUCCESS in 9m 34s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/7c7bfa13e3fc449da1177ae1d60d57b6 : SUCCESS in 4m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/534a44e4aff64121ab07fc636e9f8de8 : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/75063c2f3b574de4a20fccf5f32be23e : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/957fe5b670cc4c458f9e34e3dd4ed86f : SUCCESS in 8m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/94048f3ba6c3440285723db7d45d75f7 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44d1416d56bb4c29adee3abb50891104 : SUCCESS in 7m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/253b9237fb0d43e9a112c4791b30dddb : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b7d82c8a79674152b06a282884e7d3e2 : SUCCESS in 13m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d1da54291c4d480b801c3167cdfbc953 : SUCCESS in 4m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/45fedd92df95442380f3f6057d7c8258 : SUCCESS in 7m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/43ecce6af0ea43e98c0d6655ceea9c89 : SUCCESS in 6m 36s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6a52aa3f747d46ada01a3e031c8bc194 : SUCCESS in 10m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b6430d226cb845f0b001ec87b3b1534e : SUCCESS in 34m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/21838962b64844859a6f93ef9af7d50c : SUCCESS in 6m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/aca6a438e89b4b6e9b1c1c454c52acf6 : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0f102c2edf514254b59c06bfe29f6b81 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9241ee7aad74f219d2d81842043e9b6 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d6005468717427e87c83d6123f6ab4a : SUCCESS in 21m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2448e10ceabd4e499c88aec2fc74a778 : SUCCESS in 5m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b1c63824d35e43efb0297a1b2e819666 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce6b6ff2a36446b3bf193831240999ea : SUCCESS in 7m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6e7afafdec924909a2fa7a3b25883b4c : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccbea1ade3054c8db8a56efbcf45051b : SUCCESS in 38m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1376f3a2c1254745b337106b9aac81f0 : SUCCESS in 1h 27m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0e6943aee5e9e656e4f7019f4e282198c835065f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-12 08:46:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d58e5b2f32f6962ce56f275f3735e0fcf071e31","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-01-12 14:01:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"23ce4c5fd53c5c6981c129d12a3f3571c1f010e1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-12 14:08:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4c21f72535d6529f6e2c90da01702cf83a61849a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 14:08:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"63d1f84d1a455e9d4d156b796140ba5d98166718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 16:13:35.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5202342de7b4eea8ea2042c2cc6d465\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a61eb0a7c364b36865dc428f0c2db12 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c186672834a44e4a90b2dfea62ccf8e4 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/972648debf93466db4456d2567cf773c : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fe67549ce334c34ba9ca58457976166 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d92bd9ff23344745ac78a869b31fa7ff : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ed3729e287f424f8c3d03cd52373a70 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eaa08bddb0bc444e91f1ea3c66ebc045 : SUCCESS in 9m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c608d71a2b9546d4b68314721c8048f0 : SUCCESS in 15m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1ea04f84a5934d28aa8a7f5895a76a9f : SUCCESS in 9m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a71dbaf910284a2582024626fc1f225d : FAILURE in 11m 09s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/e190914e8e394c93a60cd74f0286f066 : SUCCESS in 5m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7fa840ad538341cd8ea48a81d67e006f : SUCCESS in 5m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/29de7812b25f47c49647907fe323b8f0 : SUCCESS in 5m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f959c4b4d64b47049e468bf8dfe416e0 : SUCCESS in 8m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1a925b56ab704cce9654680845d88a22 : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/131e56ff554c4b38ba231aefb089aec1 : SUCCESS in 7m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/686958d513e74bdd920116d28d41d2b9 : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/867e48f3b5b84956b759ec53bfedf5d1 : SUCCESS in 10m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ab135a1de8ec43abb1bcf2b98efc04a7 : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0819bc0c0f864d06a58ccffa4b4af5d9 : SUCCESS in 9m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7edab3db7caa4b6e968855592a505dd5 : SUCCESS in 23m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e819df18ca34a24b6afdda04931992a : SUCCESS in 6m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/30a7c3a4c58e4844b5cbdf2e8e4e327b : SUCCESS in 14m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/56ea8ea2d37449ccaa0045de3b72e916 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd0731dc78a243f68816302dfdee868d : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9cef88d4236d46dba5aa69922518f22f : SUCCESS in 24m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3962b7700d5646429b513de0e95fa0d0 : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/464c2b27b298465b9dcbfa0472867e52 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31731395a02547ae821563dca5cc5baf : SUCCESS in 6m 36s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4a38b1c04cf9483486e056c321f00a7b : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89336715e2e84dfa8e421db15578af2a : SUCCESS in 37m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/401f99c5077e4cfcb9e4ef5cd8534588 : SUCCESS in 1h 44m 36s","accounts_in_message":[],"_revision_number":3},{"id":"0b9cd990a1c1ac1205d78486cb60ffe19c4e1a30","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-12 16:43:46.000000000","message":"Patch Set 3: -Workflow Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f2be51d40b6cfa04794bedd8d3e71a1f4e43cdf","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-12 16:43:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb27fe27f7dc1da3e10987bb883b416b673f48a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:26:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb2acf1dd30499ebc3841bc478ea5f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2619028aa04f495db760a538b7b645ad : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1534c77f536e43b69b2599e9e7f88080 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/414f4475daa34d0fb34038dccf107699 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d74067708474b1eb98c728a52de199f : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/225413346b264aa3b0d1f4570ee1ddd8 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b1a27d00758436f96f04ab1223804c4 : SUCCESS in 4m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/07d42f5674db4938a60c690c019b2dc4 : SUCCESS in 16m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7ac5c03b2138458f95839c0f7b32d659 : SUCCESS in 15m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/81b48b50199b4edea07d4bd8eaa64219 : SUCCESS in 9m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8966ec82996342798004d60662354ea8 : SUCCESS in 19m 05s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8e0bc9c412e343c0a2f1c5e2164ee4fe : SUCCESS in 5m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/dc4fc37987df4aab8f087b52def5885e : SUCCESS in 4m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/99ee4ea1c6844c73be2ee21e41e04334 : SUCCESS in 6m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4be8d5c25d054c928deba71b8cf58100 : SUCCESS in 11m 54s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/e647c55cfeba44e196913a40851ced4b : SUCCESS in 5m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bf4f5ffd865948c9b6b9d4b7d52faff7 : SUCCESS in 5m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5994abd67a174b8bacf4cd17f2ca7b26 : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2ad15266b052499b8645e3aac6150f85 : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bd11ec1e8fef4f7785caba46b3a1cf5d : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ae3616b4c5c45e08c23d8359d5c8f06 : SUCCESS in 8m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1f755d876fdf421aaa1480b10dcccc72 : SUCCESS in 10m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/56415bf82e4b44a492943473c1116f88 : SUCCESS in 15m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7046088d9e41466a8f329e33a870e132 : SUCCESS in 4m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/623f7c73d121410082d5a8ff18a765da : SUCCESS in 10m 55s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ae0ff0af15874f1d8db3847fed96144d : SUCCESS in 5m 55s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5cb9089da23d498d8fadefd9a18409fd : SUCCESS in 10m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fc2f7855ce084aeba533c4cee7893e0f : SUCCESS in 31m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ecbe2773ea1e4e80b9cb8e9dad0d00d0 : SUCCESS in 7m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d58cbad6f8c846e0a61260e493882908 : SUCCESS in 14m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/52e5a2b8da9c45c485aae64b2fd0b12a : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/771e6458df994b339e8bfe15e8dcee24 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f52030c17e5247489b9c8de7f9b017d7 : SUCCESS in 25m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5541de3bbed843c6b483701601263e34 : SUCCESS in 7m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/65346f8d97b14da38d42cb6b2abae567 : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/313b257d00524961919feb5d61649cac : SUCCESS in 6m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/835028297fb3411fa896c2a66fe8da61 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ab93c71a5684646bd68cb6ef9093f85 : SUCCESS in 35m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b186d4c4dec24fe4a3a8f273e625199c : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":3},{"id":"077dbeec9bc729409db1b2658e1c913731f75058","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-16 09:14:24.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"603c6d37c07f380d095191006bdedddc0b8e3ada","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-16 09:35:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fdcac38bb18da9ba3ba987b71517ff326341a77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 10:55:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d7613ce8de64ebbbcfe17909ac49a93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1774424a17784098b07501e8af4d79b9 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877f51f257914da5a2cc35a2e3cb44c4 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97fa5189b17c4cd0bff2d5cc8b5f0c89 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67b80c948d0b4b10a3345b4776515863 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c56de3f6c53247418a99d620ac05de39 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a6065109fd840f9aaba4d585e3bd4e6 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99b4c97b5f3a41b096bf49d0397b1083 : SUCCESS in 9m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6d8ea4305efb4e468123c84b6212c980 : SUCCESS in 13m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/854d90164d724f3faf60c87b4f19cbeb : SUCCESS in 9m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8efbab47218643c7ba1424a09f407fe6 : SUCCESS in 16m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c46ea64f26c34e719e6a8245fa70a848 : SUCCESS in 6m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/abda88e5404f4a128b7ddf766bf0728a : SUCCESS in 4m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/dac3945d8a1b47cab8145b80499ae003 : SUCCESS in 4m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cf7762289eb3410792bd2b50df0171a1 : SUCCESS in 10m 22s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/53f347efcb9a46028ba9090e518f6532 : SUCCESS in 5m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6feec422459b4e7d916453b9fc9233c8 : SUCCESS in 6m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cb4ccd88f0424e798af4563ebe283826 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3f6648a0fe21424583a161874b9ec74d : SUCCESS in 8m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/94692982bc42440486862da9f5652871 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b213db6b14e402cb49310eb2afc312e : SUCCESS in 7m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5d56d89ddbd6489e9cb36d90ea383125 : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5c791005bd034636be895a3f9c598ecf : SUCCESS in 13m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f9aee2eb165a4ab9a10d7376adeb3c66 : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/933035ac29624df1bbd1b10f19d97399 : SUCCESS in 10m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cbf5d98e8c124696925203029aacc685 : SUCCESS in 4m 13s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e6ef703843214c5a9b5e652006425dd1 : SUCCESS in 13m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/18550e52716140e68ad34ffcd83f1d32 : SUCCESS in 25m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/448e81c692a741b09d4ece7f1d8fbd9b : SUCCESS in 7m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8130eba89434469fa3f2b33cd23db9fd : SUCCESS in 14m 49s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ea785b82b687460ea13d7e8fad5e8943 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f263e7c696de41dd9dfdf017e3ca9a38 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d971e318ddae4e218e936d18578efeb0 : SUCCESS in 27m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/14f574af776d4da5aff3d5b9a06d18f6 : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7e7de40e8d7e4d939b92e5215d16020a : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38eb42f37e664d79b3bfccc7b4af16ab : SUCCESS in 6m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1c84e364daa34ea3a80cdcf7b8c654f5 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29ea203bebcb4c0c8fae7569c7e158ed : SUCCESS in 39m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b3eb9cafa29455b8d1d2c0e61da075e : SUCCESS in 1h 35m 05s","accounts_in_message":[],"_revision_number":3},{"id":"06f4674cf1b78d7c1e89eb3d0f7763fd74e4f26c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-16 11:03:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f824cefdc7e8cd946d4b0355d1b99b3f522d371d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:03:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a80468f491392e351c3da66e4c5a929ba0359122","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 12:42:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6fc621319714fd8823252aaee785cbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fbc36e3cad4c479c5a81d8828053f4 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cb44dee859742b091d5547a2fab3ca5 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de85231be524632b65dd05e6bce1349 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4679e249364a4c8b8894d9e610b89398 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/250377814bfe4fa1a15788afed352148 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/117d73854a8b4cecb50dfacfa5e64cbd : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5809a37091774dbea5e5a4c772da100d : SUCCESS in 10m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69f909f049b14476887ec83a708363ed : SUCCESS in 15m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c232ed58e68c4c65add8c60c45d399d3 : SUCCESS in 9m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a61971c200c3485da1d481b7c9ab9996 : SUCCESS in 12m 09s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/ee72055ee7ae4c52b1bf8431d9cb7e01 : SUCCESS in 6m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1c37c5b0acff47c39096fb064466a1ac : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bcd5cc2c16a54f0cba5445de7e900200 : SUCCESS in 6m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fa90ce71a88849a18515c826481570f3 : SUCCESS in 8m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/17639e7723b54f6eb9315af0aca17257 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a44166d1d2e4cdbafe725bc1d964fd1 : SUCCESS in 7m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a10797455f9e4e27b6cdf157b2c95b40 : SUCCESS in 8m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7d0721c5d8aa4118909c2c9fceb3f3c5 : SUCCESS in 10m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bc307a112a9045fba4e43f1c4b91fa61 : SUCCESS in 4m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/45095dd7067149f7bd43c81afa9a23e1 : SUCCESS in 10m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/33ac130556784a6c8bb655960b83e3c9 : SUCCESS in 27m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b0bb18abd1734e00b4cf4cd4b1b193da : SUCCESS in 6m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/63390783f3e349a49b5be890b3580346 : SUCCESS in 12m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/27880525659b4ef5b2669a2af2fdb49c : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/888072d90ca747fea131e0dfd3f557bc : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9cd27035e3e477eab633ec3ebce0aaf : SUCCESS in 24m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/28c3b5518a9444dab5f9c136fe78a886 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/be3f13f6d7de414789fc33fc197b0ac8 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9add3ce21784a289d59e3002f74a41b : SUCCESS in 10m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5ecbe094455b40adb247c9de8065d41b : SUCCESS in 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/522a5eba519e443089b240992c7b2737 : SUCCESS in 35m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5b747bb411f4409a246203393904ea9 : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":3},{"id":"997f5c1b06704d9252ce45f04e124ab6d2256e9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 12:42:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"40ab440917e7639cd03f5a03317bf511e5980db3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 12:43:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/051f77513159431788e30e9382d8b88d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8c413afc2a42ac88272b0d6db953f4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3},{"id":"8543395afed6e751865dba9aa7836263609acbff","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-16 13:21:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b2bdc84b1fedc6fd543c283291cce621465172bf","tag":"autogenerated:gerrit:revert","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-17 18:58:41.000000000","message":"Created a revert of this change as I062be5c503775d138eb65989c2c9788e58ed0e95","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3728e28a91f25f7c1023c07f563dc0e0ce688312","revisions":{"d5166bebafb78f43ffec1e98671b60335a880dbd":{"kind":"REWORK","_number":1,"created":"2023-12-20 19:06:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/904147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/904147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/904147/1"}}},"commit":{"parents":[{"commit":"b6efc819aa8c495d3356325d29db97f711dda840","subject":"Merge \"update constraint for python-neutronclient to new release 11.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6efc819aa8c495d3356325d29db97f711dda840"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-12-20 19:04:34.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-12-20 19:04:42.000000000","tz":60},"subject":"Update eventlet to 0.34.1","message":"Update eventlet to 0.34.1\n\nSeveral important and urgent fixes are released there.\n\n0.34.1\n\u003d\u003d\u003d\u003d\u003d\u003d\n\n* [bug] Fix memory leak in greendns https://github.com/eventlet/eventlet/issues/810\n* [infra] Fix OIDC authentication failure https://github.com/eventlet/eventlet/pull/855\n* [bug] Ignore asyncore and asynchat for Python 3.12+ https://github.com/eventlet/eventlet/issues/804\n\n0.34.0 (Not released on Pypi but landed with 0.34.1)\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\n* Dropped support for Python 3.6 and earlier.\n* Fix Python 3.13 compat by adding missing attibute \u0027_is_main_interpreter\u0027 https://github.com/eventlet/eventlet/pull/847\n* Add support of Python 3.12 https://github.com/eventlet/eventlet/pull/817\n* Drop unmaintained and unused stdlib tests https://github.com/eventlet/eventlet/pull/820\n* Fix tests and CI for Python 3.7 and higher https://github.com/eventlet/eventlet/pull/831 and https://github.com/eventlet/eventlet/pull/832\n* Stop claiming to create universal wheels https://github.com/eventlet/eventlet/pull/841\n* Fix green logging locks for Python versions \u003c\u003d 3.10 https://github.com/eventlet/eventlet/pull/754\n\nChange-Id: Ib2e59a207b86ae90fa391bf1dff7819851dc9c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5166bebafb78f43ffec1e98671b60335a880dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5166bebafb78f43ffec1e98671b60335a880dbd"}]},"branch":"refs/heads/master"},"135ac878eaa77d3188a3ebe5494f5b676be16c0d":{"kind":"REWORK","_number":2,"created":"2023-12-22 15:01:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/904147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/904147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/904147/2"}}},"commit":{"parents":[{"commit":"b6efc819aa8c495d3356325d29db97f711dda840","subject":"Merge \"update constraint for python-neutronclient to new release 11.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6efc819aa8c495d3356325d29db97f711dda840"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-12-20 19:04:34.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-12-22 15:00:52.000000000","tz":60},"subject":"Update eventlet to 0.34.2","message":"Update eventlet to 0.34.2\n\nSeveral important and urgent fixes are released there.\n\n0.34.2\n\u003d\u003d\u003d\u003d\u003d\u003d\n\nhttps://github.com/eventlet/eventlet/issues/861\n\n* Allowing inheritance of GreenSSLSocket without overriding the __new_ method https://github.com/eventlet/eventlet/pull/796\n* [bug] Fix broken API related to `__version__` removal https://github.com/eventlet/eventlet/pull/859\n* [doc] Fix pypi broken link https://github.com/eventlet/eventlet/pull/857\n\n0.34.1\n\u003d\u003d\u003d\u003d\u003d\u003d\n\nhttps://github.com/eventlet/eventlet/issues/842\n\n* [bug] Fix memory leak in greendns https://github.com/eventlet/eventlet/issues/810\n* [infra] Fix OIDC authentication failure https://github.com/eventlet/eventlet/pull/855\n* [bug] Ignore asyncore and asynchat for Python 3.12+ https://github.com/eventlet/eventlet/issues/804\n\n0.34.0 (Not released on Pypi but landed with 0.34.1)\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\n* Dropped support for Python 3.6 and earlier.\n* Fix Python 3.13 compat by adding missing attibute \u0027_is_main_interpreter\u0027 https://github.com/eventlet/eventlet/pull/847\n* Add support of Python 3.12 https://github.com/eventlet/eventlet/pull/817\n* Drop unmaintained and unused stdlib tests https://github.com/eventlet/eventlet/pull/820\n* Fix tests and CI for Python 3.7 and higher https://github.com/eventlet/eventlet/pull/831 and https://github.com/eventlet/eventlet/pull/832\n* Stop claiming to create universal wheels https://github.com/eventlet/eventlet/pull/841\n* Fix green logging locks for Python versions \u003c\u003d 3.10 https://github.com/eventlet/eventlet/pull/754\n\nChange-Id: Ib2e59a207b86ae90fa391bf1dff7819851dc9c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/135ac878eaa77d3188a3ebe5494f5b676be16c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/135ac878eaa77d3188a3ebe5494f5b676be16c0d"}]},"branch":"refs/heads/master"},"3728e28a91f25f7c1023c07f563dc0e0ce688312":{"kind":"REWORK","_number":3,"created":"2024-01-11 15:42:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/904147/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/904147/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/904147/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/904147/3"}}},"commit":{"parents":[{"commit":"920ea1a6d5014fff0105bbf1371abb8302004441","subject":"Merge \"update constraint for oslo.cache to new release 3.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/920ea1a6d5014fff0105bbf1371abb8302004441"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-12-20 19:04:34.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-11 15:41:01.000000000","tz":60},"subject":"Update eventlet to 0.34.3","message":"Update eventlet to 0.34.3\n\nSeveral important and urgent fixes are released there.\n\n0.34.3\n\u003d\u003d\u003d\u003d\u003d\u003d\n\nhttps://github.com/eventlet/eventlet/issues/875\n\n* Fix security issue in the wsgi module related to RFC 9112 https://github.com/eventlet/eventlet/pull/826\n* Fix segfault, a new approach for greening existing locks https://github.com/eventlet/eventlet/pull/866\n* greendns: fix getaddrinfo parameter name https://github.com/eventlet/eventlet/pull/809\n* Fix deprecation warning on ssl.PROTOCOL_TLS https://github.com/eventlet/eventlet/pull/872\n* Pytests, fix error at teardown of TestGreenSocket.test_full_duplex https://github.com/eventlet/eventlet/pull/871\n* Skip test which uses Py cgi module https://github.com/eventlet/eventlet/pull/865\n* Drop old code based on python \u003c 3.7.34.2\n\u003d\u003d\u003d\u003d\u003d\u003d\n\nhttps://github.com/eventlet/eventlet/issues/861\n\n* Allowing inheritance of GreenSSLSocket without overriding the __new_ method https://github.com/eventlet/eventlet/pull/796\n* [bug] Fix broken API related to `__version__` removal https://github.com/eventlet/eventlet/pull/859\n* [doc] Fix pypi broken link https://github.com/eventlet/eventlet/pull/857\n\n0.34.1\n\u003d\u003d\u003d\u003d\u003d\u003d\n\nhttps://github.com/eventlet/eventlet/issues/842\n\n* [bug] Fix memory leak in greendns https://github.com/eventlet/eventlet/issues/810\n* [infra] Fix OIDC authentication failure https://github.com/eventlet/eventlet/pull/855\n* [bug] Ignore asyncore and asynchat for Python 3.12+ https://github.com/eventlet/eventlet/issues/804\n\n0.34.0 (Not released on Pypi but landed with 0.34.1)\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\n* Dropped support for Python 3.6 and earlier.\n* Fix Python 3.13 compat by adding missing attibute \u0027_is_main_interpreter\u0027 https://github.com/eventlet/eventlet/pull/847\n* Add support of Python 3.12 https://github.com/eventlet/eventlet/pull/817\n* Drop unmaintained and unused stdlib tests https://github.com/eventlet/eventlet/pull/820\n* Fix tests and CI for Python 3.7 and higher https://github.com/eventlet/eventlet/pull/831 and https://github.com/eventlet/eventlet/pull/832\n* Stop claiming to create universal wheels https://github.com/eventlet/eventlet/pull/841\n* Fix green logging locks for Python versions \u003c\u003d 3.10 https://github.com/eventlet/eventlet/pull/754\n\nChange-Id: Ib2e59a207b86ae90fa391bf1dff7819851dc9c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3728e28a91f25f7c1023c07f563dc0e0ce688312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3728e28a91f25f7c1023c07f563dc0e0ce688312"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
