)]}'
{"id":"openstack%2Frequirements~904205","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I7637dc540d3872f4c1f44e1fd47f00d76f15a7f9","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-01-24 21:13:15.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-01-24 21:13:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7637dc540d3872f4c1f44e1fd47f00d76f15a7f9","subject":"update constraint for neutron-lib to new release 3.4.2","status":"MERGED","created":"2023-12-21 12:36:42.000000000","updated":"2024-01-24 21:13:15.000000000","submitted":"2024-01-24 21:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904205-new-release","meta_rev_id":"216c30a28bd1ae71cfbdfc0db4c24ca5b6e57871","_number":904205,"virtual_id_number":904205,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-24 21:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-24 19:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-01-24 14:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-24 14:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-01-24 14:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-01-24 14:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-01-24 14:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-24 19:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-21 12:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-21 13:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-21 13:55:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2023-12-22 05:33:28.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2023-12-22 12:09:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-12-22 13:03:11.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-01-24 10:04:33.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-24 10:05:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-01-24 10:54:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-01-24 17:30:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-01-24 19:20:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"40fe4a68c1e4d5c9cf669ce296b4c3c7d0a126f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-12-21 12:36:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5d8f6633712168daf442678cd2f5c9d4457b00d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 12:44:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7478a8e00564167ac899b23393b8096\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7d8fce0ab8b24dcc9c699b92f5eb633b : FAILURE in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"16fca308efb128312622f1476dafc1c2b7a1f553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 13:44:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a32e457bd7e4e22becba879b255080f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae986291d28442996b9f6fd1d4aee46 : FAILURE in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e87cb2da6a24e3f98414055947c7799 : FAILURE in 2m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/558003da4ea64a1e8a230cda1564628e : FAILURE in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0757a11c2b74b92bae2bd8ed722a9d2 : FAILURE in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9eb1b4c90f65418f973ac8a9468e7a7a : FAILURE in 7m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2d266dac457745a282a027fedc700bca : FAILURE in 11m 19s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/24f0688891b649969af5f87686da782b : SUCCESS in 6m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c24969fd75e341fe8bab5be762081f9b : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3363bd4e2b02433da599107669a1715c : SUCCESS in 6m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8ebf61a6c2ba4c6ea5fb7b61fd01a192 : SUCCESS in 11m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/192d352100cf400b922153608fd20b4f : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9f8f32f808004e01aa7bb2ab3e80ac38 : SUCCESS in 5m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4524e72ca31447e89f735f12be98c3ec : SUCCESS in 7m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bbb90ac4c24f4f8198cd2630f8c379b4 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbd55b6ed0694eec91349048931dede7 : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e2f335f50c564840bf0aaf0fe2ba89e6 : SUCCESS in 10m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/da40eb19d8374879b9cd5f460388120f : SUCCESS in 12m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8c65f83adf6a4494a06e6911452a06de : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/104b394b672345aeb3103ec39a124c06 : SUCCESS in 10m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a4e2d56fb1d3449e96a162d5cbd2d4d6 : SUCCESS in 5m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/af30bc8a94d0428b9d6f919fb2916675 : SUCCESS in 11m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dcdc695d16684c16ab4837bd09f008a0 : FAILURE in 24m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e6c782256e6140eda00c9cf1a7712241 : SUCCESS in 13m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b8e6e8e0913f41f6a6f081d52b17e3de : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3449ff537f7e40568c57b16dc6b74e88 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/364f68a9a911484f8257ad1ed75ee9ac : SUCCESS in 23m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0992e5122b704a3488dd8fc5929462f3 : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f9d2a2a06a27445498f998c2800e89fc : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ce86eec6aac4dc9adec0098ddbc697f : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/77f414f112334bac8ed71269c1b133ea : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/801412d6263549d482fe53cbfd648260 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/001f779aa26c45578e8fb1913c10a17e : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2abfd32004ed89b45786b66fe57728d83a0d3b5f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-12-21 13:55:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4150560425718480023dd71061285063654e1418","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-22 05:33:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"686b7b08ee1695325e54934f18d793f8478c52ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-22 12:09:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5eb5aa8bb5bd67ab5edb617f9009dee68512281","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-22 13:03:11.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44bd072f116cea10dddddfb236ac905f4b78c7ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 13:09:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18e379a1b30a4e72957d4c0e85487544\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/24c3c1c1f2c740709798f3ffca4fbcbb : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6a3b1a80dfdb0d53c7d0c14f0f7ed38855a7e70f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 14:31:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dc5622a7bf14971a17cc69faed598c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98ba66ae364498088feaf6c8c7520f8 : FAILURE in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8933004f078d462b84ec8cb7d021a5b3 : FAILURE in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb7418610cfd40d29187860fe16f5a9e : FAILURE in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f1b15f195a64c3d97bf72794071cb74 : FAILURE in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4660c0bc09cf4e5faa4d2ec0a99ce4e1 : FAILURE in 12m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5fa26bbac86f4cc796a6b05d5a0ef813 : FAILURE in 18m 34s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6716a50d404843bc91ca4fe68fd5e8d2 : SUCCESS in 8m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f0753f682c7c4a09b08da9013a61718c : SUCCESS in 6m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3c580eb2562d4f4eb9f58e03bd8884d2 : SUCCESS in 4m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/21a68e1ba9a44383a737a320ff6ecabc : SUCCESS in 11m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/29c62f03c8584b8e8504fd0de5912d04 : SUCCESS in 4m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f9c15e77f30a4696a11ed9b27c432ca3 : SUCCESS in 4m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f2d450644af5424cb455e37c7b299574 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3248792a1ee641e881cc72513dbef2ee : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c31f74571f86409abd390ce6baa771e3 : SUCCESS in 8m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8773deb782084407b11cebc927e3dbff : SUCCESS in 9m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e5f75632c7a942028a3a8df38cb25fad : SUCCESS in 11m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9be769cf05aa44e9a1c3f9e75a9a619c : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7c701a2027e44ad0be3d1ccd64e61955 : SUCCESS in 8m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6a2fe6df45e4493cb63913f0a270e023 : SUCCESS in 4m 51s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2b2574347a144785b0c682857dea7213 : SUCCESS in 11m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f3eed831d7f54917a1af1a3847dabe99 : SUCCESS in 22m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/64873b7e02784fb8b9fb5a4b4e050f74 : SUCCESS in 13m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/47eeb857dabe48da95b3793192202617 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1db5ed2b11c74345a164f910b13e70ac : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/415534366627440792b80985962e0755 : SUCCESS in 25m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c12018f5144d4d55945e1e9bbd1dab71 : SUCCESS in 5m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/486dc7a07fe245b1a29341c29b26a341 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aceb53d103654284bacfff172cff02dd : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8123add743874a358e80b5ce3057bbe7 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d958ec4c0f3845ffa8226c61e82827a3 : SUCCESS in 59m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/053eeb03a2b84488a6ab00f2098e80f4 : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8ba8231203dc4c555fa34d17472f62325ab16b71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-15 10:47:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c787ba636482e6c9007dffc09d52cdc73ad12b8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 11:01:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a57a41276a3540908ec9a62376385ba9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ec50d19fd9a147119cdb91b1b8b37005 : FAILURE in 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3f56eaaa29cf08d6563d514951ac70b2a55321ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 12:20:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3e9737a42b943d5b57bfd869f5c8d7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d60a9b657513486e8bda21019e7cb589 : FAILURE in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/715e5e4de7a041e3896ccc36185614ec : FAILURE in 2m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12d8f3c4aacb4443a63c791f34010c9d : FAILURE in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ef1dddf6c2844ca8a3a866704cb4d45 : FAILURE in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/65de7fb847af4c8a832ecd050376a26e : FAILURE in 7m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b4b8605bb0b4075a61aaf8f14104a68 : FAILURE in 14m 13s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/cd4600e0a2a340d4817818f4b78d48d7 : SUCCESS in 7m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a8d7d347885247e8839b8cc795a32017 : SUCCESS in 5m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/454cf6d3aeab4d6889a449dc2a221d35 : SUCCESS in 5m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b683036aa32f480887210a57821e23e4 : SUCCESS in 11m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/746c0a81542a411c96f6922a5b4f0bc6 : SUCCESS in 4m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b544e75f0d0a4bfd9a4d347e9365cb88 : SUCCESS in 5m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cacde3a094a142f49739895be903ccdf : SUCCESS in 7m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8d405dfeba51460ca3d80129777934c6 : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1bdbe4442ee0403594268513baf0296f : SUCCESS in 8m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/641b42e5cbe947e0a902f132428edd52 : SUCCESS in 7m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/78ff0095af6849dd9d8b00d5d725d4b7 : SUCCESS in 11m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/42036fe6cb264c10a6da4b893cacb389 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bd910d29b57e474d999ca80698c25fb8 : SUCCESS in 11m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1ca4905ef1494761b0d621a7d548e34a : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e7baf2b84cfa4050b16b88457910d8c2 : SUCCESS in 9m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/73e14e30664243c99621d05242296efb : SUCCESS in 24m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/14c249db693f4fd88ee2454051f3cfab : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/74b71cc3640e49d1b3eb39037ec010c3 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/068a3c632c0740df9b9be4222d1596ac : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6dfff66bb944485baaad495c7500394b : SUCCESS in 22m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3d8f5ea5e2cb4b7980225715fafb5cf3 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7a751604a17944eba24f603862964e3c : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7518bbe2f3d4c85b4a710867a8a4c6c : SUCCESS in 6m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e4b886dafa2344f7818368d9d6ae2c48 : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a3b9ad5e13243cea31f05ea7d90ea3b : SUCCESS in 57m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5eaf68838574b578ee043ad299be405 : SUCCESS in 1h 24m 48s","accounts_in_message":[],"_revision_number":1},{"id":"adcd9a14d91498f3ad12c20567b884f323c5d928","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-01-24 08:01:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9b81a6f90e293c4dc0dd1a988d903713ddb7e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 09:44:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d98eeb846701fb4d9c927b3b5f2a5bfa2691d00b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 09:51:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3403ef376db40659a26354872ffe1d9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8902ac9263f84f948cc3ae5fc366800c : FAILURE in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"9b4aaad45f56f2431e38e1dc6eccb7fec31fa104","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 10:04:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c66f9fae697272d9c72261a70a7135493523476","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-01-24 10:05:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4307535f2420d35afbc5c3af3a63f6d336842eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 10:31:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e51ea78a9147bf4f94a7b6738f4725802af60a51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 10:39:41.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2bde46874ef44f68fe96948f2206f76\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74a2423d22074203b8be312af5dce458 : FAILURE in 6m 28s","accounts_in_message":[],"_revision_number":3},{"id":"ca8c098b3308557549e1017a2da5fd6eb8e1d49f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-24 10:54:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"32391cfbf9a968d0ec8e51ac10fe5929cb82fb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 12:35:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a66bd87821974e66afd1d95622e2c77a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0345ab2e1a3648479005fb3c0598d9b6 : FAILURE in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ff0a7f92b2b41e9a51f604f78e36b3a : FAILURE in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b21071dac313418d831aeccc19fb0b13 : FAILURE in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/660eb3a02ffe43a1bdb00753ad48cba3 : FAILURE in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dc1226c6d54f4654992f3c7ca5e47e10 : FAILURE in 7m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/044ea2d624ed4a83b71f79eaf6a98875 : FAILURE in 13m 06s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/03672e856b464691b6eed82397dbf801 : SUCCESS in 7m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/65329acb970f46f0a50bed8af286aa09 : SUCCESS in 5m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/dc65bd77d7aa433986e4764b744b3d79 : SUCCESS in 5m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f247d356cc5446f7bcf36553eac313f2 : SUCCESS in 11m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f0ff8407bb5245359dd5664b2e5730e6 : SUCCESS in 4m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f3bb10c3dc01451294de663b9651d0d5 : SUCCESS in 5m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f3870c414cec48b1b6e1b01f0978b214 : SUCCESS in 7m 23s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a1b19457288e4ee38fa01417416d1f99 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8640c40e92414391bacff50a20d5a6df : SUCCESS in 8m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fd6dfb0f727c4124ab94a0f2346605b3 : SUCCESS in 9m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/37f2319a1a3344bd9193c2df2ac56c92 : SUCCESS in 12m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2ff605377bd04d79b1b031cecc3883fd : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e8dae248d6ee422a83d92538821c2ab8 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/96726fa2d9d843bcae7eec0907a0bcf2 : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fa67310d06294c45b6ec438102c7f54d : SUCCESS in 10m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1566e8e5b412493da582d9da179e6c36 : SUCCESS in 24m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bd6ecd2fcd214f6fa41139248acbe350 : SUCCESS in 13m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/18d25d8047ff493bb7b2a71408223f80 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95cc1631ae644213924dd2a863d115ba : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c231d2530d064672afad14c36cd9918f : SUCCESS in 25m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/021dab742fbf4617861226122479ec8d : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e514c4065c374fe299f8e3b1ad0fd216 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f40c434513b5453e9117691e3349f47d : SUCCESS in 6m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fbec9cce7c224ff8b1dab2fa8c57b74c : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38eb9845b9284ce985d587ffc7587f70 : SUCCESS in 1h 53m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/984b41d2273b4b07b57f60ab563bffc8 : SUCCESS in 1h 26m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ad2b4369506fe038ce1ecf0ead6abf04727351f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 14:33:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1a80194f2c3d6c816e94234cc96307191a65655d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 14:38:27.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87f6989b6a044e97935d227a04abe5e3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1c8690a190004892955237f0d84c1118 : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":4},{"id":"4923160b740d5d6abf9122e342560421d5d04577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 15:38:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/daa1f0235dff45b3a4c18be14d5cdbaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567098d77d7f48f2aa589c834ec628ef : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8c396c1f4ee43af9de38fc16647fdc6 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4feba2b5477540df8ce6b508a46117c6 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eacb321ace734440802334286da344df : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ca6f85b452ac4b379f04c22ac134676c : SUCCESS in 10m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fc11996a150343269abf6ce5bcc6f06e : SUCCESS in 16m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1f60f99ca7234f709f40bdd6bf14a0da : SUCCESS in 7m 19s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7824166407ab42de9e168dc5e440c36c : SUCCESS in 5m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/00023ba0f92a469b8ce17773af807653 : SUCCESS in 5m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9f4892842ef644d6ad988e8066d8e39c : SUCCESS in 11m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cf238f1d53cf4cabb0514aa7d168f0d7 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/92979c5882144eaba379f919eccd7453 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/81553b70a23a4b61b8a17f0232ffcf0b : SUCCESS in 8m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d9372d653f554dd49f0e7edc458468ac : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6061adc93ec64e68b676a7aacb3efa93 : SUCCESS in 8m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ff27746d58db481794f18c5277038df9 : SUCCESS in 11m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/eee5d346a15b48a9820bbc21dd14af6c : SUCCESS in 14m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/45cd31900f0e462ba98ad76ab1669d26 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b15c4c9458564e15a593c21b74c158ad : SUCCESS in 10m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c7305c304c3942eb91dc56b9f58b46e4 : SUCCESS in 4m 34s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c8a81da1d73341afac886cae87d9ffb1 : SUCCESS in 10m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b4a209b4256847ed8db613d74fd98446 : SUCCESS in 22m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/14cc898ed847477882ffbfe62ab5fb87 : SUCCESS in 12m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/04d076552873489784a116485605e4ce : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d9554d6d7b64505a230d5342f1d5119 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10221aa8cbea435788c0c8f852baf95b : SUCCESS in 26m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9321047105424a37a72fc3a025196e66 : SUCCESS in 5m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/457a42b339724d37936014c7ab511432 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce85a039e1cc4e64911b4739fe9f668c : SUCCESS in 6m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0e872b93c5fe4602b2fc80532a714776 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea2db7090b9e4316ae6a39a28f1f0261 : SUCCESS in 50m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9fef05f161b421894550a988956270b : FAILURE in 53m 56s","accounts_in_message":[],"_revision_number":4},{"id":"47a7af67147697a5966009b991f2e12b0f99c76b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-24 17:30:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a5d4a2b5c3ac0e3d537a6618ae0ab36d6d44e58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 17:35:45.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f46f2444b24888a551e13649f5acc7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/291bb7b26fa743a8a228813eeaea437f : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":4},{"id":"69f5161fb83e2527005839e7807b640fa189b5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 18:52:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75803b1e5f82435292d2818bd22800f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51af1bbb062d462fbe7c19c32d7d399c : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3329710f980943bdaa329931828c81ea : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/499e8919be0e449e860457614db52ddd : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef575de619bb4c9394e932f0b63ad7ca : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d66f0b6f6f044529bc2a34bc593ffed2 : SUCCESS in 7m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6dd2078a434c4e40a6934e111e096219 : SUCCESS in 13m 11s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/335200632530417c99817cc9f7259bcd : SUCCESS in 7m 19s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9d39473d1ab043f388c8a35ac82e5fd1 : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/67669bf9331b468a9f4b8baf4ba1c169 : SUCCESS in 5m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3af22d142a594c69983f3a6eef82993c : SUCCESS in 10m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7ece97f8b97341c3ba09d05696de34ef : SUCCESS in 4m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fd69e07aebcc4b2faddde3a315f356bd : SUCCESS in 6m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/686a1a66436642c19c05abff9149097e : SUCCESS in 8m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a80c75e9748c41628b75c6444059d262 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b8fb99472854267bdc351ce86532874 : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/18f7d4a52f9c4fbc8aa409193e1533c1 : SUCCESS in 9m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e1d453db8d4a4771afd6ba5d9984a8cc : SUCCESS in 11m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fedead81c5b64fa18a744fb85cdce93c : SUCCESS in 4m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7a38c255137e4e2da0ff4f235d85c184 : SUCCESS in 10m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/635ebce30b5e4ad6810aa8a3fc3e95d0 : SUCCESS in 5m 37s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c4dc65f1344e49428f9c9d255e853ef6 : SUCCESS in 9m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ba00eb3659eb45c5a597178078ae7df9 : SUCCESS in 21m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3f0f19bef3ae40059b41bc200a2c81a9 : SUCCESS in 12m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f4816404d1f8404f8f1d97043a1a4dc3 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cc861f46df243c9b6b3813fbc12d5e2 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f68d5f17b4e44a89ae745e73234d02f0 : SUCCESS in 22m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0ddf6e9983b04d668d02f01f055aa1e7 : SUCCESS in 6m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8715ab50e4f4422987a56b2f524867ef : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c020c380f87a4e1fbd1aa81ca5d21e1d : SUCCESS in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c2a5b96e023847bc9faad070d6c2f82c : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45b7b14592e2482a85f3a28997c3788b : SUCCESS in 40m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26f63b56858146a2adacc076b9c56b43 : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":4},{"id":"9b2cb423ec5b05b20254d8269f11a80f373c609b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-24 19:20:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fb68453b2b1614c8c0ba4f5cf5a9bcef7af70880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 19:21:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5dc23f57e28aed725f944bc4d21b890b0e42c88a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:13:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6be3921aee1471e8321749a59b470d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14d647b0175a451f9321dc8626719981 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc14a0855451451cb116a8c1457a831f : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb5db741d4c240e58e39aeaa7c25b1f0 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61e384482886493d8f4f1ca549b95bbf : SUCCESS in 4m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4c5dccf4197498db76fb75ef11e9ec4 : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7554199bc31a4f3886b6576a14717eb8 : SUCCESS in 13m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7f38a982d938430f98167a82bdabc148 : SUCCESS in 10m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7a8fedd149394acda703f92fc9c5dd55 : SUCCESS in 5m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5043030f5cfd4339948954f99762f662 : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fbd4d1e19b36406680b9b80da3996496 : SUCCESS in 9m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/228b779ab67840e79d1e0173c024f441 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79a2a94c301d4ac59ffca9f4a0191e91 : SUCCESS in 7m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8a4e19ab6e6f4ef9836111be7a05f30f : SUCCESS in 8m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/48bae74f35cf4c36957573f6b3896b51 : SUCCESS in 13m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6efe86bb7c704f16998aa6d1b38bdfd1 : SUCCESS in 4m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bad3806737bb4fa5ac1062ad3733c66d : SUCCESS in 10m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/47ccc0c1f2e74feeb2c60ba065641d2d : SUCCESS in 4m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b11a675344ff46dcbebf3ead109e0a26 : SUCCESS in 23m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/abeb6e0669fb4d8f873564f0e4f9abf8 : SUCCESS in 15m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9b804f51ec9442bbb61e10eccc1db330 : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24418309c5d64e8995623ad8c745a577 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/235ce86334734093a720e8810dc50ed8 : SUCCESS in 24m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8e302f1ec4134de283ea3b77e8845312 : SUCCESS in 6m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9e7971ab06664cfc8fbc78f857ddf6ca : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/986c029b00704d96a7347384bc014493 : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/775e03d616bb452fae732776c5304753 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/617620600a6740d9a6fd595b19526382 : SUCCESS in 51m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10df2f850be14eeaa70fcbc93d1112a2 : SUCCESS in 1h 29m 46s","accounts_in_message":[],"_revision_number":4},{"id":"216c30a28bd1ae71cfbdfc0db4c24ca5b6e57871","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:13:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f8a887aab5cd10c897f70494aab1f4d0016a3288","revisions":{"030a7a1a3224c1b0fb722d4dd6d9dcbead84ca7c":{"kind":"REWORK","_number":1,"created":"2023-12-21 12:36:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/904205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/904205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/904205/1"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-12-21 12:36:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-12-21 12:36:40.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.4.2","message":"update constraint for neutron-lib to new release 3.4.2\n\nmeta: version: 3.4.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ieb46c4593120daf9f34883bd3ebee5d8639b2c21\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nChange-Id: I7637dc540d3872f4c1f44e1fd47f00d76f15a7f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/030a7a1a3224c1b0fb722d4dd6d9dcbead84ca7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/030a7a1a3224c1b0fb722d4dd6d9dcbead84ca7c"}]},"branch":"refs/heads/stable/2023.1"},"8b01fbe5333f47923cb66b984ea6f891b32054c5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-24 09:44:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/05/904205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/904205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/904205/2"}}},"commit":{"parents":[{"commit":"a9ecc497e708937d5acd2348893c9f049f67c98b","subject":"[stable-only] Fix gate by removing virtualenv cap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9ecc497e708937d5acd2348893c9f049f67c98b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-12-21 12:36:40.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 09:44:19.000000000","tz":60},"subject":"update constraint for neutron-lib to new release 3.4.2","message":"update constraint for neutron-lib to new release 3.4.2\n\nmeta: version: 3.4.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ieb46c4593120daf9f34883bd3ebee5d8639b2c21\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nChange-Id: I7637dc540d3872f4c1f44e1fd47f00d76f15a7f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b01fbe5333f47923cb66b984ea6f891b32054c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b01fbe5333f47923cb66b984ea6f891b32054c5"}]},"branch":"refs/heads/stable/2023.1"},"8ea22a75c65e5f02f850e6fedd3a393bef57974d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-24 10:31:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/05/904205/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/904205/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/904205/3"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-12-21 12:36:40.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 10:31:58.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.4.2","message":"update constraint for neutron-lib to new release 3.4.2\n\nmeta: version: 3.4.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ieb46c4593120daf9f34883bd3ebee5d8639b2c21\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nChange-Id: I7637dc540d3872f4c1f44e1fd47f00d76f15a7f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ea22a75c65e5f02f850e6fedd3a393bef57974d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ea22a75c65e5f02f850e6fedd3a393bef57974d"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"f8a887aab5cd10c897f70494aab1f4d0016a3288":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-24 14:33:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/05/904205/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/904205/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/904205/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/904205/4"}}},"commit":{"parents":[{"commit":"5ac6c3f2a9c923ec1d716680fcef06b26c467b9e","subject":"[stable-only] Fix gate by capping tox version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ac6c3f2a9c923ec1d716680fcef06b26c467b9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-12-21 12:36:40.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 14:32:58.000000000","tz":60},"subject":"update constraint for neutron-lib to new release 3.4.2","message":"update constraint for neutron-lib to new release 3.4.2\n\nmeta: version: 3.4.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ieb46c4593120daf9f34883bd3ebee5d8639b2c21\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nChange-Id: I7637dc540d3872f4c1f44e1fd47f00d76f15a7f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8a887aab5cd10c897f70494aab1f4d0016a3288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8a887aab5cd10c897f70494aab1f4d0016a3288"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
