)]}'
{"id":"openstack%2Frequirements~905245","triplet_id":"openstack%2Frequirements~master~Ifb75cc754de94f5751732cafe0b6732e3d305e8f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifb75cc754de94f5751732cafe0b6732e3d305e8f","subject":"update constraint for oslo.config to new release 9.3.0","status":"MERGED","created":"2024-01-10 16:08:21.000000000","updated":"2024-01-11 17:54:38.000000000","submitted":"2024-01-11 17:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905245-new-release","meta_rev_id":"19036c0900932a79b817bcfb5871b3138cbf1199","_number":905245,"virtual_id_number":905245,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-11 17:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-11 16:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-10 16:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-11 16:03:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 16:43:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-10 18:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 06:55:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-01-11 16:03:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"25d4850bb1a42c2fedb4cd770c192fdf6cf3b37b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-10 16:08:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9feb14e71822c3c09e4e5562ae24cc63a606db3c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-10 16:43:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2292474f72aaea7656f3bb14d1f58064be63dfe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 18:03:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86bbb62d8c1340eebffd8280c8ef0126\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499a197a8e0b45d984754233fc7116bf : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de459931c72f4c2da323b223253ab2e8 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fabd97371bd435087375418ef84005c : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61bd8e5190e5424ba49a9c2c5db79ab0 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82c88bcf97514984ab9a006c87f796b8 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72b2c6ac7c1543f8a9ea86822370ce33 : SUCCESS in 4m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/817582ee7789433ebdbd379f8800e44c : SUCCESS in 10m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3ec7c009ab1f4a2fb62f05b4affdac8c : SUCCESS in 14m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/126b88177ac04e7f96820fabad8878e7 : SUCCESS in 9m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/81214f3684214a258daafd98c8c5d21c : SUCCESS in 17m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a8913937a0274038ac538265d6e30e8a : FAILURE in 6m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3dc602c3a2774b1a9ece098c30a0450d : SUCCESS in 5m 07s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1caa048af01149f1b57b47736a2e3620 : SUCCESS in 4m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5f821d79af604eaead3b70999008d6f2 : SUCCESS in 11m 17s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/752939c25f4b42d6946fa18e50f2758b : SUCCESS in 5m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ba9ba87fdaed4600af7603acf8544383 : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d6b9eec2d4474dad89e41ee422df6958 : SUCCESS in 6m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/51987a20771044b4aa6e6b5b68869862 : SUCCESS in 7m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9443810154524ebeae1bf45f1836bcf9 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1e3cc35279b477286348ba0a9c562e5 : SUCCESS in 7m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7b5a94195a604bf8b5b6d6b7f6c2c804 : SUCCESS in 8m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/27b7e3456b0545938fc931f7005f3f81 : SUCCESS in 12m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/11a05747bf5a4c85a4ff7e909e554c7f : SUCCESS in 5m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e5c21a6a9cb64e30b8feec8b7ed0d6cc : SUCCESS in 8m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/15319948b55a4cbb953d307319db8121 : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/62f4c2d6a6fd4813a2d597b1bab1b1f8 : SUCCESS in 14m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c2b7daf463ac4cbf976fc8ecf405f018 : SUCCESS in 28m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/66fde6b9add4401fba899655d403b62d : SUCCESS in 8m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c906829a00bd4cdcb42c62c3a0cb399d : SUCCESS in 12m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1e3d7f827a2842998f2927383e846103 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea42a162dde740afab5b84d4ce95180d : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d0d41d2da1340059ef92517051ab297 : SUCCESS in 25m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8620f61be78d43bda3a084f0691e645c : SUCCESS in 7m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a77af5ba87aa43edbd3ec2a9f0921784 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e112a622ee7e42ae85e04467e7dc1846 : SUCCESS in 7m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/948c8a7a6d2f477bbdc7fc996c848a5b : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab9f32c4d81a4ab1b623ae89c5fa76be : SUCCESS in 30m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e4fd242d5f34b53a9790764e8524038 : SUCCESS in 1h 48m 39s","accounts_in_message":[],"_revision_number":1},{"id":"870cc3f58e18d48303a353135f41d387aff4fea1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-11 06:55:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c562bd7259f522cd615aa71be0e1c1435e4d6665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 08:35:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3da50b4a9124798ac3a56b8f83e32b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57004d4f76fb431ca0203d979b55eaef : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8a9ba8844a4839bbbaf1cf0186a608 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b935e3c507140ae9c36ecc8f6bfae25 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a415b0714fc46228b60a5ee633775f8 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d1189da730643d0a58add15491d386f : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48515775dc6a46c88377e0b48f14f0d2 : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1f4991019514ab1b3c0e44e8a6735e9 : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dc04a30cd3c041d5933e3531971313f3 : SUCCESS in 14m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6ebc10b4b3654cffa230bdbad406c115 : SUCCESS in 9m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d48721849dee40d2bac45f96e033fab6 : SUCCESS in 16m 51s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0886d68c069f46a3b00c9caec9c66bbb : SUCCESS in 5m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a9e33f5b393a49dea17175d3aaba2642 : SUCCESS in 4m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1111a5d8c8eb489db26313fe72951b14 : SUCCESS in 4m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2807af5d436c4bc28d01772c0ab82373 : SUCCESS in 9m 53s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/6a8f1b3c71944297bca7d83601d560f1 : SUCCESS in 6m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2021503cc8774b00954ae4d67228829b : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/276ffe80c6924bfcb8e12e65a8820acf : SUCCESS in 7m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/26c3be216e7442ea83e245e95bf6c36a : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e2d1a4ea7ecc4a71a10ec705b6e5a71a : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc8928d2b3174c088872bf9d749bf32d : SUCCESS in 9m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/127d1a5887ec41c9b6a43ee7a0ec0798 : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2bc8c7a9157d455b8773a703728c43cc : SUCCESS in 10m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/646eb2fa22f2440baab643156842d345 : SUCCESS in 5m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1041b27ac789493185b6b348b5b1fd3d : SUCCESS in 11m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7c9704cddeb542cba883982535822e17 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b2bfafea9bbe4f08972bdf38eabbdb2f : SUCCESS in 12m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/247180ab5ba94d6785394370f45ee438 : SUCCESS in 26m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/acd0586ddc22470da8caf87f0b553ad4 : SUCCESS in 8m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/69217c3dc4a0498eb6ca96b005234f3f : SUCCESS in 15m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/00a2d3d44c8543f69203fdce0bddb1ab : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85070b429ab54608924bbdf5f2048f66 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63fa493d23a54cfebda8c1ee28de9425 : SUCCESS in 23m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/81d37b43001f4ac5abc932c18e54ddc8 : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cd6acbea592048d0aff94268d50081fd : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6af7d19318a54fa082e607799c0a5409 : SUCCESS in 6m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9a897292743f43cf8a6cd9b533e7e02d : SUCCESS in 16m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98430d66d8fb4a9c863d3fbfcef9bb17 : SUCCESS in 39m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/536bd1e97fe54016b8e707ce1c87e679 : SUCCESS in 1h 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ab4eebe446acf5b4ca9ec889507d8ed326a06125","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-11 16:03:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"657c72fbb1d7755a54dbf999b6eefcc42d591d2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:04:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85af9068757d7c809dc9c4dc30c0323bf7ea42ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:53:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1101a70f3b44672baff6752c08bce3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afceea03dba1435a9c9a475b69101453 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7106937e0504be287561228bedc768d : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61f5bdb1462e44a2ab5d054e6f23595a : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04b34992273d45f5a2a24b6907c20143 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e92a2e1b0101419da9c2dd0e5f0a982e : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ff365ecf3354efda423d81ab8f6010a : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e803578fcd24d32ae7e536c0bd7af85 : SUCCESS in 8m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/310c5b19d24847788ea92362176659ad : SUCCESS in 13m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fab5bfb6f6cf42f888d89c7e64185cac : SUCCESS in 8m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/41f63ea002754e5cacdc75b956c70d0d : SUCCESS in 10m 18s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/e69a73ec8d4f492bb29799284ede4f5f : SUCCESS in 6m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/858c580e971f4bb199ce42edefb316fa : SUCCESS in 5m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8a56b937e2514c4483169eef8549b552 : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/22f6fa9f585047648d87f32b206d2693 : SUCCESS in 9m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2baf27a33f864dac843c0ee5a8409559 : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/064f8be4a4c14899a04cefed137ec006 : SUCCESS in 9m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9c094e14704141af93af0f5dfef7be56 : SUCCESS in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f69e86e800bd45b8ac622ce8868c3142 : SUCCESS in 13m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8e4a4d5788dd4d21b91339175cac6cc1 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/672cd0e501c54e3baa2d960dbf2d0483 : SUCCESS in 12m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ba49373bbf064be38c2decb56a74b768 : SUCCESS in 25m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d4cb105cbbdf4381816dd14066e199b2 : SUCCESS in 7m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bc38b9c413564dca826d805633d2e629 : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0ed37afa95294ed2a4b33e8ac9c1479a : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d144cf738b064a97aeb0f81c2f8807ca : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b8c621247044d378199ac7b75b2aa94 : SUCCESS in 25m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c45a51b9aa5b4b14a3b3ca8aed84c353 : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9b9d1ff944d141c89eaaac44b0a0aa55 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43dc75bd25454fc8bd3022dbd782cb1f : SUCCESS in 7m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/79884a71b0254addad39b3d1a99e733f : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/152832d191074a0086173e40c554b425 : SUCCESS in 41m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b592621ce78140eb8fbb1aeea83f32fa : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c98d9a24732409caa6266d07cbeaa0ca85859196","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"19036c0900932a79b817bcfb5871b3138cbf1199","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:54:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8c52cd56a84a31a92a8e6e6cf12ddf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80954820ce154a6e9d177cc61e470c79 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a90a0809a0d554ae1b52af7e7be568687930b8d9","revisions":{"a90a0809a0d554ae1b52af7e7be568687930b8d9":{"kind":"REWORK","_number":1,"created":"2024-01-10 16:08:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/905245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/905245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/905245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/905245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/905245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/905245/1"}}},"commit":{"parents":[{"commit":"67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a","subject":"Bump Django version to Current LTS Version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:08:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:08:20.000000000","tz":0},"subject":"update constraint for oslo.config to new release 9.3.0","message":"update constraint for oslo.config to new release 9.3.0\n\nmeta: version: 9.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie760d7a24321ead31ff5a944fc041b87b19c4f3c\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifb75cc754de94f5751732cafe0b6732e3d305e8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a90a0809a0d554ae1b52af7e7be568687930b8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a90a0809a0d554ae1b52af7e7be568687930b8d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
