)]}'
{"id":"openstack%2Frequirements~905246","triplet_id":"openstack%2Frequirements~master~Id2ec3502829961e64167b4a64012689f7c10ceae","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-19 10:16:10.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-18 09:59:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-19 10:16:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2ec3502829961e64167b4a64012689f7c10ceae","subject":"update constraint for oslo.limit to new release 2.3.0","status":"MERGED","created":"2024-01-10 16:21:54.000000000","updated":"2024-01-19 10:17:12.000000000","submitted":"2024-01-19 10:16:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"905246-new-release","meta_rev_id":"31793317ef3e53e775052c24b29582c65bf87eb6","_number":905246,"virtual_id_number":905246,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-19 10:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-18 16:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-01-18 14:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-18 16:52:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 16:42:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-10 18:21:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 13:20:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-01-11 15:24:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-01-18 02:16:28.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-18 14:17:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-18 16:52:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e0c2739720aa000aef77f506f6033ba488cd41bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-10 16:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e478a6e5c5cee8cdea2a523260917a1d0afa1718","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-10 16:42:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebb467fd3430635d98359460602236c803fd199a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 18:21:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82628f2ab82e43ea9ac4543917955292\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b3cf7d8a52451da9ddbd6b3471dabb : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d90e5146bd4f27a87e78413f98f162 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73807279b32148bfa1dea5a88afca4a7 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81c161bb91d24c36a0752580ccef11d7 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c6c1e2371ef4b768aaca3339970ba17 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/145c3bab31ff46deb34513abf92bd1b3 : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc7343efd4084db5bb5a363ec5b8b2dc : SUCCESS in 11m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7d79c6b09c524837ad7c4aa1421e9ab1 : SUCCESS in 15m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cf66344ffae340e6a267d307f1f3ac4b : SUCCESS in 9m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8b0371695be343d49103eef13bf90069 : SUCCESS in 17m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b6f75a7f223e408d97c27c648d2448bc : SUCCESS in 4m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1a60192e542f47c2b284f1a7c4662c4a : SUCCESS in 4m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/26645b743bf44e6dbf465f6fc6e4c1ee : SUCCESS in 4m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fb962a5425ba4fd59de6e7f3661d8ea1 : SUCCESS in 12m 06s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/17d2f6fec9a14f9c9dd70e55ff53e332 : SUCCESS in 4m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a55c5a405c794f81af02a3a12e683e95 : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7634390991ab4442b4230b09563ef26b : FAILURE in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/20ae1ed3d3034892987cbc9d263aee51 : SUCCESS in 7m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c8b493c1045345ffba5b1dc0e9bcab90 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2aa3c4f2c7374e7eb31ddcc324784f13 : SUCCESS in 8m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/00a74f7732eb401ab145743d74694177 : SUCCESS in 7m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/80efa86ec4784140bebcd389f458353e : SUCCESS in 11m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/187ee23828684f8e9cc9109085a887f0 : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4fec4f3ba0704d2c9c4f26cc87cf41d5 : SUCCESS in 11m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e0bd47eafe1246b2a86c3901ad262487 : SUCCESS in 4m 28s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e07eafff23704455bb7f0d0351fc99ea : SUCCESS in 12m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f949331d78b04e84ae11af0c78e44688 : SUCCESS in 24m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/281f0526b22e46d58a20476693906ce2 : SUCCESS in 9m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/79d2c815f0e04d4b8f6838b1d1ec026f : FAILURE in 12m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6d798d4a954c439fafd9f638d2a926f5 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4404e187b464e9294a8076e7924fa09 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/159f06a88d7948209a111a9f45306e0a : FAILURE in 23m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0c8a1005021f4ea98b3740528473a984 : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/05363d108d964edca86f451c304cdc6d : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e8a087a1426447488711d819ecedca4 : SUCCESS in 6m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1334f71eed44483fa10f6f0ba2d5a6fb : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/021182e0ef4048fd8aeeee2c00563955 : SUCCESS in 28m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95589261ec3c44b1a5c67d711b055266 : SUCCESS in 1h 38m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6e03717febe5d350df08a8fdcfcbd841236ec061","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-11 13:20:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba852d0989c0291134d9c9f8d9eeb9db934b72f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-11 15:24:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6eef4e584ad74240cbad0231a82a472b1ab0747","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-12 16:47:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fa36901e012e34b44fea0e5adc48155b30c629c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 01:40:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5e5ab5afd14350f9c855de8968b68170e6534fd8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 02:16:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2628c0ba9f2098b3c37ad4820eaf9791ca54c117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 03:36:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a752a50f0aa42d298cdd3b9cea29459\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3773a8729c66473b8206fc36263a008f : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9571e4530cc741fe96ae448b9029302f : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f9731054449422b9de65d6f9219316c : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d78487092344c9ca5eda7ec603569be : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce5257a51d20457d96982afed82cb7cb : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7957f245f18411eb074eee90b83537c : SUCCESS in 4m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5e7bf4ecca384807ab069dc73fb38c42 : SUCCESS in 8m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d952e791de554e7f81b7b1c531483019 : SUCCESS in 20m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/000cae479838420aad35990c87212265 : SUCCESS in 8m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/46eb75a5f81f413b964e4fafcfeb9a62 : SUCCESS in 16m 56s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3c0987b2a1db415d8ce30ce5a43cca9f : SUCCESS in 4m 40s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cd4a68e74f9a46d99013c24970f12ac4 : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/31cfc1874b734c478a475d94916c594f : SUCCESS in 4m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0d22733af8054601b48c391891ffa028 : SUCCESS in 10m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e8cf9ffa914a45e8b8a8a631e001cc1a : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/efbb1b5068974394bc415a238888b78f : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/24be14b977f44d1b98ef5ddc92f90f97 : SUCCESS in 8m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d4b3a32f975f4a1ea9e744125e1118c3 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/215d0c26bec64a30a5d0e73717f4e830 : SUCCESS in 7m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9dc2254a1e5643669e96c752f4b92f87 : SUCCESS in 8m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f42f8810a70f411bbb55b3107a5794a8 : SUCCESS in 13m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c3108d68c3bd4657ba97f6b0fbe62eb9 : SUCCESS in 5m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a3b366f0884140d59455fcd889f9799a : SUCCESS in 14m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d9efdb36dd1940fcabe027257633b75b : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0c2e264734544cbea0d031493d348b9b : FAILURE in 2m 56s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3c679401914d4e70add69708ff0e9cb7 : SUCCESS in 25m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d106f87bfeee4eedbac415f38cb6889c : SUCCESS in 9m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9044c37d4de547ff9d302b14dd22d398 : FAILURE in 13m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/84861b4fee61429495bf730f315c7c26 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/202f58dfd0ca4ad5ba6a2ec17ea460da : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17cd36acc27b4517bf17f314b2ac32dd : FAILURE in 26m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e45436d5b9444579ba3ff7d9264dc1b1 : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/facac557b5364ae7aa115c7dd3106f4f : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a086e3bef1c4340b70917f5bc51151c : SUCCESS in 9m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8544e45eb67c4fcea599143b98f9e2f8 : SUCCESS in 14m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8ad97852b1c4608acd614d3649fe600 : SUCCESS in 30m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0dcd0b08c5d4498aa40a9196c0ca2cdf : SUCCESS in 1h 50m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f680cc236bc2191b278e2cb537f60f051c9082a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 08:55:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f5e0708c812a1cd722c92ab05ef5296bc72943a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 09:59:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"58f3684609272b320e46e3204f51fb950b5e5c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:27:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0e395d707b444f1a37ab515012913f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fd0f5da17a48eea79ac8c89f78fe1e : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1a8fa36e9f46f2b0e854987f61a1fe : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eb29b13379446ce80224ee12abf08be : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52fc786e3ec64f4dacada4b93267cfbe : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/302e098a9d2247648b619a312fe27bc2 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52bf0cf88bc84ee78e49b1e217f9fa4f : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/994abbf8fbca4adfa5d53b593754f47a : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/739597124e2244928334f5d1cca8575a : SUCCESS in 13m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/73e00d15ca0e42be9e4e2c217ab8446b : SUCCESS in 8m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/483126b7de6c43e9a6790df523768c16 : SUCCESS in 17m 13s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/79fd2a8e50034774ac4a79dfc108d58e : SUCCESS in 5m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/eaab2565b1124d7b9272be82e50a7b3f : SUCCESS in 5m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f28e457c9ce74190aef860d3d9a70ba3 : SUCCESS in 4m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fffb2850b4104f2791acf31191426e07 : SUCCESS in 12m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d6897d38ecf940ea8c3a7cb9fb75907a : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/20c08daa7b88406a87d0995b8cab8ab1 : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2450e2ae4f354bf284866d777b3e8ae9 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4d723e1625fe4cc0be53605ec813804d : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6105393a11454eb6b5d9e33bde98fd09 : SUCCESS in 7m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/df9351e3b8644758a96109944361b391 : SUCCESS in 10m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8c3238c0bdbf48569ff8d7e0b5aceb66 : SUCCESS in 10m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/75fa8026516041faa5183109e43da231 : SUCCESS in 4m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/383f0b721d0e429f966759d2b2b73c70 : SUCCESS in 9m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/303b9ad6700e45519394591d6ceb536a : SUCCESS in 4m 53s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e0eeb2206bc24adeb8f1f14bf2dff218 : FAILURE in 3m 00s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d23bd761160e4faa9148a20ce1b0d9e1 : SUCCESS in 25m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/91b122f662de44549554eeb364f3b8ae : SUCCESS in 7m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fa6e73c6eb124c969f10ad2c32b3ac4e : SUCCESS in 13m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d81bf55480bf44aba17e050f30b9db5a : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69092f3fd5d04095bcba46badb6051a6 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09f477fab0614cf8923f6ade2a30fe22 : SUCCESS in 21m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7f8a289e10574101a3122e413e608dbb : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/feb8e9f4d58643d795ad4dca1d75f9cc : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1121229ff514144b93688b99233f407 : SUCCESS in 6m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/143d53b153e94528a7742b59b72334e2 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed675c325b87488685ad99e115390991 : SUCCESS in 37m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dbe93e54caf411381d21119b2fb7d45 : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":3},{"id":"43eec514ff518804fb8f0947994866218ae02462","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 14:17:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f3bea45a188decd741335d47bdbc0400cbc0a422","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-18 16:52:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cae57fb394fc2b1b5c921c4700910fb745f2b3d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:53:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"71e144b709b963d5effb4fff3fdea140736a8ba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 18:46:31.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/901907 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"0e4110a0ff5025ac4feeb689794eea931e615b26","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-19 06:48:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c6a7288936b6296c0262b9611193a2ce3a2efee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 08:41:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b9a57456414fccb26429054db804ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7af375ccf540a4baf9df0b16196654 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c427adb5600b4c0a897883c3bb842795 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/130ab64e1a2f4660a6ded3aec8cfc98a : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f72ae0aa4db9418d91e55acad68d5e2b : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfce457e95484da992755f9fdb5aa4fc : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2648ca7b670a40c8ba2077f2f8d20a2e : SUCCESS in 2m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d43c6e90b10444ebdb75bef06694fcf : SUCCESS in 9m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/977e1f10aeac46e4b519531b8dff1f0e : SUCCESS in 14m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/88f2a872c8984af5ae3b0b11e2a36cbd : SUCCESS in 8m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0ad07fe796574796979befa71bec3d5e : SUCCESS in 16m 16s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6af87e7c4b34412ea5d0e15c7d3f4187 : SUCCESS in 4m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b4ee276fb6c34f538397a8f0dacd5290 : SUCCESS in 4m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4fe737866f634a3f8d4a9373113f3327 : SUCCESS in 4m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/077b0d367f06404db34b4a6ef71979c7 : SUCCESS in 11m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fa6fa217fcc84bad8db5ffecdae9e4d6 : SUCCESS in 5m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ee1809a70ab34dddb42ed5a2f24476e5 : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7035b73fdfc641768f1d7ed64ef8393d : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2e8fe24c818449af91dd5f2107e02665 : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95b1ae2decaa41439da4f3f7ae4f17d3 : SUCCESS in 8m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/53b1bb8db2a949769819733fcc26e865 : SUCCESS in 10m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2c2f11a7349644ba8a130efdb02a807b : SUCCESS in 12m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3a06332cd20d4442b9d1efbbcad67038 : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e2ed7e2902054ab5a5088493aed550c9 : SUCCESS in 11m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/06dd4aea45c04209a7ddecd1e9114bfe : SUCCESS in 5m 53s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/042494a806c0473f874de0abcbc5f952 : SUCCESS in 13m 33s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5ee3158b4bc045d8b29851fccb42891d : SUCCESS in 22m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fed242e744694c71aabe28e073e04186 : SUCCESS in 7m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/106ac65f829448c5982a2b6a05dc2a7d : SUCCESS in 16m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0824051036014cb6b200cca416e0b961 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4621ab759e474841ab922168661a11d8 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7dec4625888345e486df9458066e1c9e : SUCCESS in 24m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bde963693865410aa8c93c2a4f8775b6 : SUCCESS in 5m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cf8a0b69dcc74c56ac5986e02eb1deb3 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6dd80ee4c2845cc9325257374b754a5 : SUCCESS in 6m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2b203a89a7f940f9a2dc1ce454cb43fd : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93dba65d51f64e5192e084b0d9e5a808 : SUCCESS in 40m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/498582487ef44461b43d162d8716a6ea : SUCCESS in 1h 32m 24s","accounts_in_message":[],"_revision_number":3},{"id":"15b16a4e374de72f6aa5eac198ce791dd2a08f3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 08:41:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3332e94a68d1222644ead0d09ea43f6e3a7ec932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 10:16:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06fa890237e7418ea93a79fb53e136e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d100332f24a24e48aa1f6ba33be89a23 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aee1e0e707445e184073a3cdf7819f9 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35b4ebc3e842499681ff2974af632ea1 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16905fb7dd174c578fdd5d98a1a9ebd4 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7aef1619477c4cf284f6e79a310c522c : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/27b275d518af49978b2f6e0b861038ad : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b8518c97eb74ad3b81ac6d6d4d685b6 : SUCCESS in 13m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3f17a169eaf14e4b87511473ef524302 : SUCCESS in 15m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8bc9ae4d565942749dbab5089a1335ac : SUCCESS in 9m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b112e3f917a04f24aa4ebe30cbc060ce : SUCCESS in 11m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e600a4f9ac5a48249dab61ae2d5e82ff : SUCCESS in 4m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1bbd23e0055a4a6f828949ea7ee230f2 : SUCCESS in 6m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2ed28cecbdf444da9a4e69917d3cec90 : SUCCESS in 8m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f94d988fb855455682461a50ae61b549 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/632985a0ccc84c19a1c294a9df3c51fc : SUCCESS in 7m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/830b4aafd68c42df92d6060dceaf726b : SUCCESS in 7m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b049f499a9e24feb9771733de0b9b8c6 : SUCCESS in 11m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9e68304a00b44b01a929e82d2440fb5a : SUCCESS in 4m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/897d10ea91534e97a37530cf5f7e4ce2 : SUCCESS in 7m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8c5483aac5614d938a3a07d4db7ca3f2 : SUCCESS in 34m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c3916088690941e5b61253d2fc0e822f : SUCCESS in 6m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b07e9d08598448df85985178794745ab : SUCCESS in 13m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1450198b1198416ea74b196bf7a3afa4 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b297b1a5d2ae48c1a406c34ef0381dfb : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d18cae17cc6840518cc3ecf28c197b4e : SUCCESS in 24m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/918d1b4706774d1fa751e59da189ef0f : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/11b17167c84c4b51825d2bc276756807 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5af21753e6904b2a9c034ba32df9e810 : SUCCESS in 6m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5f85982d4034444a809477d594e48f92 : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84b47abba5e040cfab2b9e27af274a4a : SUCCESS in 37m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f757cf05f8f438c95df849c1e005a3e : SUCCESS in 1h 29m 10s","accounts_in_message":[],"_revision_number":3},{"id":"4e963e157969d8c81db43e99b6340924eec2ac86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 10:16:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"31793317ef3e53e775052c24b29582c65bf87eb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 10:17:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e132b7e496148429ed0f15d61788aac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc6ff67453348a99690c7276c985388 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"21630fe33339e3a35ece498d3f12bcddb91f1da8","revisions":{"5928c5ddf52feb8de2964b139e4cbece425fa300":{"kind":"REWORK","_number":1,"created":"2024-01-10 16:21:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/905246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/905246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/905246/1"}}},"commit":{"parents":[{"commit":"67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a","subject":"Bump Django version to Current LTS Version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:21:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:21:53.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 2.3.0","message":"update constraint for oslo.limit to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic63ac2cb7d6599809f70609ec874a865481483aa\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id2ec3502829961e64167b4a64012689f7c10ceae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5928c5ddf52feb8de2964b139e4cbece425fa300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5928c5ddf52feb8de2964b139e4cbece425fa300"}]},"branch":"refs/heads/master"},"ad3b4d6be9dcf479c0878fb6003abaf7d263b584":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-18 01:40:11.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/46/905246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/905246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/905246/2"}}},"commit":{"parents":[{"commit":"8d67991af046bb5a5c5a172440588f4b0bcf6249","subject":"update constraint for python-tackerclient to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:21:53.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 01:19:22.000000000","tz":660},"subject":"update constraint for oslo.limit to new release 2.3.0","message":"update constraint for oslo.limit to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic63ac2cb7d6599809f70609ec874a865481483aa\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id2ec3502829961e64167b4a64012689f7c10ceae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad3b4d6be9dcf479c0878fb6003abaf7d263b584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad3b4d6be9dcf479c0878fb6003abaf7d263b584"}]},"branch":"refs/heads/master"},"21630fe33339e3a35ece498d3f12bcddb91f1da8":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2024-01-18 08:55:24.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/46/905246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/905246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/905246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/905246/3"}}},"commit":{"parents":[{"commit":"f255ec17ebba4a1bc8000aca003fa50a5bbcde79","subject":"update constraint for os-apply-config to new release 13.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f255ec17ebba4a1bc8000aca003fa50a5bbcde79"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:21:53.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 04:02:27.000000000","tz":660},"subject":"update constraint for oslo.limit to new release 2.3.0","message":"update constraint for oslo.limit to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic63ac2cb7d6599809f70609ec874a865481483aa\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/905314\nChange-Id: Id2ec3502829961e64167b4a64012689f7c10ceae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21630fe33339e3a35ece498d3f12bcddb91f1da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21630fe33339e3a35ece498d3f12bcddb91f1da8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
