)]}'
{"id":"openstack%2Frequirements~905247","triplet_id":"openstack%2Frequirements~master~Iac0cdfc64c74c4ac4935b66f8adebf7021f247e9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iac0cdfc64c74c4ac4935b66f8adebf7021f247e9","subject":"update constraint for oslo.messaging to new release 14.6.0","status":"MERGED","created":"2024-01-10 16:36:29.000000000","updated":"2024-01-11 18:11:02.000000000","submitted":"2024-01-11 18:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905247-new-release","meta_rev_id":"86f0f8555dd070f0c1b6f5fb646491fc1d6677ac","_number":905247,"virtual_id_number":905247,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-11 18:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-11 16:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-10 16:42:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-11 16:03:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 16:42:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-10 18:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 13:21:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-01-11 16:03:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"baa5552cd0b9a4f026f4631ed55c99f52c89662f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-10 16:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13ef5e7dbac926f83103265f480db910eb75ac43","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-10 16:42:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ceeb6a1c64bf4ce383cb78be733f8c9f694f349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 18:06:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d584a17a5c84fc9b4a42f89cfba495a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa326309dc4943eb8814846eb180ceb2 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff82af4d75f441bac970b1f8ec5bd1b : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac2749169d9541579ad3a7aea5f52cfd : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46423cedd3f04760a68b05f728b5483f : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c482483b90a44877ba39b0fdd8099d48 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2254aee02194a8aaf670964d2c0cb67 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f53e46cd889a4e5fa1a7dad063de7522 : FAILURE in 6m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/77808de7a3ff491f808a27381dd1fd42 : FAILURE in 4m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a828d57ddb5e4e9bb3a420f5b32aee8f : SUCCESS in 9m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/613706925fd043e29310b9dc19ad45e8 : SUCCESS in 17m 35s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/40d6754f0078465c9847f3c7f5bfc977 : FAILURE in 3m 22s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/96092ba7041c471887d995f402f41a89 : SUCCESS in 5m 02s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5a798cb597ee491da4dc6e20fd28a3b2 : FAILURE in 2m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e8e8e246fb3c4f65a2cb67c6bcd61f89 : SUCCESS in 12m 01s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/6705269dab22495f925643c98e2d6718 : FAILURE in 3m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c11c7770edce47c28af054b6716424e5 : FAILURE in 2m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/52fe3aabc9c44137b95b0806328f4d78 : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b41e907d2d5b480984f8a1d087d37d47 : SUCCESS in 9m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1b534a6163bf49f2828692d200c3f80c : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3367503d10ab4efe8807657d36a5ffb2 : SUCCESS in 6m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f7ded0f20f0b454abaaed45477a2e4b3 : FAILURE in 3m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5a77c99447524b3ea29cac9405142651 : SUCCESS in 12m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/050e4e632eea4d7bb41afb8ee1e1676a : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c07e4bf5f9604deb83bed1c9b45f0ac5 : FAILURE in 3m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b72ec97c5b2f449b81ec5e7b97c69690 : FAILURE in 2m 58s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6f4a94e10dff423ea6fea5245845783d : SUCCESS in 13m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e0b03435109c43888a94d16d1540ce81 : SUCCESS in 27m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b1783ad9eb8242a693ac22d673dee383 : SUCCESS in 7m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/687efb06206c41a99a54bdf1c9b37a6b : FAILURE in 5m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/038b150cb12c4425a88480807c2a3f6e : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70efe354d3834503b074a78f204da5b1 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f045f3ba738e4c368a20f067707d536b : SUCCESS in 25m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8ebc728f80804465a38a68b957267f06 : SUCCESS in 6m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/197acb4591694f03b067f7c1c58584ab : FAILURE in 3m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/681a8a37e58b4f7aa885ee00e2c60f78 : SUCCESS in 7m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f61493cec32c45e79cf7443b1512476c : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07887f7cf06746128265771f08aa55d8 : SUCCESS in 31m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb2f4443073440548f28303cb372fec2 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0dfe82de9447daf77abfefe0f92f3bb4e16c0160","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-11 13:21:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a53563c9ec88daa2eb88fb2da40d17fc281a184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:09:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8888dc8c89ec4465a2b0ee2f5639823b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c849de0540e84f049235043aa86c03a9 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2fabac75ad7490e81ef0cdcc22f91d2 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1091dafa70ce4ceeb824bcfa8b9882de : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/109c23a3d8414ea8879f6e4c8dc0c4c3 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a31b41a6971b4c0b8178dc20f22f6a1a : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9b85388a55d6469ca87b265117f29524 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5da6aa47a0a04279ac2d7333fe7d8dbd : SUCCESS in 10m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2aedc8a1a6c0495483b1f046a616ba19 : SUCCESS in 17m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/16d492566b544aa883a7b6cd0892634a : SUCCESS in 11m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/479438a519864c188099f55002c8be55 : SUCCESS in 17m 22s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/36ab74946a6b49798247469752935f52 : SUCCESS in 4m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8d6c5586c76d4382b6fa38f92ba9367d : SUCCESS in 5m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7a0124a08ad9400a9c83fc2a67f20baa : SUCCESS in 3m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e071a1d17f3d4178b1e4bb026bdbfa6c : SUCCESS in 10m 39s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/e7e99a60b61a4987b063dcb6ca8c9ca7 : SUCCESS in 5m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4dab1d133457465594711892223a7be8 : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a18a540661074f5a97a89452aeb5b48c : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/49207015cf5045e8915fffe7b6caf2de : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/653c87c4b12045ddb5a84a436bd7e098 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af38b20fc4dc4a9f859270ddb6e2aed2 : SUCCESS in 12m 43s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e8d0d2c3220c469a84d7162e849e16de : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ccbaeae6837a4411b5e9e5b722bef785 : SUCCESS in 12m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c6ce2c22b88742e6883750a90af597bf : SUCCESS in 5m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/eef8eb6869fb4188b1a9a9077245e578 : SUCCESS in 8m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fa2817daeb9744c6828770120f480b3c : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5925d9a8130c4dbfbb0b341b58129d8f : SUCCESS in 12m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c3adaeb8f65f44b983c7f522f69d9e6b : SUCCESS in 27m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/db2b2c958a8341419c3a970919452698 : SUCCESS in 8m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5fc30915e13343bebe379b56471c4887 : SUCCESS in 15m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cd9b6dd8bd9146948194e504ba84f927 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6081e905ccc4833a57bf323d1a74021 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d58e0a9c33c49788d10f9f24b574259 : SUCCESS in 23m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d22a39e91b0240489bee1d6d029a02a7 : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6b1b854c2157450a9fb091111b8e09a6 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2364ae741095435588c8f0547bd33c64 : SUCCESS in 6m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8c0f88791b9741a3806e4c7e38a2e7dd : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e50583d8c8b4575b1479ea057c97871 : SUCCESS in 34m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9285074ececf4bf3ae891233b845abf9 : SUCCESS in 1h 37m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1db1fbc3a3312f6cd1c3298149aacab35659b1a8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-11 16:03:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68febe294a30cb935a798e1d155920059f2ce6d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:04:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a61e3e35a397e34e398deb9671806bdf076c1bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 18:09:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0901e8b8382546a08036818a3eec76f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741fc8e473a044908116bcb699e78d01 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25fc871a0db45b787fe6f11d7941081 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20eb830c597f421ead83035043ba6482 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b7ac4a576904142a404c5bad93e3470 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e24cd76642b43f5b6cc0efe59a279dd : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/138fb1312a6946d48803ea54d7f9d0fc : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5cdd228c0964310aba49c8d41854eed : SUCCESS in 11m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17393a2751024fefb37bf8abb5e93855 : SUCCESS in 15m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e20a47fa791a468faa3b9b389b4097f1 : SUCCESS in 11m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c289410e69fe49139d87bd5b28826295 : SUCCESS in 11m 58s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/366b91f87b1b48a5b9802f7970c7d9d3 : SUCCESS in 5m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ef213e36011040b695c392149bb980f8 : SUCCESS in 4m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ec8f541ae8d1446dbc12d82f56b85695 : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/232acfe08bc1464e884cc19c99532ec9 : SUCCESS in 9m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ed1ffa17483c4ebdb100d5f0fa0ec644 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47da74fffc354c01989863321e653c5b : SUCCESS in 8m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e575fcab95964434b3a22d1b3984c639 : SUCCESS in 8m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7fceb6d1d19a41a5958ac6a721172eae : SUCCESS in 12m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/44109817e52e453c82b9a7e150fd3baf : SUCCESS in 5m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d347416b54ea4892845039bda3845d8d : SUCCESS in 10m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d367ecbbd9534f9caadc682846ab8692 : SUCCESS in 25m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5a6a6eca869a47ea95b3f627d6fac844 : SUCCESS in 7m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/649544dbd1f54d6bba0c524b922b2276 : SUCCESS in 14m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/34361f24bc554b19866e040a647a9ab7 : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/962844d2c4004354bc0dfc2febf49763 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2bc56c960bdc413380a7743967aaaee6 : SUCCESS in 23m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d2abec02440a4883b873dd287d0cfcf8 : SUCCESS in 6m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/aa6f24b3f424430384c73529dcca9458 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a9e61e9f02045e3b8e74fc8cb157c3c : SUCCESS in 7m 00s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/12b78b7e46b143d89e86de9fd67c2bcc : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d431acfb95414b3490da369c66c52b7b : SUCCESS in 36m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f4dfd8e542a411294e6dd396e7eaa36 : SUCCESS in 1h 56m 16s","accounts_in_message":[],"_revision_number":1},{"id":"de97c7f5dea0a9f27c415707a343d939fcabf3b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 18:09:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86f0f8555dd070f0c1b6f5fb646491fc1d6677ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 18:11:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aee42487dbf4dee9c4c9b05251d6ab7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e22bafb7d94691a0a38c1f366a2275 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b42702a4d4685f91338f3ae0892685b696177f8","revisions":{"6b42702a4d4685f91338f3ae0892685b696177f8":{"kind":"REWORK","_number":1,"created":"2024-01-10 16:36:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/905247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/905247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/905247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/905247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/905247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/905247/1"}}},"commit":{"parents":[{"commit":"67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a","subject":"Bump Django version to Current LTS Version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:36:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:36:28.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.6.0","message":"update constraint for oslo.messaging to new release 14.6.0\n\nmeta: version: 14.6.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia84a1ab36ee3114535acdb3cf24aedbd1c4fa664\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iac0cdfc64c74c4ac4935b66f8adebf7021f247e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b42702a4d4685f91338f3ae0892685b696177f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b42702a4d4685f91338f3ae0892685b696177f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
