)]}'
{"id":"openstack%2Frequirements~905249","triplet_id":"openstack%2Frequirements~master~Idbe420144731f9801d644ccdd72efcb8db740a8f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idbe420144731f9801d644ccdd72efcb8db740a8f","subject":"update constraint for oslo.cache to new release 3.6.0","status":"MERGED","created":"2024-01-10 16:42:08.000000000","updated":"2024-01-11 15:24:17.000000000","submitted":"2024-01-11 15:23:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905249-new-release","meta_rev_id":"5aefee3dbf5a5445e8f3dbc40a2c0ca54d9caac1","_number":905249,"virtual_id_number":905249,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-11 15:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-01-11 13:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-01-10 16:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-01-11 13:48:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 16:42:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-10 18:26:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 13:48:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"10d945ea19b711cc2872cd4574458f04115a8583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-10 16:42:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc9e384b0606fe77bcf16684f1a525e01a1c9ae7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-10 16:42:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8858af562f2da9babe180c93163c76712cca01a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 18:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb5f34d3265540d0b306794d534150bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21cf022ff3b4447b8e7f9201c6529d1 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0406cdcc4b24bae865353c686a700d8 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b471aedc700d4ec1bd902e666d685800 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5d9610dc87041ec899eee99ee6424ab : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17e2bffc870e430cbee594b8b3197e12 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18330433ecf2450abd306913e16a1d8a : SUCCESS in 4m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b27bb2453e30475a985836f6d30fb1a4 : SUCCESS in 9m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/33fcf249c0c04e05a427b6889b8de4e5 : SUCCESS in 15m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/70d97df5670244f9b1544fd6e4db19cd : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/58b2e9aa5a6a45f59388157639a3f531 : SUCCESS in 17m 05s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d5c25eeb1a5c479589cebfe4c3a3be87 : SUCCESS in 5m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/543f29b0b7c84b53bdda880f160be87c : SUCCESS in 5m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/abee85d334b74b8598fa7f13252debca : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8c19623cb8fc470c9a9709f0c268b481 : SUCCESS in 11m 01s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/0b6b043f1c0f43eba564d45b86494294 : SUCCESS in 5m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4ad1997e534645cba17b969246c4c16d : SUCCESS in 4m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9409e3f2514b4787a0b273916bd954f3 : SUCCESS in 6m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0f815d704f134d38a44a2e64e3bde6f7 : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0c0d513bab44479d9cd66a8df8463c10 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c12b8c18c24d4d389d52cb12d824846a : SUCCESS in 7m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b798f74f482a48cbb090c82819903639 : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/939ab931cc1849c1ba90f6270f13c8f2 : SUCCESS in 10m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b48622e7f2064d92b03325fd4f14b2c0 : SUCCESS in 4m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/25f288d6842d41a0b0159b4597136686 : SUCCESS in 12m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0df194fb6e104979a375808cb9bd8fd9 : SUCCESS in 5m 03s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/04a0807c22dc429ea56281588bbd3747 : SUCCESS in 12m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0a885425b22841208569a213255ac5c9 : SUCCESS in 26m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9a3560828be444a8ad9f76776faa0322 : SUCCESS in 7m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3866e485a92c47b98ab0596634be0bae : SUCCESS in 15m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fcaa94296f2347e4954dbe15920fb66d : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a042578f089445dafd6d2fbeb603fc3 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/822259b70e194c3f93aabf308cfb4af7 : SUCCESS in 24m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9eaed2cc008f4c3c9c7f988c376c48af : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d625abd8c5024dc8b59f07581cda80b7 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/117d90437f3f4e2595b12a6d2ee28d05 : SUCCESS in 6m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1cd708e5b08f4935858d13f6a6c3727d : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d41b96391587459e87f79ccf67e63e7f : SUCCESS in 33m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff3199047b8e402d8f5ad9235b9699f8 : SUCCESS in 1h 38m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e2186c17a19e5ce1f83936cc7ab8b86349300f5a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-11 13:48:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"809b3d04a0541c62f528caba7bd316dceae7a4c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 13:49:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdddde790026941184a07ab1a5dda095eb2e0d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:23:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235ae130b0f446a7aae1e7042a3195a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/235fd231a0354905ad2bfc6d93f828f6 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580e6d814fd040cc9434d17e20f7f0bc : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb83ce116706455ab697f378be754eaf : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb28a3afa99d474a9b7f1150c6d43a15 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fc5449ee0c44976b3f9ffbec2602dc0 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d625bef5354d4f50ac4f278fb582b59e : SUCCESS in 4m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/62c319960bd54bf29f5cf190f04b1ace : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e37df9e686d5449eb21d2bb33eb72d3b : SUCCESS in 14m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/db5efac674ed4f9fb6cb9695601bfb15 : SUCCESS in 8m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0eecb3560e1546afa04b884feff401da : SUCCESS in 10m 40s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/8c96859930ad4d49b509c174e8459301 : SUCCESS in 5m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/76ca92e931444a31af3683537252e3d5 : SUCCESS in 6m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/94c15af381304540b5a99ee3f06a5658 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cc9b5aecf22b4460829447ae00eb9a34 : SUCCESS in 8m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/19e674ac3beb471997c66eada5a2a6f4 : SUCCESS in 13m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae846bf451024a75a30f2c29ce01cdeb : SUCCESS in 9m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/475642b60aab4fcba2910804a7a19b87 : SUCCESS in 7m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e7d3e85ddc6745bda3cc784dbbae76fc : SUCCESS in 14m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1a6647372b9d40e5bf8047f13f89f1d2 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0f2c407ffd0f42d8aba7ce640dfcddfc : SUCCESS in 11m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7b4d820f9f034b468e9f210c48a79fe3 : SUCCESS in 26m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3f6674b1332d4d3f8be06eb9a178bdac : SUCCESS in 9m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0c92d05c15974eb981b98420f1c260fc : SUCCESS in 14m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4540693b3efb4e4b872ba11c9be31b76 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a3265455f7d48cfbbe91524586eb278 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6923ddac893e4c588cbf8da28031f114 : SUCCESS in 22m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c09c51591cc94e0d8de106105d344aa7 : SUCCESS in 6m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2bb2bc5c2cdf436a8eec68283032be3e : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4af738bcad304f35820ebab147e9061e : SUCCESS in 6m 36s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8f62e585892041b2ab9b27ebde2d4361 : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95341c9d67844362a04ce0edffd5ec24 : SUCCESS in 22m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b1fd4f819e748458e911c26c2a90fe4 : SUCCESS in 1h 31m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f7331f455586619b27b70079cf00d7d50a740cf6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:23:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5aefee3dbf5a5445e8f3dbc40a2c0ca54d9caac1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:24:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c89292a08ae24543b178a45aefa99487\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87e124ca9674dd29ace532a0806655c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8413ce12449ae75d8125bf41a171391bcd4b146","revisions":{"e8413ce12449ae75d8125bf41a171391bcd4b146":{"kind":"REWORK","_number":1,"created":"2024-01-10 16:42:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/905249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/905249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/905249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/905249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/905249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/905249/1"}}},"commit":{"parents":[{"commit":"67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a","subject":"Bump Django version to Current LTS Version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:42:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 16:42:06.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 3.6.0","message":"update constraint for oslo.cache to new release 3.6.0\n\nmeta: version: 3.6.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iee4dd95c5dc4cb242511f4b7ded0f65eb8b07674\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idbe420144731f9801d644ccdd72efcb8db740a8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8413ce12449ae75d8125bf41a171391bcd4b146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8413ce12449ae75d8125bf41a171391bcd4b146"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
