)]}'
{"id":"openstack%2Frequirements~905258","triplet_id":"openstack%2Frequirements~master~I7065169c7544787b99e22483ebb7c2cb41e33bdb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7065169c7544787b99e22483ebb7c2cb41e33bdb","subject":"update constraint for neutron-lib to new release 3.10.0","status":"MERGED","created":"2024-01-10 17:27:35.000000000","updated":"2024-01-11 18:10:04.000000000","submitted":"2024-01-11 18:10:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905258-new-release","meta_rev_id":"3ff26451c8ace30c3cbf3684b6ecf10c34673995","_number":905258,"virtual_id_number":905258,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-11 18:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-11 16:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-11 16:03:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 18:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 06:58:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-01-11 16:03:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"200069867dc6d71146966dd0d4d71d5a4912fab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-10 17:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa482929ff97790d22a5245d043df7a4ad5f28a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 18:50:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29f65556fb5e488d96439ed76687eadf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb4f55689e543e7a68fe984f047bdeb : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74e11fa82b941a8a147053f927ad7a9 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68823d165e214b7c8ef345ee8442bd61 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a9fc17eb9384f8ca2fc04b25095da9e : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5209c625e413495eb318638f2088469b : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/165500b66f3c4d3b89c0feb89d173b99 : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ff672affc1f04ca28569cc6f22d55183 : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/075683e1dd9740948519c2db7abcf704 : SUCCESS in 15m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/52260ddb7d1a49acbd1f0ce2ed7452c0 : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b78f7452a982473c8bb758a10453b781 : SUCCESS in 16m 20s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/70b68e7730af4300a6a517435b16991b : SUCCESS in 5m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d96f2956b7364decb23d93df989e267c : SUCCESS in 6m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b93cf737209d4685af9e94d4ba61a3e2 : SUCCESS in 5m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/af9786dffcfc40b0826710adbd600b9d : SUCCESS in 11m 40s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/44bb65123ef54e49beb187af1021ff96 : SUCCESS in 5m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f1d45eae8184444c85279e0e1f84469a : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/21f21492a62a4294ad7d02c17f5fa2e0 : SUCCESS in 5m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/10a18d3c365149f7bfe7f0468d8ee14e : SUCCESS in 9m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/dd1bd1735be7476185643c70c03a7118 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22b20575d8764cbb90e98d2d62858c2f : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6c0dde41f8bf4aa48fbc413c966df832 : SUCCESS in 9m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9e7b3d347d0f42bb84e58cd37ceb39a6 : SUCCESS in 13m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d8aeaeaa57b54b698121c1dff7b02555 : SUCCESS in 4m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f62ec764e2994f7a961c14b2cff00029 : SUCCESS in 10m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d0b7b2552ab1492f9b340f3c8708cd39 : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7654e27a99b0410da4f22c9e4bcb3de7 : SUCCESS in 11m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fa4befd434d8462fa0c46de37c839205 : SUCCESS in 23m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1701b72795f84a9bb51d5aaafe6e2781 : SUCCESS in 6m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/36bbd6e60981407c8d2698a9db7d1d27 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/12059e35fcc145249b0d719aac0a5203 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/caab55aced154f6d9e1324b58edfca38 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6086e4964da34228bc63e6adb7b77f9e : SUCCESS in 22m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7cd2de16242345f7b9eec333b2f13ef7 : SUCCESS in 5m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/becae136cdd7445b91c58ba74321dd4d : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d08e07e01a2e43abb2b64efeffbcbbf2 : SUCCESS in 6m 32s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e0429e151e4846d593975612e81fdc74 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e324c76e795410a8fae067e6025352a : SUCCESS in 26m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e82c381aef5a4b8e87229c50e2c7ae3a : FAILURE in 1h 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fde8af67e5858a2744be275ae346b4c6549a61b5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-11 06:58:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"259536ed7aa651d7440bc578e81784a20329b5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 08:49:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f812a6edda204502b584296c6c681ccf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0542a357af0843c084e0929db7ce80e6 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4072f584ed474b81a83a66096c930c62 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f72f94cd0b1b4990a22edc5a8eff3aa4 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f085905ce8c4bd1aaae33495af363ae : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a629a4aa48cd4d199a1e70c3fcce4555 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fa7cb30984d4a7287ef2c17d8dd8209 : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0031c3ebd28a4c3f8ed4c799314f108c : SUCCESS in 12m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d643f44aa17c4e2faa794cf685bbd2a9 : SUCCESS in 13m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6f4a7a0f42fa4bbe9c286d0c8c059bea : SUCCESS in 11m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eb0255180b0641c2a228ee145ad809f7 : SUCCESS in 16m 07s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4a430ce2a9d24b08ab5896f00958342c : SUCCESS in 7m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6fc370d43d754b658a2daaaafe84f852 : SUCCESS in 4m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a0bf0a56774c4d569315007c3f409ded : SUCCESS in 5m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/00c42ee9f582497ca1d90fc1a7d488bb : SUCCESS in 10m 38s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/fe46fddc8ede42fc99286fd1e4c5962a : SUCCESS in 5m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f3fbd72877d842f0a55e57634dca8403 : SUCCESS in 5m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ea08bede9a934938aa6b9ed7750fcd2c : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eee21382cc75401c85fdf3b4f3e688e2 : SUCCESS in 9m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/cf1b9016864d4edf972cdbcb88b928f1 : SUCCESS in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0412c054eed4357a20d0cd787ef2cef : SUCCESS in 7m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7bbe5a193854494285ec2a13e30ac396 : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/881e02b8a189436fb7ffacd4d0f2607f : SUCCESS in 11m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dd586220a1b447068154d4308f77d924 : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5c9ac27075104118bc79dc4688bcff2f : SUCCESS in 9m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ccf0a64024094f64abf7d68fe511b3c3 : SUCCESS in 5m 03s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/aae7aa28c1044737bd401a0e1a484f57 : SUCCESS in 13m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8a61fe4ef4464f84989d9c9636dec9bd : SUCCESS in 25m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3b115dd76e1c4ac59cce0c07f90f17c7 : SUCCESS in 7m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/50a5f3b8ea01437aa9bf202c491ac81a : SUCCESS in 12m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/44c6f1e658604e808d7be74171f390c4 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e166cbb2491a4d64843fb039028cd660 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f318e471202e42149bd62d719bc52470 : SUCCESS in 21m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/21dcc5fcfe3349728c60d54dc14c4873 : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4e828ae094c840b0a33ceb65b947a62c : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc958287f8dc4d869aa0c266b139eb2e : SUCCESS in 6m 50s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4613ce37a71942c79daebf2076a28d9e : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c8d399ab94246f1922e407c94a02a55 : SUCCESS in 37m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c522f83cd6e4403b81b1b6c306196c5b : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a8689d14a9569549d38b4a741b92a239f3589ca0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-11 16:03:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f213f06385c1d2b9b11e594b67d56ef118fc12a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:04:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9bf47b178bd5e89c789c04d61f98a99a6073610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 18:10:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d9ccf82610f4c0cb5bf73e492fa8e9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc0c3571e5f4ede85b677f7320142e6 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1e6f92e4744053bd742c1b8f5ad286 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b240702d081c4c37a3657edf42b5fc91 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56eba60637fe4a26afd5ffdfe52276a6 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd690bc790fd4d238009f8f060ddb627 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82f9ea02d06c44b1826de3a53b80d9a4 : SUCCESS in 4m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6adcfca5d5c04dd7b57e5b232c8afe12 : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3b834c213c8241fe98b478e66fb7eadc : SUCCESS in 16m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1b049a78fede4cd7a7e30602eb59920a : SUCCESS in 8m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5e18bd50c7a744d78b9e4314d632fefd : SUCCESS in 11m 44s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/e306a3f00def44f0bdeee29e59be587e : SUCCESS in 5m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/874acb86f66b4494b93b581594fc1383 : SUCCESS in 5m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c3fe583e0321484aa9a0ca94e41aae73 : SUCCESS in 6m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f1336bd192e24e3e98a2ee3d176a4666 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/42025d22dece47a3b0174f767ae776c4 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/216ef2504fe24875a3739f23586500c5 : SUCCESS in 8m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a33091d32b9d4924be31cfee636c7ab5 : SUCCESS in 7m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cfbeccca15ae4d87920faeeb529ac81c : SUCCESS in 12m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f6437e9fec874e53bd2cd7aa9ea3e92e : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/893249d36f7544329f060c3abaa6aef5 : SUCCESS in 9m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f4588e477ac54e8cb7254d299e30b7d8 : SUCCESS in 24m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2d95a3c9977440319cb678b88938a928 : SUCCESS in 7m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0dff5fe4881845cba8a69deec8e21e44 : SUCCESS in 13m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bacf2539d7f249ffa204faba1348fe44 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecb28fd496cb4aea901529736a24e66e : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92b28811206f4496800fdf666f6fe8ff : SUCCESS in 26m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/de7c5395afa74da8ab5090fbaa94ab4c : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/89a1d4e7c5bf4726911eb6aefbf86a44 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d069a45732834ff1b5868c7d71260ad7 : SUCCESS in 7m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/799838a759b24535957cb8017be79c2f : SUCCESS in 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/725b96f445fd4888a2a7b2cf7e2ed66c : SUCCESS in 37m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19451cb12b6847f2bb2c0edce4381b35 : SUCCESS in 1h 39m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3ff26451c8ace30c3cbf3684b6ecf10c34673995","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 18:10:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df042959db53703c517156ce8895884a9769d92b","revisions":{"df042959db53703c517156ce8895884a9769d92b":{"kind":"REWORK","_number":1,"created":"2024-01-10 17:27:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/905258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/905258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/905258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/905258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/905258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/905258/1"}}},"commit":{"parents":[{"commit":"67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a","subject":"Bump Django version to Current LTS Version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 17:27:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 17:27:34.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.10.0","message":"update constraint for neutron-lib to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I030510bc3b98477af3fbdaa46684d68e9b969fa3\nmeta: release:Code-Review+1: Shanjiayi \u003cshanjiayi@yovole.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nChange-Id: I7065169c7544787b99e22483ebb7c2cb41e33bdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df042959db53703c517156ce8895884a9769d92b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df042959db53703c517156ce8895884a9769d92b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
