)]}'
{"id":"openstack%2Frequirements~905259","triplet_id":"openstack%2Frequirements~master~Ifaec6f13a15ae0a6c78816a545e2a5062bb7d3ab","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-18 05:17:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifaec6f13a15ae0a6c78816a545e2a5062bb7d3ab","subject":"update constraint for python-tackerclient to new release 2.0.0","status":"MERGED","created":"2024-01-10 17:28:06.000000000","updated":"2024-01-18 05:21:50.000000000","submitted":"2024-01-18 05:17:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905259-new-release","meta_rev_id":"ba9b2e090e5247a3c881fc9d6d6ed4350acf173e","_number":905259,"virtual_id_number":905259,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 05:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-18 03:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-18 02:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-18 03:38:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 18:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-18 02:14:41.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-18 03:38:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"30b1bb357bd3acd240553fa14b6ffbfd0b3be137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-10 17:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4742acf009b3812b39caf981cc6c0b250e603b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 18:39:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a24e8831cfd9462398c9ca9d5a0ad432\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d3dc2a03664bfaafa3bba26cd62c3d : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd04c8540a3b4f3cac0e152cc69382e6 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/342366bf733842d08f3de5d48481d520 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb844231d4ff407387e90c57d69f9590 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/971925d9b8214c0baac84bb5e8b1c4b5 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79636e7fe8584939bd238e8c88ebecde : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1341e6e79b5349bea4fd65b14bf7c4e3 : SUCCESS in 10m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f3007dde45d7479b92e3101d47a44053 : SUCCESS in 15m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7d7f152d8d0c41ff94f8d56142b99a60 : SUCCESS in 8m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a73239917b3749ae9481246199c621d5 : SUCCESS in 16m 58s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4c073a90a5b645b8add0c354fb7820df : SUCCESS in 5m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/afab6a59cd16490ab15f691f7aa16cf8 : SUCCESS in 5m 17s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b24f3ae2997e4707a7aaa00f295fa933 : SUCCESS in 4m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/05ab3b5335704ac2ba28207eb2632010 : SUCCESS in 10m 14s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/b453453514bc4530afeb8f58450e8fa5 : SUCCESS in 5m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/21b77d71e2514166beb1d89c0eca6a83 : SUCCESS in 5m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4cd0ddd716d94d389e83f0a9d4514f07 : SUCCESS in 5m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/330581655e094b409a524cdec2dce0d1 : SUCCESS in 7m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0aa9776953674fa6b35fabe2689839d0 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f5b098678e474e10a62b82f001620ee9 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/393389b3f9e74761844c9852926624fc : SUCCESS in 7m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/aca06286e9104c849e17bbd8d924d087 : SUCCESS in 11m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6e4de8b7315a4f0ba0452b194ba0af60 : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fe216c5d47dc47c2818d0ce24ce4b246 : SUCCESS in 10m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/917a6911b58d443a833b8d6c84de5783 : SUCCESS in 5m 25s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6ca8a15a0bdf48d1b25dc7371220c1f0 : SUCCESS in 14m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8a0c52c53e7b46dd833acd7115a14c93 : SUCCESS in 28m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a9589b9334cf4cb8a2e40cbdf095830d : SUCCESS in 7m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9b63b377092140a595b85d7a0d9a292c : SUCCESS in 12m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/846dc8337b8240548b847d9cb34664ff : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d15859c0609c4aa684008b8d45a5b330 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00c15c4132db455783e92bf5b854e203 : SUCCESS in 23m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2d483ca4cc9c4ee48b0ebdb03e776e3a : SUCCESS in 5m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4ae28ad80cb44dc1bc5a9f16dd0e0a20 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07bacb1866b344f1a30fa75ef5bcfd85 : SUCCESS in 6m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b8b0b878c5b84ab4960a24e5dbe2116a : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0635ccdc45b494396069b5065573252 : FAILURE in 33m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f893620d30e47a3971af338188a7a04 : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ee6e2ae3e302bbf48cabefa6dff75b2d1726aeaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 01:40:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e7f4397939a303b6faf144471d9baa742f9244a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 02:14:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b2b3afd24c5f562fc91eb65600f4d81f41d55976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 03:30:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e794d3a82a5141f6861133d69c2dd5db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60561632ae154f9bbb2a4ab6e2627a40 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/691372ad68574414af984d88004547c1 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b611d8de1f543bf84bed3a06e863e9e : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b856153abc740c0b228ba43bda41753 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6b251ab3d7b46dfae4a83dcde144d52 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5c45630ebd8474ab8f4e87a441c1e8f : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cbb48e8432ac4eaab4fea79e697e025d : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5ea751a42bc74be2ab33451a1e6c7527 : SUCCESS in 14m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3a7835bd6ce14a17886429d41e350649 : SUCCESS in 9m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fe858ab3303a4130b74f34c27edfb2b9 : SUCCESS in 15m 25s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6b1e6ff1f55a47f0b6e3f4fad16b98a0 : SUCCESS in 5m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c565a24756874b0b8ce2adf6a898c924 : SUCCESS in 4m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f2173104e01a455bb35179dc098e70ac : SUCCESS in 4m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a52c21145eb345448ad14d932fd90379 : SUCCESS in 13m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f2733b2a37684664a43a3a8801eb2903 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/11d44c7bb03e469cb4d3baef115a2670 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bc1170aa8c8f4f5f99057f71496310d4 : SUCCESS in 8m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/debf91f3175d4e88a049760dd82974f9 : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e11b8469932499697b3f34fb68fe156 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/28986f586c0a4952be4dc496d5c15c12 : SUCCESS in 8m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/68edbb0d33fc4601aec95d4de013d0dc : SUCCESS in 12m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0e01e575591048a9ae67edc2cc606c29 : SUCCESS in 3m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/37c752124ef248fa88af3bf56c3f8428 : SUCCESS in 10m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0682dc1b86a94fc68e9ca42ac7491b06 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6297a0e93ada4df798bcfb00d2b7aeb5 : FAILURE in 4m 04s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c808b51d858a4ceaabebe1c80fcb3f4b : SUCCESS in 27m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d244bf4ec6ae4b6cb16ce63af899eb94 : SUCCESS in 10m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/74cf606ff1e64f64978709d626b11ae4 : SUCCESS in 12m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/31a61b1f515c465ba09399898fa98b7b : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab3d862c9fe542b3a92515a396685851 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14b978a2c84f46dcac763f0227ffe2af : SUCCESS in 25m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/987f6b14da9f44a093fe9dd00a09ea6e : SUCCESS in 6m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3e6c7e84c4db4ed795c5d5303e371ce4 : SUCCESS in 7m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aab227980b764e468891c428ffea1d76 : SUCCESS in 8m 08s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bd82b4d98b304b22a9e24dd63d24723e : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cda75ef72edc46b1862cf2482ed5f8f0 : SUCCESS in 32m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90b444955c264e15ad8874f869df1e23 : SUCCESS in 1h 38m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7ef0a78b870589c288b02d87b1368fe76f42bb75","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-18 03:38:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a845629d2c3c38a82701d8087ecff11a23ddc2c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 03:39:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1432032f1614482783db1d6ff3c64e232255009","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:17:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33d780ba0f4a43feb0cabca1a6ed8939\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c27f21455b846578248c960cbc5ebd2 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47169575845443898e055c8c50a25b0a : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e210c941c291441c8a45d5cb6f3e1906 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/478caae2c6fd4d6586037712fb1120c6 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9b4889e1f2840dd99ae392206a77acd : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7c5ecee29af45ffac91a1ed0636ca1d : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c9c3f58c4bb46379fb3e6a0c975b752 : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fbba39488a074bf0b5e68567d188051d : SUCCESS in 14m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2206b43f11ed47dbb9086aea0c656eb0 : SUCCESS in 8m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/57f1b19451b24529bbaab4aa62af1698 : SUCCESS in 12m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c67b7d94149d4f028f415db1eb9b083a : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f02173d0d1cb4118abfaf4ea254cf376 : SUCCESS in 6m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/530c88a1c4e348f497c6aec80d94d437 : SUCCESS in 8m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e935c9d217fb40d7a00ed1a68609a43e : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ef9aa4947d84d0fae5a8f74ed378d40 : SUCCESS in 8m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/91189da3f80d4ae38f0f675ad7fc18ed : SUCCESS in 7m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7f7155f2df4c4058b8fad0689b692d7f : SUCCESS in 11m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8791169fc9064586b62c1e896875be00 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/04dc1b5c5ea841daacffed2c3c2b60c1 : SUCCESS in 9m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a39427fe2d1c4d63a5058a0e82aed056 : SUCCESS in 32m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/45d0fa41a25a4b539382300daa1373c0 : SUCCESS in 7m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/825288e94c7b42d3b84de445babe265d : SUCCESS in 14m 30s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/694b9d5b5eb146d5ba3643f56f110ab7 : SUCCESS in 6m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e55ff15c2fd94385b75cf61a1547a604 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f2a3d5d1cb4c440ea34ee5f84d996855 : SUCCESS in 23m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1d3948b417eb4689a7caa28db0b245f7 : SUCCESS in 7m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/644d36abac5f4c4f9f18b7d1fb03a70e : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1cdb9f43eb484e9ab1cb804a2dde0ace : SUCCESS in 6m 10s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c407dd0dc7234b1d995e800f62de0cc7 : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f7c3ceb3b6a4308a8344c264010eace : SUCCESS in 37m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26e3d8058a624d4ab3edb55a2e3a9e48 : SUCCESS in 1h 31m 05s","accounts_in_message":[],"_revision_number":2},{"id":"49c394e19d8433abef34e0aa931e65715fb589d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:17:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ba9b2e090e5247a3c881fc9d6d6ed4350acf173e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:21:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed86bff961046a2b93cbc9e180c9bf4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f83516bd17e4d5ca984d67ae46755cd : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d67991af046bb5a5c5a172440588f4b0bcf6249","revisions":{"0c68472ff5692d81e6f1e4aca7b9124f1e756aa5":{"kind":"REWORK","_number":1,"created":"2024-01-10 17:28:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/905259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/905259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/905259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/905259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/905259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/905259/1"}}},"commit":{"parents":[{"commit":"67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a","subject":"Bump Django version to Current LTS Version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67bd2cb1ca331ee0ee893d8b0d3eedef519fe66a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 17:28:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 17:28:06.000000000","tz":0},"subject":"update constraint for python-tackerclient to new release 2.0.0","message":"update constraint for python-tackerclient to new release 2.0.0\n\nmeta: version: 2.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I05ab405e457c0f35bd5ac66a227348333da189a2\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifaec6f13a15ae0a6c78816a545e2a5062bb7d3ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c68472ff5692d81e6f1e4aca7b9124f1e756aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c68472ff5692d81e6f1e4aca7b9124f1e756aa5"}]},"branch":"refs/heads/master"},"8d67991af046bb5a5c5a172440588f4b0bcf6249":{"kind":"REWORK","_number":2,"created":"2024-01-18 01:40:11.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/59/905259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/905259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/905259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/905259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/905259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/905259/2"}}},"commit":{"parents":[{"commit":"b353fd308448e510faaca9a068d22b51858d68a4","subject":"Move mypy to blacklist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b353fd308448e510faaca9a068d22b51858d68a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-10 17:28:06.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 01:19:22.000000000","tz":660},"subject":"update constraint for python-tackerclient to new release 2.0.0","message":"update constraint for python-tackerclient to new release 2.0.0\n\nmeta: version: 2.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I05ab405e457c0f35bd5ac66a227348333da189a2\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifaec6f13a15ae0a6c78816a545e2a5062bb7d3ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
