)]}'
{"id":"openstack%2Frequirements~905471","triplet_id":"openstack%2Frequirements~master~Id52e38e44a0d520ed0b27130e52095e9713b27a9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id52e38e44a0d520ed0b27130e52095e9713b27a9","subject":"update constraint for openstacksdk to new release 2.1.0","status":"MERGED","created":"2024-01-12 15:44:25.000000000","updated":"2024-01-12 23:49:30.000000000","submitted":"2024-01-12 23:49:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905471-new-release","meta_rev_id":"d9162cbe5dc83cf78fae454cfa1e4b7d152de548","_number":905471,"virtual_id_number":905471,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 23:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-12 22:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-12 22:03:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 17:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 22:03:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a51e49d3094d57cd35fa770775571835022c09ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-12 15:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5732f7bed1382abe434278d72e0e1018b229e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:13:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f37bbfced04c6387c7062e534d9e1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e48daad265e5405a98c8d862efb7e633 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52307acf34e04ef2a5470ae690664dc6 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c97363644f4e4d6aa933befc9a9eb4cb : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd64f275f37547238241fb7ef4818bd0 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd08ecfd2bed47049c153f56ddddf50b : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c2a2c0b8f75410ea93ebe6a2c4b75f1 : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/362aa1ce440d4055bfb55ece936cc65e : SUCCESS in 14m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2232ef3b66e44e25bbadb7d843f2b41d : SUCCESS in 14m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/10f07d2301494dc48b69de66fe2543cc : SUCCESS in 9m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7adcc14fcb9c4b2b974bca79b9225d14 : SUCCESS in 21m 02s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ba4a55775fc34949a2f9d84589249610 : SUCCESS in 6m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/73fa9ef28e1540b799d20792ced52c8c : SUCCESS in 7m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/cae42ddf2c2646e1be2dda1f514730c8 : SUCCESS in 4m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c31ce00261734864afb60c543093b16a : SUCCESS in 11m 42s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/27ba19c7a1fd4a63b8750059ae9c8107 : SUCCESS in 5m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/82931a9413894254ab3b24069f21907a : SUCCESS in 7m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3c4dc02bb90d45debfa583c7d5da98f3 : SUCCESS in 7m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2d582a3300a343d8bc8e37238f01b131 : SUCCESS in 8m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/89d3c4a706fd489e82f7cd450c3b30ee : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/edceb283e59d4b45b9ed74e100504b40 : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/97ef817d920948ddba1b7ea44615f401 : SUCCESS in 9m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/62c777ea25b14f76a63806ef0622e9cb : SUCCESS in 13m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fd62f1a667d24387a33854bb2e23744c : SUCCESS in 5m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/58eecf31cb7d46f9960a5e288e5f63fe : SUCCESS in 10m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/86e06f049ba94c9ca997409b6768df2f : SUCCESS in 5m 06s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2cad092ffdb64e65b4aefe9a1f49e2df : SUCCESS in 12m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c08dd9c13b264ffe86a07debe87fb73c : SUCCESS in 37m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9b13b04468b04694bd768dd0596be5ce : SUCCESS in 9m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/979eccc12f804da48f70b5724014922f : SUCCESS in 14m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e0f35e2ace3e4d48803fd5ba0db031ef : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bb5a582becd845d382a382290d3b471c : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8420cb3509174dc58c114b655e049d19 : SUCCESS in 28m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/910c36e5d60e475ea670842c5fd7b9fd : SUCCESS in 6m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bf616c2eeb9f431dadedbffaa21b8551 : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b8680862db548a29eb321ee60dad5de : SUCCESS in 7m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/39a6fc770e684336a737b043650b412d : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf9e643d2b7445f6bf57c55908419c01 : SUCCESS in 40m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94d62d21dfa9480190a231b29dabc504 : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e913faf33603557adbe7d200009ecef968c0f03a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-12 22:03:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"852e316ec9e1e094ba0fc3d137cc12e1d0a3c424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 22:04:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29a9ae6fe0bb3d96c98fecc0e34fa9cde8a92728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f17d9704843b453ea9210527eb01e6fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5a9cb409a646279f22ec34f5cc7792 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163e890eb17a4ada802725e64c0d63ee : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a943e5c47a234bcebb1c34687b4f03c3 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69484b78ee8f41958e65bcfb14226e08 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f66d0ec7b99549f4848df0516bc79fde : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0fc3a9c9b5d4515880724da5af54378 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/16a2af475d9d42f1b52ffd04b8a3843b : SUCCESS in 11m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/904e1d7c3e8b4f21949a8ef675eff102 : SUCCESS in 16m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b31a249b0b50454dbcfc8b2ab799eb10 : SUCCESS in 8m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5c8887cb9eae4c81bf969c4b8117d4cd : SUCCESS in 11m 40s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/d9ec8fdbc8b947b99e363262a5ab63e2 : SUCCESS in 7m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/adfe3355da8f40ed8d4e06172b94fc4d : SUCCESS in 5m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/01e1c306507742ac95a94dc5646cfb65 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6e2486c190544cf4859b737aaeb02189 : SUCCESS in 9m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5b1b4ec3c9734c9093c25e383f4941af : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dbc63d4cabc44c08c2c96d741098ca6 : SUCCESS in 7m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2d71aa4fb29641039002e00d2082ffbf : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0265e17825f841d8b8d6fcdda94e2028 : SUCCESS in 11m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6d0175b987b54846bf4b8968f1e2b858 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1a34963a1dba4f37a8f7fed87c437441 : SUCCESS in 12m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/00a2aa87605f45b8bb7d030449f56ec4 : SUCCESS in 30m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a32316eaa8b5462cbb638973dd088ad0 : SUCCESS in 6m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3894646122164af49c9bcdd6b6a9dfe4 : SUCCESS in 13m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/72994118e3fc4da083e867b920fa06cc : SUCCESS in 6m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f2a216345c540cd8c15cccfd971df10 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7150ba6e93814f6baa029601f6bbf6b3 : SUCCESS in 24m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6b5a790f2da84276891c0352e3c9c0e1 : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b2491c06412b4e269676c43ada44ec61 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8142e192d5374d2592e0cc06d9e99ee3 : SUCCESS in 7m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8ffa717ef5b1452c92db7034a75fa892 : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/613b148f823c445db154a74990cfce7b : SUCCESS in 38m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/512486a4a53c4ea78b32533c9f5f6e16 : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d9162cbe5dc83cf78fae454cfa1e4b7d152de548","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"398b7455a30a397d144ad5b8721cc7dd0b660cbd","revisions":{"398b7455a30a397d144ad5b8721cc7dd0b660cbd":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:44:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/905471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/905471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/905471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/905471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/905471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/905471/1"}}},"commit":{"parents":[{"commit":"5181514787ab65c4bbae40e7dac4fc9e73a86da6","subject":"Merge \"update constraint for ovsdbapp to new release 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5181514787ab65c4bbae40e7dac4fc9e73a86da6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 15:44:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 15:44:24.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 2.1.0","message":"update constraint for openstacksdk to new release 2.1.0\n\nmeta: version: 2.1.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc161c9e08f07b83f16fc7e54aef4bc24a11f1ff\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id52e38e44a0d520ed0b27130e52095e9713b27a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/398b7455a30a397d144ad5b8721cc7dd0b660cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/398b7455a30a397d144ad5b8721cc7dd0b660cbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
