)]}'
{"id":"openstack%2Frequirements~905472","triplet_id":"openstack%2Frequirements~master~I0e2d14774adbc62ad12673e4315156b023cc17c1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0e2d14774adbc62ad12673e4315156b023cc17c1","subject":"update constraint for os-brick to new release 6.6.0","status":"MERGED","created":"2024-01-12 15:45:44.000000000","updated":"2024-01-12 23:49:33.000000000","submitted":"2024-01-12 23:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905472-new-release","meta_rev_id":"b1b89a1981c9b7013358c8d86ed8c3b8ce04c9d7","_number":905472,"virtual_id_number":905472,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 23:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-12 22:03:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-12 22:03:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 17:27:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 22:03:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5df9b81deda7a792c1ccab97a39db547d30e457b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-12 15:45:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"115b9bc10c3b3b40c38a451eabf0200fd7a4210f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:27:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03ce10ed2d0a4b7c90626fed6ab7286d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7683c1288bc4305b7eb2f5ad930c67e : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59072223dc894c12bd38a88bbbce3065 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/400b08cf365247918848862aee499b8e : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13b484b1e6b5488b8179080def685f3a : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/229e9238e76f40acb817b3f65298e39a : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/239ca661f27945769ed1bc76ab3362a3 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e3f719a73b8943ffa5cb4749ca7aee3f : SUCCESS in 11m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e9205b2ca0bd4e4db6dba6779dd9559a : SUCCESS in 16m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/630ca10185964cd7be8303d51630d81e : SUCCESS in 9m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1dc7a34e028e46b19c9602924472c158 : SUCCESS in 16m 34s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0675fc82a56b492bb6b3cb2476f9106d : SUCCESS in 5m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7acfce7c36364ef89180d789cfffdb44 : SUCCESS in 6m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/264a9d9471764549b68da3190a211316 : SUCCESS in 5m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ed897b02d0ec40d49757374114279e6a : SUCCESS in 12m 13s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/8d0c5eb6e1ff404faaa3d4c5ccaf9b0f : SUCCESS in 6m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0e101e14a9904dbe88d858ce0b9b827d : SUCCESS in 5m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f93cf80a3150440bb958aae31148c99f : SUCCESS in 5m 33s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0155b3c3fdc04f96a42af3d7382e22d3 : SUCCESS in 9m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2c9982ba75c44faab4101ce41c7c07ac : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43ed6fa2b1c94f5ba12bcf3a3f5635eb : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cdc3a5be7e5942929fcdc67641f8d3e3 : SUCCESS in 8m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8c5d4449313c4b49959e931e0efad238 : SUCCESS in 12m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d6e9e6df574f42be8bef9ca46fd9d188 : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c7f1a3ea0d884588a20b284d01bad243 : SUCCESS in 10m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d63ce3cb085344569ebe9b5368e78f78 : SUCCESS in 6m 39s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/22d2a892f0464578a3843861f0ecd9c9 : SUCCESS in 13m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a22c0cc732424e039815f2b0113e73c0 : SUCCESS in 26m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0c1ed6980b2d41a8bb8b226bc75ba975 : SUCCESS in 7m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/92bd682975c6441da2a86b2e7130f988 : SUCCESS in 13m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c9ef25902c5840e3b40b379d382da578 : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7cbef279132d45eeb81e3b76aec4d2f0 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c43f66038fe47a8b4951c45c9bf785b : SUCCESS in 24m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/97d93ecf5e5140dda4ab39cf7fb71fde : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/46494828585f487bbdc9d3a034ed4245 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3de38b8723c84a90a8dbefdc75ccdea0 : SUCCESS in 6m 17s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5ebf36eb63a6443ca2bd5145ca9fe300 : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c254d9df0fe4e39b820d5902507d2ae : SUCCESS in 44m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/917659e296db46a794ddcedcb271f8ab : SUCCESS in 1h 35m 40s","accounts_in_message":[],"_revision_number":1},{"id":"54721a57b133beb3b891dd56a7addeb3c5ffc193","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-12 22:03:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"626244b1f43eafe4ed054cbd7a63fdd2ceb6f674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 22:04:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85f8b0fb1311a153408bddfe8d99279e58769a7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a91f96c29a49d0942fe9c568315030\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc44fe3335c4050b56c7b335703c550 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105a396a300a4dadab345023778d10a4 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/924795a9b26248c19c043b98ba8a006e : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b787fb58f1044dd810328ae9ce06642 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d7d3c3663314cbab1daf4507cb59084 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e43b7ecaaeeb459c81864ac06c9b56a6 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd512ea0a9e646a187d40839887309cb : SUCCESS in 10m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6d6eeb83c3a84d7382dd0dd295357a49 : SUCCESS in 12m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed85981572f940709f59fa0ea9a23318 : SUCCESS in 12m 09s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e935a4f3e563410796cdd386af74ef56 : SUCCESS in 10m 02s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/8f6d93cff4dd43d5ae4e5dc4973d0aec : SUCCESS in 5m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1ba73598ad1841639adf7667eb98566a : SUCCESS in 4m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9ed938857c4f4ce3b44aba9794e63034 : SUCCESS in 5m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/011f50c7f3504adb9d3d0cda91cad725 : SUCCESS in 8m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bd9b11d6805e46c8b618ee51a0ec43f9 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86b8fabe579942bdb1f8a38ea9bdc8b8 : SUCCESS in 8m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5c28773f1a9245aabfa355c005298a13 : SUCCESS in 7m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4165544849f043cbb8f51ca6c0cc07ad : SUCCESS in 13m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/320d591664c94cc5975e2e67f3352e10 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c76d1249b1eb425f9b820c7afff16657 : SUCCESS in 9m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/de198b8a7be74f33913558097a25fae0 : SUCCESS in 27m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc2f657f6a424e52b978b79109d4e9f0 : SUCCESS in 8m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/54ea194f1aec4593a9167cc7a9e32025 : SUCCESS in 13m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1fcc2f589c934850a3b834fd9fdf25b1 : SUCCESS in 6m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d152c50d353d4976b409117ccafecb6e : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1add84e137e846e7a216cf4b3f7428dd : SUCCESS in 25m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c98d9936deeb4ac0809e98a779017842 : SUCCESS in 5m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a241944f05e34b25b2d08cccd3462dcd : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c0152ece274049b692211121ab8f5288 : SUCCESS in 6m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4032bdf5bd824d05881863201ba6df08 : SUCCESS in 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3df2f9dbe18403baceb7fa7833a91eb : SUCCESS in 35m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/179de3c260ff4bcb9c363bfc3a6a2b06 : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b1b89a1981c9b7013358c8d86ed8c3b8ce04c9d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db8ea268c738b317f39c40dff1b3e610ab2b3955","revisions":{"db8ea268c738b317f39c40dff1b3e610ab2b3955":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:45:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/905472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/905472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/905472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/905472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/905472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/905472/1"}}},"commit":{"parents":[{"commit":"5181514787ab65c4bbae40e7dac4fc9e73a86da6","subject":"Merge \"update constraint for ovsdbapp to new release 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5181514787ab65c4bbae40e7dac4fc9e73a86da6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 15:45:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 15:45:43.000000000","tz":0},"subject":"update constraint for os-brick to new release 6.6.0","message":"update constraint for os-brick to new release 6.6.0\n\nmeta: version: 6.6.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: Ida21a81515b3cb928e3a9994f236906daff613a3\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I0e2d14774adbc62ad12673e4315156b023cc17c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db8ea268c738b317f39c40dff1b3e610ab2b3955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db8ea268c738b317f39c40dff1b3e610ab2b3955"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
