)]}'
{"id":"openstack%2Frequirements~905477","triplet_id":"openstack%2Frequirements~master~I67954f362b3719e20f0992d38ffb4ea8bbf22419","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I67954f362b3719e20f0992d38ffb4ea8bbf22419","subject":"update constraint for python-blazarclient to new release 4.0.0","status":"MERGED","created":"2024-01-12 15:48:32.000000000","updated":"2024-01-12 23:49:35.000000000","submitted":"2024-01-12 23:49:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905477-new-release","meta_rev_id":"d5f3320d98c531ce07ef39e659182b9e59a45aa8","_number":905477,"virtual_id_number":905477,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 23:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-12 22:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-12 22:03:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 17:29:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 22:03:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c39820e2cd9cf5750bd9b626766265b63b0e3529","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-12 15:48:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bee6a9d392df6a4a9159e6aaf33fe82f39a65afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:29:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83a62eb2ee4d4555ac6cf270d9b5f563\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca852a4ee6e495cb1c4251a3b585443 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d782d84608574ca3a2ec64e128f37d7a : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c61ec18249e4cef9ae175c13c7d0d67 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc83686ad460468a9f801f28652db60f : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99ccfd17a4124584ac193129d5f45833 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac8410dcacab4cf0934bd1bfb222cd01 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/77392eabd4174e7caaa30a3200a6e15c : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b6e99c1c9fd64758b9ce52e3a9f3e2ec : SUCCESS in 20m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2362718c8b2f44989dbe8eb5140060e1 : SUCCESS in 9m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/916cb7bdf0a6421bb9de05bafc93d2c1 : SUCCESS in 16m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8a847a73477f421d8acc7b8763d64df2 : SUCCESS in 5m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f7a01feb7fe74d19a438721f925ee1a3 : SUCCESS in 4m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5d8dc0501693494e8db2be72d413094d : SUCCESS in 5m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ad1076428b724cbe98095b4d718cbf54 : SUCCESS in 12m 04s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/48edb58519cb48308cb5444426691787 : SUCCESS in 6m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c1b22d7ae1c147589885b5dc2a10d50f : SUCCESS in 5m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/27e3687c0d9e4ecf8a914ea96df81cd1 : SUCCESS in 6m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6bc57e8a433349679cab54285c7562f3 : SUCCESS in 8m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bed95f8f9b4b4b4083fec80534caffec : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb981d5aad1a4a7dba124893d68595b9 : SUCCESS in 8m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e170364ad87a4a4282e8c8cc40c2afb9 : SUCCESS in 8m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/df4115b44e1d4eee8b03c81dc56f4566 : SUCCESS in 12m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/86a6bda8e62241cca831decab0fde824 : SUCCESS in 5m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1df454e79bca4d2d9b996c65370cf2e8 : SUCCESS in 9m 38s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/05ef9e158b82431ebea6b848e024c63e : SUCCESS in 4m 20s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e61dfc34e340498ca1cd8381d4717dc8 : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1d8be2aa27054a4aa6f2bfc0082cd592 : SUCCESS in 29m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/26e49d5099c946d1af617a049805e3a0 : SUCCESS in 7m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2fe459e99dbd4df1a78ec21cd4be8500 : SUCCESS in 12m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8c96a2f9c31246aeb89aa55c25d3177a : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed47b89601dc4eb787199dd50da48ffa : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a444aa0773014bb7b0937291f1e66e01 : SUCCESS in 28m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a05db510ef0141a69c85449878b666b6 : SUCCESS in 5m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e88ad84d2fea4783ba8e00aabe13cf6e : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12ebf4420dd04de2a5796c2b14026b87 : SUCCESS in 7m 23s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/38b21d9147aa407b9b0ad5dfc6c81a42 : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a67899c4e5ba404f9623585e617ddd9c : SUCCESS in 35m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b20d6183a694d269a493a3198ba187d : SUCCESS in 1h 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"57337f693bde9180b2248405f631164fdbd108a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-12 22:03:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86a114c05c0d933d4d048e8ea72655aa0de25236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 22:04:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d8b87a8163df57b5948ce2a245b950ddbf92811","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c62b3203e549f185dfbbf5cb1d281d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72b9a5dad3845d4af040aa2de573c31 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130aa1befbe44f548a02a2ddf4044c85 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6fa2d6233114b67b90074c37c927a3d : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f975c0fd9b540a8ad9f6a35459f950b : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db26fd9d659d4a3886f1bf38e86381df : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa4e5332cfe74f038b1b288e4b22d49c : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/287d2604f4234ab5abe682e9228ea1d6 : SUCCESS in 11m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a507d2cfbbe24273bd99667ec0d6b0ba : SUCCESS in 17m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/67f3c20250ae4d3eb7ee1ece3b15c88b : SUCCESS in 8m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/56da5b5a1943443289a5228467d766a6 : SUCCESS in 12m 59s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/13c181c70231436a989dfd1b0bd145b6 : SUCCESS in 5m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9a80d81d7ab24cf7b9371178caa4e805 : SUCCESS in 5m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cabe99bf2b624ed2a6ec4c817e60b5a3 : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7e729e7cbf56497eaf6590298a247f08 : SUCCESS in 10m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1fa20b8b86be40f8b1953ebace042879 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cff6fe515d3d43a286bed0f45b5d9d71 : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5128d00ebf6c4f5089568c08076ad108 : SUCCESS in 8m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2e660937b1e043eaa6a17bb277f921f7 : SUCCESS in 11m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/004c1e980b5f401e9aefa0ac54d8f0f9 : SUCCESS in 5m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/18e6f774e2724400a3bfb0441c6bb541 : SUCCESS in 9m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e35c04e042e74e3794c018eb01aabc28 : SUCCESS in 24m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/efd79a101c4343f28c8554fc26eaee01 : SUCCESS in 7m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b29ae552c81b43f7bd215f13ed84b2d3 : SUCCESS in 14m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7586f3a8230a4875aae820b4c32f6c9f : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc3a8fce33eb4da69203e2f2f1666e54 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a8b9b06e6664265baaf5cef50713fc0 : SUCCESS in 27m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d5463a2854824f88b2ee8d8b00e9ee39 : SUCCESS in 6m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/25551f48a8304f469dd6d7807bef5fc5 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7d02c8c4a529405292cd1b6929bbdc6d : SUCCESS in 6m 27s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ae5709d54bc5446d9a87cf44046b935c : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/602626014a4b4186a6a18e683286fc98 : SUCCESS in 37m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c16c4febf9af4645a47edd809329f489 : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d5f3320d98c531ce07ef39e659182b9e59a45aa8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a388c492bf980405378ad7061fc0366004fe73d","revisions":{"8a388c492bf980405378ad7061fc0366004fe73d":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:48:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/905477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/905477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/905477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/905477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/905477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/905477/1"}}},"commit":{"parents":[{"commit":"5181514787ab65c4bbae40e7dac4fc9e73a86da6","subject":"Merge \"update constraint for ovsdbapp to new release 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5181514787ab65c4bbae40e7dac4fc9e73a86da6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 15:48:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 15:48:30.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 4.0.0","message":"update constraint for python-blazarclient to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3e378d0f08ea509de6c887e833c19e071b79fcd1\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I67954f362b3719e20f0992d38ffb4ea8bbf22419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a388c492bf980405378ad7061fc0366004fe73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a388c492bf980405378ad7061fc0366004fe73d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
