)]}'
{"id":"openstack%2Frequirements~905488","triplet_id":"openstack%2Frequirements~master~I84d9c6c434b667115ea8a826b4c237b002ef2233","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I84d9c6c434b667115ea8a826b4c237b002ef2233","subject":"update constraint for python-troveclient to new release 8.3.0","status":"MERGED","created":"2024-01-12 16:00:17.000000000","updated":"2024-01-12 23:49:38.000000000","submitted":"2024-01-12 23:49:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905488-new-release","meta_rev_id":"23c673a97cf975a6fb35e0637204ba80539db241","_number":905488,"virtual_id_number":905488,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 23:49:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-12 22:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-12 22:03:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 17:39:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 22:03:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5d3c4ca5d51a63a4cae03972a9090707ffad8682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-12 16:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88ff891254ef11f4b02b738b015cad9054e8ed96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:39:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abfebae10f0b49caa39b84cdb4c43b4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5a6a4a6bf24a0c88432b02cd6717af : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18fff48c4804c949127eb0d2f54904e : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/583c6e7ba1fe4e469e61fce2617fb713 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/508b1d6b158f4512bbceb6b271111786 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb6ffe69ce564f198086e2347acc2b45 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bea0e26a3cd45e6978ebcbf13c80cac : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6299fc84619f423c94024e8ed85dd2ea : SUCCESS in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/265c8f86e9194556a3e7f345d1af4a51 : SUCCESS in 15m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c8514c2356aa4dff99b632bf78fbfa4a : SUCCESS in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/06c8193f65a0476299b778bdee598c5f : SUCCESS in 16m 37s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d2194f489dac4f5a83e136cbcdb7cc5f : SUCCESS in 5m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bf35074160c44ebaac8ffb6316e5b4ba : SUCCESS in 4m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b534cbf0252a4f2da3b68fe9d78d222d : SUCCESS in 4m 31s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7d0009d98f214d4f9427f17960c23acf : SUCCESS in 10m 42s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/e60c8bdd1c564d8da5ad12b83bc981af : SUCCESS in 7m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/64c5d14b2c4a466181b4155f52fa00ec : SUCCESS in 5m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/01462c4fd43b40eba6a15d1bb9e46471 : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9af83abc128d48e9a3d379d6b876ad8e : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a757422763024b1ebdd3de7c8f24f44f : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c5f26daf76c41c98dcd08c2634335f4 : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/553ca4ccfe524b6e8382c6f2a57d06de : SUCCESS in 7m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3ac056014363401ebda2a5e94dcc98eb : SUCCESS in 13m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3901732fa7024f3aaf00b14fd69c6e65 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a9fbef9e28944d9db32a84c471740eaa : SUCCESS in 11m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4c8107caf2914ca5b840d2a3ffac64f7 : SUCCESS in 5m 09s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/37e025e613a84e4fad99c8ea5109a130 : SUCCESS in 16m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9d3c75668fc042b6833c09b94edc5494 : SUCCESS in 28m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/292e4d09d60247879fcde8f8ce97b235 : SUCCESS in 7m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c4a324a8c0f8466f9b61587d76a75738 : SUCCESS in 13m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b41b7846a0554d9e92547dc4fb6bcc21 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc269e852c3f4f2c95f20e42ebbf5853 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0bc5eb9f73fb4ddb95d0de82eb42518d : SUCCESS in 23m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/239e6220e1b04b19a5131a2b8a4d3dbd : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5707c33a5a7d402395406d93a4def7b2 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d02340fb9d04f6da2108e554a417645 : SUCCESS in 6m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e60b13058b7d47edb28eece7d5d5ae8f : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91942355b8e244c49185bf489a86ed4d : SUCCESS in 38m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49c01b98b30246068cd7b8175bf77641 : SUCCESS in 1h 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"104b01e0df3442353c82a723572bc4092aadcc5a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-12 22:03:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b11295f7e26bcc04ebf379b5d75e696de84b4d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 22:04:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c1aad42f1f5538153d742af50f1efe8f4ab1140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6922206f454c499e1900627136612e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2da9b06e297482ba39993ee425dc230 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b292f735ad40f39e3c6b5385929671 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/363898c78b4f4591b8a64070063573d6 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d6f272e5328442f9ee69a0d917f9086 : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19021ada370b434a95a3e8485d0f8527 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03cf78c5717945ae91fb4350bd3cdb1c : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee203c6410f346f882845420082bc106 : SUCCESS in 10m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3167b799e2cc4047946f26f84277c13b : SUCCESS in 17m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5c607c4f432441b3900bfcef7ad39af0 : SUCCESS in 10m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/118baec3bc6a4ca287a33b506a39b309 : SUCCESS in 10m 43s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/1e8f6465d6b14eae9549fd24f284d109 : SUCCESS in 5m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a90a41f5984d4cf39961c6f326f6e5dc : SUCCESS in 5m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/028ed3143ffb4416824666a1a1c950a3 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1d4aed43af934bdbae45e7dbf7d21166 : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d864c891b53146dea300145eb93d0bed : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/30de9d2db3e749a6bfb720a66727f2a7 : SUCCESS in 7m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/df281fdbfcb74886ad252ea0a62cc7b2 : SUCCESS in 10m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/27e024e091094ba09c35df7373a98c9c : SUCCESS in 12m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/62e8c79c59ac483188891927bec2e7c5 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2c9fdb214895467f9b40d9b8b18ed2a6 : SUCCESS in 12m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3b9d7c36ebe346ec8054e0d975e92607 : SUCCESS in 24m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/456111c3a66246c6af1a7bd80caf8ff7 : SUCCESS in 7m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ed14a08b154a4660ac7a4ea3dbd7db60 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/de1201ab7168414094779e787dd8e08d : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5cf40e29358e4282b253fb0ba51a4d3f : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2680d7bdb1e649b3bda04fa72f4ebc4b : SUCCESS in 24m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9fdefd5d7504486894229a3174f8e10c : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/39a30cc1c22f4a4f87ae938c901f0468 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4cfc65a9b6b430b9cfdbd6dc39d97a0 : SUCCESS in 6m 58s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c2ca729763d84f3380bbb6f3d23ecf3f : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/602a728b4b4d472c8913fb8c23accc7e : SUCCESS in 36m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e7f4771972e4c468cbe3347189151b5 : SUCCESS in 1h 29m 30s","accounts_in_message":[],"_revision_number":1},{"id":"23c673a97cf975a6fb35e0637204ba80539db241","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4a67559972f2d7dd95f964d33c28fd1fc13f70b","revisions":{"f4a67559972f2d7dd95f964d33c28fd1fc13f70b":{"kind":"REWORK","_number":1,"created":"2024-01-12 16:00:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/905488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/905488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/905488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/905488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/905488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/905488/1"}}},"commit":{"parents":[{"commit":"5181514787ab65c4bbae40e7dac4fc9e73a86da6","subject":"Merge \"update constraint for ovsdbapp to new release 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5181514787ab65c4bbae40e7dac4fc9e73a86da6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:00:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:00:15.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 8.3.0","message":"update constraint for python-troveclient to new release 8.3.0\n\nmeta: version: 8.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I27e08158a739d95b72b7cb2eff59b9ac9335c29e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: wu.chunyang \u003cwchy1001@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I84d9c6c434b667115ea8a826b4c237b002ef2233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4a67559972f2d7dd95f964d33c28fd1fc13f70b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4a67559972f2d7dd95f964d33c28fd1fc13f70b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
