)]}'
{"id":"openstack%2Frequirements~905490","triplet_id":"openstack%2Frequirements~master~I844a10967f54d75652372c3d6698b255bd34648f","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-01-18 16:52:49.000000000","reason":"removed on reply"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-18 18:46:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I844a10967f54d75652372c3d6698b255bd34648f","subject":"update constraint for cliff to new release 4.5.0","status":"MERGED","created":"2024-01-12 16:01:56.000000000","updated":"2024-01-18 18:47:56.000000000","submitted":"2024-01-18 18:46:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905490-new-release","meta_rev_id":"ae7eb2ea2cf4445cea6ea2eea8226425de3c3b72","_number":905490,"virtual_id_number":905490,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 18:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-18 16:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-18 08:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-18 16:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 17:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 22:03:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-01-18 02:17:21.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"db76054b66f8e7f523c118cf6c95bc6e7a617398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-12 16:01:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b071897fc05ff33b686a90cf1f79961c56bd955d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:32:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/801370b34f1e45708768e1e3411bddc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/283a0d6f91ff4447b5e3b7720b5b3a6d : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/004879b6a5194b82aa557fde63bb2544 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f490506c32c490da2176a0145fa1813 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38fb79458c244f2a8bee68271b4295e9 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ccadff1415504f74ac44a4634157259b : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e5b8fffc3724869bf9d086d37358d21 : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dfe8598086304272aca50ce47d1c2c79 : SUCCESS in 8m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae41979227ab4c1983c50cd250c4d065 : SUCCESS in 14m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0103c9373a1472bbcf3c71287c83672 : SUCCESS in 9m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f2a1ca2c35e44b1aab98a1166fe3a210 : SUCCESS in 20m 14s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b45dbc9a234d496cbecc3230cae6361b : SUCCESS in 4m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4737734d0d504976ba517a0d43b98660 : SUCCESS in 4m 23s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9796cbe0ec324226b5fad7fde14fd569 : SUCCESS in 4m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e2899ea457094a0db215fe873f3f1b0c : SUCCESS in 9m 58s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/7705e02ad3f24b57950291469f9cf9e8 : SUCCESS in 5m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/647b1008f4984d1c9e832705b9f92965 : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/23bef0ca94ba4cc8bf6693f5577a6070 : SUCCESS in 5m 35s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fe86bb92641846c6adc9f4c54f637a0d : SUCCESS in 7m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c7aa7d1ad83e449b88ff5e5980fef44a : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e4c33afed474fcb992b2a13c1d80631 : SUCCESS in 7m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4231aa07eb2f466fb5965847aff5d1c9 : SUCCESS in 7m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f9c75542da0646f7b6bf60b2124e3398 : SUCCESS in 10m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/38602c66a7144b3d92275716d13efd00 : SUCCESS in 4m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1198eec17b8b41949c934f40580055bc : SUCCESS in 10m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/68b85942b7e944abbb1f4cad7032abbd : SUCCESS in 4m 00s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c7a62f1652ee41b5b8c867bb6616aacf : SUCCESS in 12m 40s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a0b9168ba8b14eecada5071236258c3e : SUCCESS in 27m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9766cd592ad549d1a19c4dd24b7cf2b1 : SUCCESS in 6m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4e5049b6a91347f0b5fab6aaa2b3adea : SUCCESS in 13m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a7233474e7b24b2cb8b8bbfadc61e972 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dffab4c1af0a4740a55ca36235d0e17a : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be5698e9518b4e07a13a05362b10c387 : SUCCESS in 25m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/57428de0ba6741c58217eb2e540bcb71 : SUCCESS in 6m 51s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/419090782b234e85a02c1da7d275811b : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/584814ff3ad54896a4edeaf67dc2220e : SUCCESS in 7m 08s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b1537c8e68134e7dbfbb4190e128d4c5 : SUCCESS in 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f575ed17f0d14c2584d6da759d7f35f2 : SUCCESS in 32m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a629ce714f542a490a191673e93ceaa : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"322a4d1701b069aa0411586cec080bf43e459c1e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-12 22:03:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88ad63921fe838f0f417e32e0fefe7571c5cb871","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:49:40.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 905490,1","accounts_in_message":[],"_revision_number":1},{"id":"04f4c6208941548d3bd6995e73ae1251da690134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 01:40:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cadd9f8ac2a8881574bdd6e2a0efd27ad1635ba0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 02:17:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"68e8d2eaa5512ac6c512817617440b4dda4d63fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 03:42:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f64c866ee84a4ad28bf3e9762e67c142\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0103de59624444afa2293e06730e1e85 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6e50bc107a4c179b64955105ff9dfb : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/265409ce6bf847e7be77c4b0c0216369 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9579573380e42b0821826d5b7571837 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0196445ed046425bad56ce0052832aa1 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/853ec11af2e244e5980e61592dda1052 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c0634edd542e4779891a3946cd81828e : SUCCESS in 8m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/878558258cd1414f88f2318d4ceeb29d : SUCCESS in 13m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/71a1883e938a4fae9ba95c26403c9416 : SUCCESS in 11m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ef9058b4893a49dcb2b83f9f25f4e938 : SUCCESS in 16m 37s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8b8be9b6a5114c7c9b8de3286e59d6a0 : SUCCESS in 5m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e656837a018d453e8ea1b26b5aadd780 : SUCCESS in 7m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a9e9bfc770a54f409ac8ef1785ad3260 : SUCCESS in 4m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6a4adfca5e8b4a88aef0f1e0d86197e8 : SUCCESS in 15m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/be5b108646f746aaaea48701f7c78a43 : SUCCESS in 5m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/de4cda1b17214db98aa21d535bd79ea7 : SUCCESS in 5m 42s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fce0812c04da4e54bf0acd3064c63db6 : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d5977c794e2d4a0a9cafd42f61312fc2 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4937e40f7763456a9e7668d73c936841 : SUCCESS in 8m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6752b7fd7a5848019ff74b1e81a22ba6 : SUCCESS in 10m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/98e9eb69620441e8a779bb90e397f56c : SUCCESS in 11m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b701b4cb880642abbefede2a0641e377 : SUCCESS in 3m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6ca3559e6e304e159d7b91b4e6969e89 : SUCCESS in 10m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/db49379fda334d748c4bbb8e521f4b16 : SUCCESS in 5m 56s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/85a80e1232e844f79a6378d6e460fc22 : FAILURE in 3m 14s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e71e71a3053540f58bdd5e925d965b0d : SUCCESS in 30m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc270bb2757941edbc814ef87828e0f6 : SUCCESS in 8m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/69ebd19f7fb1468ba2c02958f9b60345 : FAILURE in 17m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/007a715e0352404798d8cda00e581d03 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd53f53bca3b492d88f7dc4ed2308074 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9dd0220134d84102b8f7af99e77b5896 : FAILURE in 27m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bd5c156a1b7449c294063795281a4746 : SUCCESS in 6m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/daeb1c4e891a48a691ac5a8ca0633664 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eda1e21a103043e8b1edb460862ddcad : SUCCESS in 6m 37s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/371fa592840045d48a00d538123b71cd : SUCCESS in 13m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdd5654da43c4ef5a5aa1080c00fa3d8 : SUCCESS in 36m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f869103949bc4b539c1e2b6e413789a7 : SUCCESS in 1h 51m 23s","accounts_in_message":[],"_revision_number":2},{"id":"948304a65b90e6c3f03e03d70cfd398a30308403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 08:55:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ca2c1a96bdb47d96627ca4c28b75bbecccd5efe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:30:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d79e1dd054a94532b2d1e193b4501176\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515b677f10c5436fa54eb675d1d8fe89 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972a8662c1d446e688a2a6e5fa3c8730 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ce8c21a7f15464fb091fe038df2caa6 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55363fe3aba248f08ecfbcf8921507f1 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52ed54d6429c4270add44085c1c934ad : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3fec29585a0b4a06accf330fe91bf80b : SUCCESS in 4m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/392b2e90e70b4e358a3b6d8154abc548 : SUCCESS in 7m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/41f32c685a694592bebad01d5b6f0a11 : SUCCESS in 15m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ae5cdcb6da0440f2adace76b2f2671eb : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9fdd00fa314f4f899a5d7eb113423208 : SUCCESS in 14m 46s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ce2a7909ac2045ef86d0a9137315bb6f : SUCCESS in 5m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0aaf31827c3f4286a2585b04b30e2ce1 : SUCCESS in 4m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1629f08faaa4462788ba5f8fd5b50bad : SUCCESS in 3m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/70301ecf7fe7430bb9b0afbdaa37ac23 : SUCCESS in 13m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9dab2045d2ac4fa4b9a38551780798d4 : SUCCESS in 6m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e45f5d0ff7fc4b2cac8f4750bd255a6a : SUCCESS in 5m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/531ce514a5ec46a69a13fcc183b59cbf : SUCCESS in 9m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ed2e148247a74f809be7932a892e0e64 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52ac820e71774f41b4a99346af593e12 : SUCCESS in 7m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a6489359589442aeb1fb4fee85e8bdb1 : SUCCESS in 9m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d1ab62fa0e8f4dd59dbfd3106cf61958 : SUCCESS in 12m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f1edaaad8ae64b1c98d60ffdd908f202 : SUCCESS in 3m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2511e672733a49adb85e5ca32577085a : SUCCESS in 10m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fa2b45d8192e4fdcbe2765fdd4559ef2 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8897855be6364b8c97e65aaca499e3c9 : FAILURE in 5m 10s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4544d8f8851b4e2793b0eec9b0729a45 : SUCCESS in 26m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/778e7462700a41a69515dd707ecbc069 : SUCCESS in 8m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/48f389e7dd3742559abcc1b507856af2 : SUCCESS in 11m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/190aa0e79e7b4085a46864b515abf6bf : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab6b702a58334714868fd01dcdbed304 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96b0089fbe804490954b11ea959a88af : SUCCESS in 20m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/29fa868c2d074170b982a64901cb666f : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/035ba1af7f92401d9e91c48cfdc06273 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/60e47f1ac1ae4d9b920806afbc89dd88 : SUCCESS in 7m 16s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6455e8d888244ac3a642bfcd583a5fa9 : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abf09181286f4fe7aee77ee6c5fb1338 : SUCCESS in 24m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce3d9a2af8454c48a827f6240e0e9628 : SUCCESS in 1h 29m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a67cf3bc43c633414b3a5a0299b37ac491871440","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-18 16:52:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8c01da56a05c5bd29fe36b6be5c27ea57479d8d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:53:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"020144390c39eb98fe276da0a7019288f7a2fd27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 18:46:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cdca9a503fa4dd3a7851bc8754b6fb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41d59db20cd48a5b3339cb4d66f8eea : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e774a6a9471e46afa781f1d9533964d4 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c44cfe0e3be47bd88273b446ccfe967 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62297ddfd5fa4e94844beda4eef46996 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/478e2972c09141148803b0518723bf80 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71f55e55e188422e99a578952e258391 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14d119045a3045afa401ad019a149d3a : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/622b4478d3284456b1fec87a8a49f148 : SUCCESS in 17m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5bc8d531154f43a4b813a695e676dc66 : SUCCESS in 10m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d4a82de3c9624b73b82f152652883f2d : SUCCESS in 10m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8aea639f26fc4891a6e6b89070314ef9 : SUCCESS in 5m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/44386ed9b8254e2daf7e2c719bd855f6 : SUCCESS in 7m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1afd0bccaf904079930cf03cc04f675c : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0e0c84dccb41485d941518f29f3e2049 : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad6d0c35561c4cf3b07ba4938ee07f38 : SUCCESS in 8m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ec8baed51e8b45c29a13f50947cba5c0 : SUCCESS in 8m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/16a942fa9add451aa972e3ef60a33e77 : SUCCESS in 12m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6ff6072154084f37b96345c76449fb99 : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ba612032359c40b6bb8fb2912e9ae36b : SUCCESS in 9m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9c1230cda6044ff582a34def28ca7292 : SUCCESS in 25m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/78dc56f07f814ef5b9056bab255d8412 : SUCCESS in 7m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/deb2f9ff904a4e93b72b1e93430c5c31 : SUCCESS in 13m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/66c3370d84e94fdca1474deeda851323 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/402656c012174d49bbcb2bd8639af6cc : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5752923c14d34430b971363bb967fc2d : SUCCESS in 24m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/177780f962e9423a8eb959254b8bd71f : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/02c0e56637654a79921dd17831f6772b : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a754cf5971ed40aabc6519486d125d82 : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/32998e388b4f48b591204c0de0b5249d : SUCCESS in 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f871d4b1097c4463b9cecf2a8b47050e : SUCCESS in 26m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1f3f202c643411e8043560b26c22daf : SUCCESS in 1h 31m 51s","accounts_in_message":[],"_revision_number":3},{"id":"8b1b729cad447569aee307914b3a71ae824c1ac4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 18:46:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ae7eb2ea2cf4445cea6ea2eea8226425de3c3b72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 18:47:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edadc3d2ee0e48fab751375075e910fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a152f86ec674dbe929085cbf6ed5203 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"50b05f99264638161c0dfd65c056bfacf2a90c49","revisions":{"33fca681a5669c928b4e094b55fdff37e3b4305b":{"kind":"REWORK","_number":1,"created":"2024-01-12 16:01:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/905490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/905490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/905490/1"}}},"commit":{"parents":[{"commit":"5181514787ab65c4bbae40e7dac4fc9e73a86da6","subject":"Merge \"update constraint for ovsdbapp to new release 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5181514787ab65c4bbae40e7dac4fc9e73a86da6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:01:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:01:55.000000000","tz":0},"subject":"update constraint for cliff to new release 4.5.0","message":"update constraint for cliff to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I190204568dc82b9155ba54851c780a113a342977\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I844a10967f54d75652372c3d6698b255bd34648f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33fca681a5669c928b4e094b55fdff37e3b4305b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33fca681a5669c928b4e094b55fdff37e3b4305b"}]},"branch":"refs/heads/master"},"87477af3c6d1c87ad14cd4191438a6220780187e":{"kind":"REWORK","_number":2,"created":"2024-01-18 01:40:11.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/90/905490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/905490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/905490/2"}}},"commit":{"parents":[{"commit":"14d6fb646b81cd2e9951d44b2b72e803b375b949","subject":"update constraint for ironic-lib to new release 5.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14d6fb646b81cd2e9951d44b2b72e803b375b949"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:01:55.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 01:19:22.000000000","tz":660},"subject":"update constraint for cliff to new release 4.5.0","message":"update constraint for cliff to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I190204568dc82b9155ba54851c780a113a342977\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I844a10967f54d75652372c3d6698b255bd34648f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87477af3c6d1c87ad14cd4191438a6220780187e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87477af3c6d1c87ad14cd4191438a6220780187e"}]},"branch":"refs/heads/master"},"50b05f99264638161c0dfd65c056bfacf2a90c49":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-18 08:55:24.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/90/905490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/905490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/905490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/905490/3"}}},"commit":{"parents":[{"commit":"ef58ffbc3e86d146f33f4f3ce2812edef7280f47","subject":"update constraint for ironic-lib to new release 5.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef58ffbc3e86d146f33f4f3ce2812edef7280f47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:01:55.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 04:01:18.000000000","tz":660},"subject":"update constraint for cliff to new release 4.5.0","message":"update constraint for cliff to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I190204568dc82b9155ba54851c780a113a342977\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I844a10967f54d75652372c3d6698b255bd34648f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50b05f99264638161c0dfd65c056bfacf2a90c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50b05f99264638161c0dfd65c056bfacf2a90c49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
