)]}'
{"id":"openstack%2Frequirements~905495","triplet_id":"openstack%2Frequirements~master~I6c968c88115295a5fea3c2882769dcb7b155e60e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6c968c88115295a5fea3c2882769dcb7b155e60e","subject":"update constraint for osc-lib to new release 3.0.0","status":"MERGED","created":"2024-01-12 16:09:35.000000000","updated":"2024-01-12 23:56:18.000000000","submitted":"2024-01-12 23:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905495-new-release","meta_rev_id":"caf2d11e683a70a27af7ddb6cc6845ae96d84cca","_number":905495,"virtual_id_number":905495,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 23:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-12 22:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-12 22:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 18:16:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 22:03:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bac9f7a8ee1c67774b194d3012a38502bc2ba9bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-12 16:09:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ae66e5840236a1884f6be234430202a7eeda515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:16:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6080f342fe2e404f9ac66d49d7c5eca0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12de2b1ca824d70b4770c56b99faba2 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cddc78a641a40719c804f1b7ea7cb85 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeb04e6d2fcf470b8e121bd797aa9581 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b746aa500f104adf926485b117316147 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c25e99198704ebfa176749f7cc82426 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da7a041d7edd443aa6bbe378bd88a348 : SUCCESS in 4m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a37d81e812de4ccfacac35be09147ef1 : SUCCESS in 9m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7727806c41c24b9eabed04b97fd2c74d : SUCCESS in 15m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fbb9d8e0e0044a4f95dabf72d5361bd0 : SUCCESS in 9m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31c262aff39640258b9d61f4b2c7043a : SUCCESS in 20m 26s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8a45c92d562f427c8c7131773f78625e : SUCCESS in 4m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d0b187981c6546909d67b0edc563b70b : SUCCESS in 6m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/38eb72ddac1f43f08d0e00eadb59fa3e : SUCCESS in 3m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/158de3b559f941cc9475ba975fc64c33 : SUCCESS in 11m 07s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/cf54f92566294377ac646d9020aa264a : SUCCESS in 5m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/37bcc72d1e99483a8c69fb40602a7a02 : SUCCESS in 4m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/df6954c5294348b8906a34e3432e8ab9 : SUCCESS in 5m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/40d6b39d2b4d4e83bba93008e78c7a50 : SUCCESS in 8m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bed88e373bfb4ce79134a50e6656faa1 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8dd070e405114e80983ba2cb0f1c8a0b : SUCCESS in 8m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8bead7d7786b41bca714ba02f5f5b9f7 : SUCCESS in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9200158f2aee4eecb213436d1dbdcb06 : SUCCESS in 12m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fc9b424195b44568a470149464c7b563 : SUCCESS in 4m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3687548593d44098be56ced7370a3476 : SUCCESS in 11m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7b18a22939c14b88b72aab1f3441aaf7 : SUCCESS in 5m 02s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bad7edc5bd624cdba242fd1b78c47b8d : SUCCESS in 13m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/03b7775d3346476eb7cd97bf346ee848 : SUCCESS in 25m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e8a878441e2a44d386a07a092c35eb7f : SUCCESS in 7m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/24d0b1351dd74adcb71938137be098bd : SUCCESS in 12m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2da1fa0800a74b1face2de419a467cf4 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6eb8414a34b44dc8883d85f1919eb50e : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/530f8fe1c7744a92a0bfa11f7635403d : SUCCESS in 25m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b232daf184994496b9441bc001911c64 : SUCCESS in 5m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/47ffc2ed1c6447ebbe6b51279cbb3efc : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52ce37401df743c488b11d181011cdba : SUCCESS in 7m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c0fdde84bfee47b2866bdb2c99a7df5b : SUCCESS in 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1744ab959524af6bc79fed0ccd3fced : SUCCESS in 38m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb5d83c836ca4ce3995b6a1a920c8a04 : SUCCESS in 1h 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"24313f4bcc5e62d29161936e0151d4b0c7f93a22","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-12 22:03:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eebfda1cbe0b1d619516d25db87903e8b6ed0263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 22:04:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1cc31c3f2812a0d3e7f8a816b79878d4fbf1143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:56:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ec27639b6a9415d818a31e8317a74fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10cca51689a6480aac79638bcf0c4e10 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a1ebbc4c334e46be203afd8448c3ca : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9821c074e28944f4a2ac5f1626bc978d : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ffbdd4aed1d40a0b4c5604092098744 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0dcb6f14dc6d4d51963d8d579a86c742 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdddc1aff707487588199b4a9b25b25e : SUCCESS in 5m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3753145640bc457286a0bccea1285106 : SUCCESS in 9m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/439e73255c4d4a649f52ba3604970869 : SUCCESS in 14m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4e2ddf03636a4416b38b00af2fcfeebc : SUCCESS in 9m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bcefe1c728d442058e1204a5d7a07223 : SUCCESS in 10m 24s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/4752171880b24ef89c25538b35ef2d32 : SUCCESS in 5m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7948611922e0409fb954ce6988514287 : SUCCESS in 7m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b609422edd79404f8758bc12d69df4bb : SUCCESS in 6m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1f4f8e7ff35c409cb31dbd053ec47f44 : SUCCESS in 12m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0e48479de661447ca6cccef09f4866bb : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f57d7ca6c1c4d51a05a6154764d5cd8 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fd00c07ceb4b4cf5902d532f4177ef09 : SUCCESS in 9m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a95ed283fb6f46d1b66c2e6e33803c1e : SUCCESS in 12m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/50d60a15098041ee8e62afd31b360a9c : SUCCESS in 3m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/781a7ce4408041b7ac2ecfa66c9c36a0 : SUCCESS in 9m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2f8e485a8cec4468bfcef7e821e14399 : SUCCESS in 22m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8ec210611a9d4d9e985039cd875a84a8 : SUCCESS in 7m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bd855dd37bdc47a2ba5bf3524cddf6fa : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bf2c34a020bf4efab81149fff9ad64d2 : SUCCESS in 7m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d3ab4236f0041a58bea356a4afa9718 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9e412e6a9ea48f688eeec43163c805a : SUCCESS in 23m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2e26291859234f47b975c4d414d58d4a : SUCCESS in 5m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dc3e72aaee0b44de82be19cb3d0cd596 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/002426f9ebf747a9b5fb9419c24d21eb : SUCCESS in 6m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1e435113b91345ef9ccb9d5b458259dc : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5145b4079f4b4fd6a9478b9d33ef8e4b : SUCCESS in 37m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dbd99c7e50b4fb19bc762fb426007d5 : SUCCESS in 1h 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"caf2d11e683a70a27af7ddb6cc6845ae96d84cca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:56:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01605042b000fa22ba9ce8a44db38e42d07b76ad","revisions":{"01605042b000fa22ba9ce8a44db38e42d07b76ad":{"kind":"REWORK","_number":1,"created":"2024-01-12 16:09:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/905495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/905495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/905495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/905495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/905495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/905495/1"}}},"commit":{"parents":[{"commit":"5181514787ab65c4bbae40e7dac4fc9e73a86da6","subject":"Merge \"update constraint for ovsdbapp to new release 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5181514787ab65c4bbae40e7dac4fc9e73a86da6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:09:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:09:34.000000000","tz":0},"subject":"update constraint for osc-lib to new release 3.0.0","message":"update constraint for osc-lib to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I324e5aeb9800668bf21987bd3c1bb5664356a488\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I6c968c88115295a5fea3c2882769dcb7b155e60e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01605042b000fa22ba9ce8a44db38e42d07b76ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01605042b000fa22ba9ce8a44db38e42d07b76ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
