)]}'
{"id":"openstack%2Frequirements~905496","triplet_id":"openstack%2Frequirements~master~I77dda41100347609257677ebec3800e88d984468","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I77dda41100347609257677ebec3800e88d984468","subject":"update constraint for os_vif to new release 3.4.0","status":"MERGED","created":"2024-01-12 16:19:22.000000000","updated":"2024-01-12 23:56:21.000000000","submitted":"2024-01-12 23:56:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905496-new-release","meta_rev_id":"b265d32aa7d2f0d26ea15daa2bd5e7ef026d857c","_number":905496,"virtual_id_number":905496,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 23:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-12 22:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-12 22:03:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 18:06:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 22:03:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a7d8de928afe3abb93a1e7274a8ee38397c7f787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-12 16:19:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8226e6dea7fd7d4d7a0dd80818bbe66994d0db5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:06:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b9e194a372b4806b478d488bc4d6411\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70741e68773c42d8a2a3970bffdf5958 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/710a65280fa245eaa13371093066681d : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ea43c59449349fb8b6d2d864a18bcd4 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77ab6cbe2cfa4f6aa5511ee7b95fe678 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c76815c892fd41e5848874e3f5e1c0ba : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92ee850455a643439950896b17851e47 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ca7ca5da27043bca0e1f47072e17492 : SUCCESS in 9m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/457bafc0584f421db42e2559f2bae723 : SUCCESS in 14m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/39791baca40745718809535b95249a38 : SUCCESS in 8m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b5498644455a49829116cce63b90233d : SUCCESS in 17m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/684041c82faf4e9cbb057e1dc0d201df : SUCCESS in 5m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cfd6a5c6e80b4522ac92d4cea17fc6a0 : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fb56d049369f41dc86bf220c9ce47b7e : SUCCESS in 4m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2184e93d8f98408f8f65d028255bece9 : SUCCESS in 12m 14s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/4d2995b703134c798a7c86cad266407b : SUCCESS in 5m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3298943474384521967b7a86616ffe04 : SUCCESS in 6m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5224fce8942744729f89fde86624b83b : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/17eb7a3f5b1d4933abfa0593ee566805 : SUCCESS in 8m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8214482e2a474951a377237b8ad6bfba : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc3f849c3c564c4d8b7852933a75e570 : SUCCESS in 7m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f7fc80c8e4dd4548a4a64db7a045dc29 : SUCCESS in 9m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8278e291b4484dfba9b90c9c0392f9a1 : SUCCESS in 12m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b57d46b9f9ca466fba6df734a3533da5 : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/22ad51906bbf49deb2cbc34a8d7feaf3 : SUCCESS in 8m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7374d020fa33432498d73dd4b148fad1 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9d23cea1236d4f95a083e4a339d7972e : SUCCESS in 12m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ee378fe346534c8b97c6072cdc7af7dc : SUCCESS in 25m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6bbcb213f1334b8b965438fa21c4b041 : SUCCESS in 8m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e484592db2534d888a57c7353fe6dfa4 : SUCCESS in 14m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e9feec091b50491b88c1e1953fa2154d : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/613c6cc0037f488f90449afbaad3a38b : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b860e780b2c4846b81ca76b20721cbb : SUCCESS in 22m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b5fa674f2522462b88ab0de80eb3b90e : SUCCESS in 5m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/aff0a358f9ce43f58dbf682fda7020d3 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31180dca02be491796ec12b2b19e5c9a : SUCCESS in 7m 18s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6aa7d433ec2c4d9cbb356b0024ba8052 : SUCCESS in 14m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cc7681a1b5f433fa34bdd2be7016e92 : SUCCESS in 38m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cbbe891f807490b9d50fa4c72ca7540 : SUCCESS in 1h 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e82d3507ffb0cbf254887ba30d4caef0c5542ab5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-12 22:03:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbf77ef5953643854a43e660b699f910bc1bc0cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 22:04:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7596fb32b723d3ec990a903d98bfb1705912372b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:56:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d0f2eaf0f34e0594927842774d5f68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d22d1287c1473e9b1a967474086508 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6246b24763474bbf6a3626fad43ae4 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd82c348d22e4358842724bc5a5a911d : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1c2578da4474745a29d469a4b825aee : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1521382e18e4a92a1966a4efea1424e : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea7d11303f524d2fa8c382244ccca279 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40ed17a2962b444b9ffed178c1790a3f : SUCCESS in 9m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ad58ab7b29bb40a586908d0e4771f192 : SUCCESS in 15m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e95c489400654c1a8065ab7a5b5c87a2 : SUCCESS in 8m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/83d717e7585c4580b2b335785b5b9689 : SUCCESS in 13m 27s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/041026a2125d4f1e98b78f4201b4491c : SUCCESS in 7m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ab89d645afa747a3a23b2d00f640bdca : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3c1d37faffa843aca00239992915a040 : SUCCESS in 5m 33s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8a44f65b79bb4f888a187f6bcfe6e581 : SUCCESS in 8m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/31c0c2a4a6064e1fbefef1ed2a9169e8 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/839c68147a54441ca1e8caf7201ee7a3 : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a967ecd987d54a2db70faa7a49060bdc : SUCCESS in 7m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/84e1bf10a8fb4c228d0468b5f78c4301 : SUCCESS in 11m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3c326cac5f9944679cd9337e1920583c : SUCCESS in 3m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6eb658d8c6db4ab1b20f0fb8563433d8 : SUCCESS in 10m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d35b97a0f70e4722a5f4508c07c3d1d9 : SUCCESS in 27m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/21c06d0ef88f4f8f805f43cceec32008 : SUCCESS in 7m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/95ddab33481c43289235919d1c3e0077 : SUCCESS in 14m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/636e89d002b5492c8fa811b2a568f916 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2407f85b74034c9f9ffb4c5740ed1812 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7602fd9416b4ee180be9d6a9985f383 : SUCCESS in 28m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/34c11ad6b21e4b61b4c0be772cad8845 : SUCCESS in 5m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fb59d66c8c0a42c59a811a24a90777da : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fec6ec515d6244518049804001f052d2 : SUCCESS in 6m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/29b873ae7e8c48e4870157f21857ab65 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4376d736cd549a19278a39965f4ca1c : SUCCESS in 28m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2239a525f494e73bbe23ce1d79b659b : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b265d32aa7d2f0d26ea15daa2bd5e7ef026d857c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 23:56:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3baab4fd8841652a01a5b223a6c1d0b00cf9d842","revisions":{"3baab4fd8841652a01a5b223a6c1d0b00cf9d842":{"kind":"REWORK","_number":1,"created":"2024-01-12 16:19:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/905496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/905496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/905496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/905496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/905496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/905496/1"}}},"commit":{"parents":[{"commit":"5181514787ab65c4bbae40e7dac4fc9e73a86da6","subject":"Merge \"update constraint for ovsdbapp to new release 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5181514787ab65c4bbae40e7dac4fc9e73a86da6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:19:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-12 16:19:21.000000000","tz":0},"subject":"update constraint for os_vif to new release 3.4.0","message":"update constraint for os_vif to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie195d18864e91ab7c948448156144490b8644d9c\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I77dda41100347609257677ebec3800e88d984468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3baab4fd8841652a01a5b223a6c1d0b00cf9d842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3baab4fd8841652a01a5b223a6c1d0b00cf9d842"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
