)]}'
{"id":"openstack%2Frequirements~905604","triplet_id":"openstack%2Frequirements~master~Iaa4e1bd8e1b355b206374f983a290ae80bacdc97","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iaa4e1bd8e1b355b206374f983a290ae80bacdc97","subject":"update constraint for python-observabilityclient to new release 0.1.1","status":"MERGED","created":"2024-01-15 13:52:19.000000000","updated":"2024-01-15 18:22:56.000000000","submitted":"2024-01-15 18:21:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905604-new-release","meta_rev_id":"bbf7111d13da59d3a5bcb643f18379ca9aaf323c","_number":905604,"virtual_id_number":905604,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-15 18:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-15 16:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-15 16:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-15 16:33:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 15:33:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-15 16:04:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-15 16:33:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"583ce8af3148f51acfa5acb32edeb7cb223ed3f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-15 13:52:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"102707ee1ef2aec961dee75f9161d5d33deb55a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 15:33:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74f7b954033b43c3a38cdcea77172227\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ae0f90d0dc4e2db36694ff211b0069 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302fd782db1e48a4a6693f1702da34be : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7128c4f5f0d1429093547d9085e594cb : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cc6bba729984cceadce326548eb2994 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac794b7502b1477b870e916d49436004 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04cab4a699fc4960a5cf20c824b4fa50 : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ed6713d76807405496157f3409e4d8b8 : SUCCESS in 10m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d91643e490124738bdde4bff6bdfd321 : SUCCESS in 16m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9325cd2c4447459ea8db319d27570392 : SUCCESS in 9m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/267f0a4db2bb4a71adcf3a839e611961 : SUCCESS in 15m 36s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e6c9f418b64f4b95aa949f0053c13862 : SUCCESS in 4m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3ccf288a22594e7cb82d3cb9416b9933 : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c764a9c4da804784901763cddcea5bd4 : SUCCESS in 4m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fbf8c16adae54274b3aca35d4da33166 : SUCCESS in 11m 44s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/421c0adb936344b489a9cfca2817838e : SUCCESS in 6m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a5ea3fbc4d7d4e7893b73490c4d53d9b : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a3420ac27fcd44cbb6e4b65b6ac69eec : SUCCESS in 4m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6ab67e0bc9f645b5830ae590fd3fb60b : SUCCESS in 9m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b719c7a619a14355b0821127cdfadb8e : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/551d5154f24c4ef7ac469283aa8631fe : SUCCESS in 7m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b535ad49b2454180b347ca1e3bc5725a : SUCCESS in 9m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/198ff5a2749445beaa43c757dc016697 : SUCCESS in 10m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d29a20db24904ee1814650da711710cd : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/61b36a0ffdf645739646427724089b71 : SUCCESS in 9m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d13599c43c794222bbe2e49bc720f9c4 : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/45451c76c0044515b50c6946bfc54b68 : SUCCESS in 10m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/64e0a3f8be7e47178f07bf55a7765d7b : SUCCESS in 26m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/15291a19fc914860a990b5c503921891 : SUCCESS in 7m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/da94274fba8b421c8b573e34c3b618ef : SUCCESS in 13m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/02016eb06ece46dcb9d905145c1f2d21 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02281bec51d849fdb9d75ec080baf959 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e7e7f8ed69c4fb58b9fb4b6b8403471 : SUCCESS in 20m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/67342965e8e34930af37f553d94c7514 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/52092a6d03634c358f837d52ba84f286 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ea7bfe20bd34aafb76165464404aa0a : SUCCESS in 6m 38s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f100a83827464fbd84f4d032f0780117 : SUCCESS in 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8cecae60b394c98b7fb0a10dfd36366 : SUCCESS in 38m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b690fc7a54d418480fda49672747597 : SUCCESS in 1h 24m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f24b89d1d269ade690c32a3cdc9b51b347d0974f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-15 16:04:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2526c39b47db276bd877487d88f20bb720563a04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-15 16:33:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2956f7a70f2a4100f6268b5258857f6d12b5e509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 16:34:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"160e9f8c90188c4954f608cb4818f4bf8b65a887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 18:21:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68643f72b1443bfa7cd6080dd23afe9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6104d0e03e42048b41b0ebd651d249 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e843d850f9b04cc7be83eb1d659b09c3 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c12208acb4254f649c00e40dd4cc0977 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a9e39c97c154aa8b479fbf2bf39bb51 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ce98dbd20174d01abb0b3ace95b530a : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a9355b7e560422dadf92b29ea111ccc : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a4c4c36d22548c58796baa82cb74e81 : SUCCESS in 9m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4da5a11590af4a65b9779b1aa3b46f58 : SUCCESS in 13m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ef5cb28be6a74b579703ec52a955a972 : SUCCESS in 9m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/20ed6fab86f54b948c2b5a48e390cabb : SUCCESS in 11m 26s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/54117e1064e949f88121343c09ecee5f : SUCCESS in 5m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e4d6456aef5f4be9b9a17773b993a793 : SUCCESS in 4m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f8ff4ec3647045fd804f44389fcccb01 : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cb53d5c6e5d14ffe9982dbbd7d370643 : SUCCESS in 8m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5581602d27f149c5ae7953c58c3c3ee7 : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d60aa772156461193a42df8f30d58b8 : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e0567669309c4a7c971063175a8e7db6 : SUCCESS in 8m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/81a3b45f93234a7d80e9827927a994bb : SUCCESS in 11m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/83749c0ae6114dd3ba840cc6bd665587 : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0900a3f75ca64e6d93f6ee0240368d09 : SUCCESS in 10m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/05ac3c1570b041c9b650595218616b1d : SUCCESS in 24m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/56c8ce2da4484693b9af8dce4a267f90 : SUCCESS in 8m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b5cd595a97ee405684127132be0f0f9c : SUCCESS in 13m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9d478e8bbe154e15b8c8cbdb44b21c56 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27d64530ad744cda97e5e4c3312b13f8 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d81edab5c404483aeb07fb98e02c502 : SUCCESS in 20m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/062949584d0e43b2890c588ae9967ec1 : SUCCESS in 6m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/287abdd9b2c247589f5e6eb3f972bef8 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/538172c757614213a8ca21f05714742f : SUCCESS in 6m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/60e963504d4d477c9fa9761674f89c70 : SUCCESS in 14m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7bfc3e900c84fbe903ffe4e7799198e : SUCCESS in 20m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cac8eb122dc340b79e09f2f75cd9229b : SUCCESS in 1h 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1963596e7084f58f52fb405f0969f835623443e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 18:21:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bbf7111d13da59d3a5bcb643f18379ca9aaf323c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 18:22:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3902a8c33c274f9fb00b6e061a9a5946\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914ebcb598b64958904740947dae9dc6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"250162b07a987e9d98430df9daa7adb75801f34b","revisions":{"250162b07a987e9d98430df9daa7adb75801f34b":{"kind":"REWORK","_number":1,"created":"2024-01-15 13:52:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/905604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/905604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/905604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/905604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/905604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/905604/1"}}},"commit":{"parents":[{"commit":"25b22199e5ab7aae337bc61f2de950c2b77cefe2","subject":"Merge \"update constraint for python-keystoneclient to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25b22199e5ab7aae337bc61f2de950c2b77cefe2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-15 13:52:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-15 13:52:17.000000000","tz":0},"subject":"update constraint for python-observabilityclient to new release 0.1.1","message":"update constraint for python-observabilityclient to new release 0.1.1\n\nmeta: version: 0.1.1\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Martin Mágr \u003cmmagr@redhat.com\u003e\nmeta: release:Commit: Martin Mágr \u003cmmagr@redhat.com\u003e\nmeta: release:Change-Id: I2783a090adeef77bda3685907aec44bc26791b62\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Erno Kuvaja \u003cjokke@usr.fi\u003e\nChange-Id: Iaa4e1bd8e1b355b206374f983a290ae80bacdc97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/250162b07a987e9d98430df9daa7adb75801f34b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/250162b07a987e9d98430df9daa7adb75801f34b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
