)]}'
{"id":"openstack%2Frequirements~905606","triplet_id":"openstack%2Frequirements~master~I6f8cfd01ac9d224e3b83afbfc69b45e1493e98ee","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6f8cfd01ac9d224e3b83afbfc69b45e1493e98ee","subject":"update constraint for yaql to new release 2.0.1","status":"MERGED","created":"2024-01-15 14:11:45.000000000","updated":"2024-01-17 10:49:52.000000000","submitted":"2024-01-17 10:48:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905606-new-release","meta_rev_id":"1c16cfef75860892646a2953e0d7258dc887852d","_number":905606,"virtual_id_number":905606,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 10:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-01-16 08:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-01-16 09:12:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-01-17 09:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-01-16 09:12:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 15:59:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-15 16:08:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-01-16 08:05:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-01-16 09:12:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-17 06:33:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-01-17 09:05:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"a7d10bdd29de2d330fbc40eeeab4b71402f60e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-01-15 14:11:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93493b3cd6c5f0816980bab60d8eac9c2f4a037d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 15:59:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d650ef95a61444548e2338df7167a639\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8749d548784677aea3158fd0e56833 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41083a6701d64c43a2b72c01a412d936 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac23fd46ce94f4581e3a33bf567deda : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a26d20bc273422e8a407389d38b3291 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/defabcc4902c445d8458831dc758cf22 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0e58257a4674f0bb2c70da11b4684d0 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3f1fe044286f4d088ece14a0f3242a28 : SUCCESS in 9m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b2d2a9efac814f7aa199ca022c4b39bc : SUCCESS in 13m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ac790f61d2154450882daab938f12329 : SUCCESS in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/126e5f2afddc4d9eaf478e671006b2ca : SUCCESS in 16m 57s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0585f2ed936a410990aed4a2d7531468 : SUCCESS in 4m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7ed0bacde8d34634951149bc8741bc88 : SUCCESS in 5m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/517dc4a9103740eb903ff267065e7306 : POST_FAILURE in 34m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e4eeaa68cd4a4a40916c4a4167ea99ff : SUCCESS in 11m 18s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/f984e5b8f39845389f4605c0407a8e72 : SUCCESS in 4m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/70a9cdb0491049f58c45d4bfd069b7e4 : SUCCESS in 4m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bf26b7b4d211415090a042ed2797e006 : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b35940afe50d4dd49ffdd731c2d433c8 : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b62e74aa942b40d78cf1c1fc9fe160d6 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcaf1c8f172c4b4c96bbeffd09680800 : SUCCESS in 8m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3a363a1141de4308bd6cf350cabb23d2 : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/55f8fa30127d4e939883631d0280cd2b : SUCCESS in 11m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d19096d7d6da4f93a4c93cd230c3c9b1 : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4ba3a025902e47cd899fff6cb5c14b75 : SUCCESS in 9m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/42f7a975b28b4d6ca4c0d625d9fee6a6 : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e7c10909875b42f8a02c6566cea45072 : SUCCESS in 11m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a933c4e6bc634875bac952081d254934 : SUCCESS in 28m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a78c36df89ba410da9247b45f1e566e8 : SUCCESS in 9m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/225b19c4094e4b1eaef467c6e333ed9b : SUCCESS in 13m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/531bdcb5abdf46438d63b48e4d7b419a : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48eb4a63128a4d248d3a519d448fe2e7 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3faa993db48b4cdf836e68580e7e373b : SUCCESS in 24m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d3449c5cc109481baa4fae60876bf74b : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3e0c5707650641a3a93d04b26a693526 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74cb30115eea4eacb5b8f5c27a2e2c3a : SUCCESS in 7m 00s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8ab6c4076df64ff7a66d0bf14363860c : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f34d274aea6f42408f49134cb91ce925 : SUCCESS in 36m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/047e8804eaa44979852a8e4090e1d8e9 : SUCCESS in 1h 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"439e873243ba650f50e8f92f21811f594081adf1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-15 16:08:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc69eac6238a56b7ec42d6287e264f24cc06cd62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 17:47:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5d399322db64ced87cfd4c7471099f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9954a29942ca47df934276f7b8ff94ff : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2834e46f69ef40b7bc3581e11dfbb618 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/991dc377f1684fd0a26946c31b8a0a3b : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1e0dac7d87e43929b9b99eb3f744ebb : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8697c115626e43eaa0914c9e4ab59828 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dcb328578d4d4fe1ac57ea59213ffe9c : SUCCESS in 4m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78160e3bffd7412d985de10dfd821853 : SUCCESS in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce7aa740d9154800b328b037bab37cf1 : SUCCESS in 12m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d1ccab6e0c4e4fc0a6305065e92ec6a9 : SUCCESS in 13m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/10ea2878bac74ad49ed8634b74f4ff6c : SUCCESS in 15m 06s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/066fa6686c524d4e87e43495c967daff : SUCCESS in 6m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fe766d6215d24af1a0cb92f5303db559 : SUCCESS in 5m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8bb38e825d9c46168f569cf8efa129e3 : SUCCESS in 4m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a118835c297f4997ace099c6a4191e5f : SUCCESS in 11m 56s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/d7d300a5ca434c3197523ea0d6d17546 : SUCCESS in 5m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bc8bc52e92a94bb287c4d3221f095f7d : SUCCESS in 5m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7fad729ef3334ac69dbdeeb1bfee4ad9 : SUCCESS in 6m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e4977d90c49f49b58340f4d28c927558 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/afce42b23d4343568c66ac54a5f72854 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c83b62e17594a348ae00a8ea8f0bf2c : SUCCESS in 8m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fa7d093de93a47dcba47461ba194cbf6 : SUCCESS in 7m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c4fc5c47637744339ac2a18129cb2c17 : SUCCESS in 12m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ff80cc7ea114400183aeb5780b26c7a0 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/298400a94bb047538d53a0e0ee41f121 : SUCCESS in 10m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c2567ea892a94d8eaa48ed5a27e34415 : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/aeaae5221ad64e059df9fdcc647f11a0 : SUCCESS in 11m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/037d1dfe33164e45898c30a8bc5a1ea1 : SUCCESS in 26m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bc105749560641dbacf58a666dd2c4a2 : SUCCESS in 8m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/be1a05c2e70846df9f2fe73104b0e184 : SUCCESS in 13m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b21fc31a9e274edfa4d7b4108bae165d : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04d4ded8d76445b5a2a9409c497c09b7 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/beef5b39a40b4bbdbdd1bb10078a518c : SUCCESS in 23m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/066536fa6ce24fb28dc9ec33560ace0f : SUCCESS in 5m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/67b45dd9f5db4fbaba1df3da098e8d83 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00fa3367e5694f51bc86bf427030f859 : SUCCESS in 6m 41s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f1a9ffb0bf5a499f921b02d3288f2f25 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa08e57107ca43be9441e786b50a3890 : SUCCESS in 35m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e95901b5f6c84bc5a2f6ae184ba41735 : SUCCESS in 1h 32m 36s","accounts_in_message":[],"_revision_number":1},{"id":"10658e91a46a96644a392b4be78c93597b0e888a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-01-16 08:05:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64b8bddfdec26b40a31326fa34cfd7232717822f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-16 09:12:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edc4c3e3caf74b10c3c1cac673d6357ef0cde866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 09:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"719406bc233e091107cf02d5ce0296c75cb07b45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 10:54:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8852fab8c417478fafe675647884f444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed7bbafb62045af86e21eff86ce9703 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abbac05500644fe9a9305dd9cd52e235 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ffbb8935c234a7492430032bc1550b8 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c46f28de6e242ac86e52de752f88c9b : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26d9226f516d424f98bd843f1020e3c8 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b15197709e6a47d7a501219312b05d5a : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7392d9ad95d44d9d8cfc98b1cc7e9e03 : SUCCESS in 9m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/73e0bb7292ac406d870650b9cd690676 : SUCCESS in 17m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7ab6f524b6ce479b83643cbb50c7b593 : SUCCESS in 9m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/98040a24326547ffa36fffa4b1856aad : SUCCESS in 10m 32s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/273741e985e146f7a0a0525d624abcfe : SUCCESS in 6m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/98dd70ecc5c6467c899a62815f8b89ae : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c810e9a42d734935b705f2d3e3abdceb : SUCCESS in 5m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/85fd8c27296a439daed432d9ee81db38 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/69c7e608854545e4a5a591dda3b94f75 : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27547f18b9714d2d93cbf5d64299fa77 : SUCCESS in 6m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e07e4ef97d46490cae982f1e81bb4694 : SUCCESS in 8m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c135d14068954689a03e14fd7bb4365d : SUCCESS in 11m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/14d57346eb25476aa79fa6e8591677d7 : SUCCESS in 4m 55s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/38cc384805d844a2b415bcb4cdfce3b5 : SUCCESS in 9m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5babf389316f465ab305252b2a21e840 : SUCCESS in 28m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/557a7b5323b141e9a9b3429c4e1b1c93 : SUCCESS in 7m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a7c04b40dfa74e91ad06fe6aa90edb7e : SUCCESS in 11m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4f966c2624584920bfac68c4e8e35a8c : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c262091ee8844275867e2d5217660671 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e90181c8e8504be492841c0155935842 : SUCCESS in 24m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0b23e0e1b30845d8b4b7546d46c15854 : SUCCESS in 5m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/010f6c16465348af89112098afb19209 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3474d5bf55a34a6fae6ce641692cec81 : SUCCESS in 6m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e96aaa689c284ad4a3720c71c3f351a4 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49b56a7f5f804df1b445792f09b1f4e8 : SUCCESS in 29m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88c82288220d475c8c5fa69a8def0947 : FAILURE in 1h 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ad86aa9ffa3e44ddd7e052a4c386445adb0a31db","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-16 11:00:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ef14ab387d8e75d32801a12aa3776ce21b5c0fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 12:25:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e061573ce1ee4700af613fb271e0e88f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9eab84221514c1399e60224fba34c2f : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d0f0f60c69c40d597fb1a0edd482615 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b56450fdcd834d67b58f22594c582141 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cff6652a4f7e42318fa8d50f0b3a0e39 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fce6d431757f4509b430fd0efe8e42a0 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b5b33ad80664050a134507e10a44b8a : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a9752bd047eb423a94f47def278b6eb1 : SUCCESS in 9m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8fcf7ae5458b47bfaae92864d125aa98 : SUCCESS in 14m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d8813824a2854909b1253b120a53079c : SUCCESS in 10m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a9150d3106c40969a4e6055ff878d09 : SUCCESS in 15m 43s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/646cc87801004ee58c060625bac4ad76 : FAILURE in 5m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/787a47e273744331983fecb996b7e75f : SUCCESS in 6m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8008dbbc79a34cd2b882aaa94917bbea : SUCCESS in 4m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0ecfc4d7c86147de9b0a347e45f3ce53 : SUCCESS in 11m 42s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/b42008f36f35426fa01a43ef7fde33a1 : SUCCESS in 4m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/333d82bc6bf641c9bf8396682562e547 : SUCCESS in 6m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/69a7322d122a4e42817f4f5238328f53 : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0c2fda54f1b342d89bbeb434b15aabad : SUCCESS in 9m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6b72400bc5ac446b8a4d6a4c9b436345 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/729d822710694d1c893c4080e125fde0 : SUCCESS in 7m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4f4d62857c3140a59c3849b416a8bd8e : SUCCESS in 8m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bb6e3735161e41f2a93e9e3c44fb56ce : SUCCESS in 10m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/01969ffb06b24ccc8b9f35aaa175c15f : SUCCESS in 3m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/795eebb186cf42e59597bb77b9916af7 : SUCCESS in 10m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ddce82f89ae04abbaade3f855d13d297 : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c6af4c172a53455eb1f3cf2162d2c11a : SUCCESS in 12m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/390cabedb8fd417ca1aaef7808fc2beb : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e840b2c41cd84e22be7fee04cc447321 : SUCCESS in 8m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0cadd5d370d34f85a06b80b6b2018cb6 : SUCCESS in 12m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d0104241a5084cb588b08cec09ace269 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f63a87f3ac2c44e6ba6fe1a338bbc952 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e13593f3bb7f48d8b514143689a01ffe : SUCCESS in 21m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/505a2d8c0b1046eebfdc77507f9c6a6e : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cba8121fcdc941edbade63dd7033ba50 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc019cd1f5ec49f5a72872d98c06573a : SUCCESS in 7m 09s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c74d0c558f664838953e6bede5980b2b : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93935bb67eb54f15a1b650f0631ea322 : SUCCESS in 29m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6679e0aafcd49fc80d2dceb7effb594 : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0c41302b24f2fb93e11412b47313845d32d65f58","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 06:33:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14ba14e7b4d59cc1a775080f5c5ddd6bc6e1c41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 08:52:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1d09e4cb1064ef494e24adcdd8ef4a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/227ed9b4a1f84951afab4e9c38c286d5 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46df81ce51924264a8832fe9b3a8b365 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50965091770d48698e0a40c5749dbd9e : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32adc771d1ca4db3beab4c7f3986a500 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b8d444711204f0594d1a6402c703766 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f0f9bd94bec4480aae5863de5776562 : SUCCESS in 6m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f21aac0b741d4ef1af7c59f6bce62461 : SUCCESS in 9m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce5c77732b784e6c91930ad9aae1ea47 : SUCCESS in 16m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4c5e877d4cad4344896497047c293100 : SUCCESS in 9m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9fac83b9d1074ce2a91fa051dccb689b : SUCCESS in 20m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5c13817420ed4ad48dc26fa03331e3b6 : SUCCESS in 4m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1cbc7f76647f4fecac799875ae3b7034 : SUCCESS in 5m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1c9f1803e52842acbe8ce6cb54533d6f : SUCCESS in 4m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/60aec16e06664867aeea0c2c4c87a2d4 : SUCCESS in 12m 24s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/993ddade843d4c858c5d279a3acf76ae : SUCCESS in 5m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/71232bd04c324ce291869c61f324b1a3 : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9f3dec144cda4cae80bfdf4029b06e5c : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fce9ef158dbe45be8d63edb64bd8d98d : SUCCESS in 8m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a95ef0d86ae243448273c5f9a7242662 : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9c9dcd4951424006912149ef34998897 : SUCCESS in 7m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a45005af85e047e3b0d8edfb5f3a888f : SUCCESS in 7m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1544b613c0984f6d9af21fa62f0a6027 : SUCCESS in 12m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/964c5f2436c04bc0b98b258f3dafb49a : SUCCESS in 4m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c7a683df3817475ca8cd6c1d5c1c403f : SUCCESS in 9m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ab80cbf28100460882f010c3f4a0ac9b : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/194d21ced76148348028b0a3d841b3f2 : SUCCESS in 12m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/80b02608bb6f49a9b746f54fb48ef365 : SUCCESS in 21m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/442a023490124d77b521537c453d7181 : SUCCESS in 7m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/82a11c0d5e1f448c9154a4fefdca2746 : SUCCESS in 13m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c3c759dcdcf64651b685eab5f421a49e : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d5a13021cda143cc8bb3aa49af98e07e : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0fef9ef9e4aa40a6a5b26560194c0cdb : SUCCESS in 24m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e49d34e770af4e809e6ddf5a99229bf6 : SUCCESS in 5m 51s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5a39774815a94bdabcdc3a7d03352730 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2bb4dc603a0458b84fef776e2edac84 : SUCCESS in 7m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ccf2c2809d0a492eb2853520575a649a : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f3967507ee0480a8419b9eb7ee28ebb : SUCCESS in 29m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/566085b8ac8b4ae1916fdeec4090eedc : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9adfdc7ee54a5aac49814486c35d1776b0a1ca56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 08:53:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc988679b3cb168da98867c66af0584f4197277c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 09:05:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aff2570cdb59db363ae16a27f2d507374312d70a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 10:48:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c60b162ed6543fca8e796acb055437a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8125474832684483999f6bf40471a26f : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49017dcf635d4a5a9cd3a0a5beaed5d2 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a15b75285e7d448096cea64725fd4638 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52f7001fcd1a4a37aceca91ed7804be2 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d02a3d815e124cea94c80db9e832dd28 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9f7d79f9b6d44c991de2083923e7fb4 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2cfbc97680904173ab0e8f9e454f4f30 : SUCCESS in 9m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/72b14f86efe84cffaffcd4d52c3e5ca2 : SUCCESS in 14m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c1ac16e10a2746a8811418d90f5a3b42 : SUCCESS in 9m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aba7178b585d49d498e313bbda7f7bee : SUCCESS in 12m 50s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/fb7f0f67290b4411aaf55d0251be6590 : SUCCESS in 5m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/285c3999250d4a76a9e5d20e641437b1 : SUCCESS in 5m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/387ca80e45754276a7db940549ee8e52 : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9a87a3c63a6e4d8e84d930ddf0acab03 : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7913d5bcd4fb4e0386d02a9e1a076456 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27a5c0e9d9484df495d9fbd2a19eb43c : SUCCESS in 7m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3eb3dd18374944fc95b0e8eac0889fc9 : SUCCESS in 10m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cd619d9d4cb148c787cd465b1e68bac1 : SUCCESS in 9m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e75513a951314d4fa9eeb6bada0c3fae : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a3069aea7e3d488f98ea2ad38f048857 : SUCCESS in 10m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c4b6bbe485e54ffc8226024b1f93ac7c : SUCCESS in 27m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6edf10dc76144ec9882e3ce4d08ccc76 : SUCCESS in 8m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b2b0f2e0b8bc49589245c7f0b1e0becc : SUCCESS in 13m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f5a0a9883ace442bbb000e478149e1f4 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79e981ebcbcb49f281e599104052d484 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af64b27d55d347c99e5437b9584065bf : SUCCESS in 25m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fd75ce59adba4330aacd575bc763fd7e : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/796a4c2965374e559f7a50aa8cedd708 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2efae8344504d80b15c20ea0b7db271 : SUCCESS in 5m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/470dbab10ba94e3395233e7c4d611b80 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5940028168741ef95203b25a6693be5 : SUCCESS in 36m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/250ad825876940a28f01b4d0bdcd4496 : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3b003313a0b4b95d0462a9dee38ba6fbccf2772d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 10:48:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c16cfef75860892646a2953e0d7258dc887852d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 10:49:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239140482371419489346a5a8c3c2d7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62eee226d0914f3aa4a5034d62720ad9 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04d623e0a20464c19167fbcb4fae034d7a57f512","revisions":{"04d623e0a20464c19167fbcb4fae034d7a57f512":{"kind":"REWORK","_number":1,"created":"2024-01-15 14:11:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/905606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/905606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/905606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/905606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/905606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/905606/1"}}},"commit":{"parents":[{"commit":"25b22199e5ab7aae337bc61f2de950c2b77cefe2","subject":"Merge \"update constraint for python-keystoneclient to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25b22199e5ab7aae337bc61f2de950c2b77cefe2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-15 14:11:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-01-15 14:11:44.000000000","tz":0},"subject":"update constraint for yaql to new release 2.0.1","message":"update constraint for yaql to new release 2.0.1\n\nmeta: version: 2.0.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I60d42268b3d5517300dc98f675d8bb87cfea0587\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6f8cfd01ac9d224e3b83afbfc69b45e1493e98ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04d623e0a20464c19167fbcb4fae034d7a57f512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04d623e0a20464c19167fbcb4fae034d7a57f512"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
