)]}'
{"id":"openstack%2Frequirements~906022","triplet_id":"openstack%2Frequirements~master~Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133","project":"openstack/requirements","branch":"master","topic":"constraints-tools-updates","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-06-24 07:28:12.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_12898\u003e by clicking the attention icon","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-31 00:30:36.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133","subject":"Many much refactoring needed","status":"NEW","created":"2024-01-18 05:16:07.000000000","updated":"2025-10-21 05:24:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"98ea2068189d3f548ecd5f1a3869bbf71c5523c4","_number":906022,"virtual_id_number":906022,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-21 05:24:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-01-18 05:16:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 10:56:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"1061b4343f1ec0c21e1901bdae466b54099a3b6c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 05:16:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1f5d4a12f4351072da3800b4420b4eca6a0de43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:16:29.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027name\u0027]\n\nThe problem appears in the the \"requirements-generate-constraints-multinode\" job stanza:\n\n  job:\n      name: requirements-generate-constraints-multinode\n      parent: base\n      # pre-run: playbooks/constraints/pre.yaml\n      run: playbooks/constraints/run.yaml\n      # post-run: playbooks/constraints/post.yaml\n      branches: master\n      nodeset:\n        name: constraints-multinode\n  ...\n\n  in \"openstack/requirements/.zuul.d/jobs.yaml@master\", line 84","accounts_in_message":[],"_revision_number":1},{"id":"a4c307ae590fb2a936b6dfecce146cb62ffd0922","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:16:37.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027name\u0027]\n\nThe problem appears in the the \"requirements-generate-constraints-multinode\" job stanza:\n\n  job:\n      name: requirements-generate-constraints-multinode\n      parent: base\n      # pre-run: playbooks/constraints/pre.yaml\n      run: playbooks/constraints/run.yaml\n      # post-run: playbooks/constraints/post.yaml\n      branches: master\n      nodeset:\n        name: constraints-multinode\n  ...\n\n  in \"openstack/requirements/.zuul.d/jobs.yaml@master\", line 84","accounts_in_message":[],"_revision_number":1},{"id":"12d8a7f1307dc7317a8a37a95b277456294b1425","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 05:23:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"864c0567045120e73e4cf5b95fd25d1c7b51adbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:24:14.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027name\u0027]\n\nThe problem appears in the the \"requirements-generate-constraints-multinode\" job stanza:\n\n  job:\n      name: requirements-generate-constraints-multinode\n      description: |\n        Generate constraints on multiple nodes and merge into a single unified constraints\n        file\n      parent: base\n      run: playbooks/constraints/run.yaml\n      branches: master\n      nodeset:\n  ...\n\n  in \"openstack/requirements/.zuul.d/jobs.yaml@master\", line 84","accounts_in_message":[],"_revision_number":2},{"id":"eae01fe6eaf3b3cd505b37dbaf88b5215eef58ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:24:24.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027name\u0027]\n\nThe problem appears in the the \"requirements-generate-constraints-multinode\" job stanza:\n\n  job:\n      name: requirements-generate-constraints-multinode\n      description: |\n        Generate constraints on multiple nodes and merge into a single unified constraints\n        file\n      parent: base\n      run: playbooks/constraints/run.yaml\n      branches: master\n      nodeset:\n  ...\n\n  in \"openstack/requirements/.zuul.d/jobs.yaml@master\", line 84","accounts_in_message":[],"_revision_number":2},{"id":"b44c29f8d7d6de77eb6467bd8fb5eeb770885e45","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 05:29:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0a1133661cd8da87b2cb94972149c24fd529fb14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:48:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5466fb592d3b4b1790dda42840f3b259\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1126ea1bcd54e3dbaf38732cedcf88b : SUCCESS in 3m 25s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/778e757828764a948ea14224f560230f : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7343fdfdb19ed6656fc1146437888a1e4706daaa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 08:52:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5a9e5f19a51d91c1d43803028465a94da901bb6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 09:16:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c0911706f2b44dc89daa38c1c85f723\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6373f5d74eeb426ea48457ab8be8aa67 : SUCCESS in 4m 50s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/ff6d2bb8648c4b49ad45abc16a273f28 : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":4},{"id":"0894b596e5e013e0a172d92f925510ddd7214f8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 09:51:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2ab87e885d5afeb668d44cfec62177b0556da508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 09:58:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03f7880e1cb544f5a0a671600c4444f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b782ecd85aa4fe68d25121c70763b81 : SUCCESS in 3m 11s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/880810fbf075483782f2b9d86655a0e7 : FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":5},{"id":"818f919597640064eec84da4146cd74676372b86","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-18 10:00:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"93568ae34d24bba858831ef41487c89220172370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:08:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3476a560d6b4d869e7c121cdbd01904\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d2a572ee5a452aa3b87872e7fe9e6a : SUCCESS in 4m 42s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/5adc2c8fc6214f3abba2b19839da8cd4 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":6},{"id":"227883b331cd3fe637bca07106889f457d393556","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-23 20:36:32.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"96e83ef30a02026ce685c571a67e83502f929337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 21:02:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc48aabfb94f41efb13229194633ebf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5470caf754644fe28880570bfc8e0133 : SUCCESS in 3m 36s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/71bb70c108174c0eaff857e10bf11fe3 : FAILURE in 1m 50s","accounts_in_message":[],"_revision_number":7},{"id":"f7a82e0373e89f06a460ae744798fd6ece3252f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-23 22:56:32.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4f0c62b83bbf1d9c78795521ec8c40d61ebd7341","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-23 22:56:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"69c1c6a7fe354389eea0292c5b0d81fa9fd075cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 23:06:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35b7dac6789f4a06a5efaf158993ae1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc4d440dfaa40dc894e6e4ebff5e5e5 : SUCCESS in 3m 23s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/7c0e1e80ee3446038d13cc93703502e9 : FAILURE in 2m 08s","accounts_in_message":[],"_revision_number":9},{"id":"d0f373f2d993091cd55b27b8170812d792cbf820","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-23 23:11:41.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"9ec37a8eeb065d7248aaf705d62b2c5a665c648e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-23 23:12:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"feb3e482f007087c2d89cf2c56ca97a3f59209cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 23:18:57.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9c6ebf127da48c79699abbf511689e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232b06620bc74512916d8054aed3b0f8 : SUCCESS in 4m 46s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/6623166a691f4226ac3776d1bd2e0b91 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":11},{"id":"637a8a1cdbe3174283190141a191b14edbb506f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 02:06:46.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"f7f4d5eb2e16818a374d9c220f583f5604d1ea7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 03:09:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f131fd49df5d4316a9bbc0e1fa3d63e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861bccb06c194d249a504fab7f3e392e : SUCCESS in 3m 53s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/180eeea70df7479aaafa957bfd12ec01 : FAILURE in 2m 21s","accounts_in_message":[],"_revision_number":12},{"id":"cd752d015a580962e3fb5cee1e94915002f4fb1c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 03:14:38.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"a7263bf19ad7b655725d684caea675b43e3d481f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 03:44:04.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04336695e4b5423d977ba2ae4c5b7a22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260569e85e5e40088634d3fcc9a7ba81 : SUCCESS in 4m 09s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/25431b5b34ae4aa1a1d872ed13e24a94 : FAILURE in 2m 15s","accounts_in_message":[],"_revision_number":13},{"id":"372d546c1a4a2a2a0e7c8d5f51429a3c653b3d1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 05:11:57.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"d5be6c2637797ad810ac01d33e06b3338405e242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 05:39:05.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c0bc0a4c704a0c9194e353266c6a8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfca24c51b38482e82a0e0f17dfc0120 : SUCCESS in 3m 28s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/c95b40c800e14c5983bfbd4e833fb056 : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":14},{"id":"fb8aaffd9be4b197f034d3fde620131bf5030fbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 06:49:47.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"3d2d87c73798d087186da625676ca0525baa866d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 07:12:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c44e6a666a6b46f0aa067a0f21ae64f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9c8bb9e8ae485f9653dccced7ea260 : SUCCESS in 6m 31s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/659846b0f1a8402b9ea19bc097682f1d : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":15},{"id":"bc66bdaaaf5ddd735c7dcc40c9cbd6c19e8a3953","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 09:58:46.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"da12a220f5fa835cdedd9919aa0f3dbea75d9700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 10:19:31.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d347d234cf3a4f77b69aeb53ac07ef16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f97f5f248f43238fe71eaf6a43dcdd : SUCCESS in 3m 48s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/bc53e4b9ba0a49b286f2cd2244936413 : POST_FAILURE in 18m 55s","accounts_in_message":[],"_revision_number":16},{"id":"8be451a78f517f7e7c9ad6ff9ee9d3ff1ac8042e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-24 10:56:03.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"71d1ebe63f5713907b694b5af696020e474b726f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 23:26:20.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"934bd3d9ed712e08403d03d88202ab74d76293a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 00:16:49.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b007c8c2d04eb9a4107b87fa22e064\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cdec35e79514f11ae66f153fca54956 : SUCCESS in 3m 10s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/0e791068beb049739c422e898e7399b6 : TIMED_OUT in 30m 38s","accounts_in_message":[],"_revision_number":17},{"id":"6b88834dccef96b4e884ad9354eec6870f824c58","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-25 00:37:33.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"342ca958c20ce655f7991bf6762f376aff66b90a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:13:57.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fef9cd33d8b54726a280c29efab9e29b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb6abe4c37e495490b33f6511adcb63 : SUCCESS in 4m 18s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/32f5833fc37f48718cd6072182dd40e4 : RETRY_LIMIT in 1m 33s","accounts_in_message":[],"_revision_number":18},{"id":"f503e793974809796857cf8d3d0ecc58592fbd6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-25 01:16:37.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"d6fcc414cd77e8f4622013f2f541c51a3088c7ae","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-25 01:23:57.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"52df2ddeacce1b5544f78053ee337dcfdb377e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:48:14.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5665ac7f5254b81925d9bdec8a2fff3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb898972b62e4cd9bf9c0b0b3cb0ae83 : SUCCESS in 3m 16s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/7f94b108a4f146859d72722cc713cc43 : POST_FAILURE in 29m 04s","accounts_in_message":[],"_revision_number":19},{"id":"f0338e87a4eefbb651a04520dd8b3c3915d4ce6e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-31 00:30:36.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"88eaf48882f6cdb2fab504dbf92b3b720079916d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 01:00:39.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c2292343e2145708f05af9c94a1a51c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15d77500aac4be88d8081fff95559e6 : SUCCESS in 3m 21s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/affcf6f338c44c50b1866c12f9b9cc23 : POST_FAILURE in 24m 35s","accounts_in_message":[],"_revision_number":20},{"id":"8fd69ba45a6373d74e1e059768ac02ac88c88c88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-31 01:40:43.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"f119576a0c43cff5d213cf5210f4c4666d152ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 02:08:24.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4266f326234808a559f05f0ebc5add\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6af3ed92084992857c0b4aa741be36 : SUCCESS in 4m 28s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/e666bc447a564f5fb5800d822b52782f : SUCCESS in 25m 21s","accounts_in_message":[],"_revision_number":21},{"id":"5f090188120363562a8864c395e5c60b005ca2d8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-02 01:19:19.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"f0e52baf5e77d2e35c10afc072953795a8c8200c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 01:54:37.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c67dd15e4b254dbeba2d1a5003235a78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa47867e3d64acca18e13cc5c13da96 : SUCCESS in 3m 33s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/b811e07c27504c91b03c3383cd345d8c : SUCCESS in 24m 17s","accounts_in_message":[],"_revision_number":22},{"id":"1b4273e72c2f4d7d859213d138853b84ac4f644e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-11 23:43:12.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"e1662274ff7c01356942e57983619c92028f9e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 00:00:41.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d30a38c1b474bcfb6eb0c4e55ffd3cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c578a7203a784adfa750edaa7572342e : SUCCESS in 2m 53s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/3b979d183c2c4e0e96c041f11939f7e6 : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":23},{"id":"62fc3441f4701285147cb7448ec8f4b750c2d196","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 00:33:14.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"64f628c3f57310e897914dcf7fea924162d58dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 00:53:33.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0aabda97cb40c685557867c5c742a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ef16a6eba54b2ab7a837d3d7a8eb21 : SUCCESS in 4m 15s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/ae9c8a510f2d445ca5774e0b5694c3bd : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":24},{"id":"d4c7eb48afe3fa88d6e56fdca796a0babeb79ed7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 01:30:32.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"bec9924a4ae547f6cd1b0f63ad2e638699d9b8e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 01:36:16.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d28a1f041bb496c9e5aaf64b0cbecc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec90aca1dac74c2d83ff10e1f3e4a47c : SUCCESS in 2m 51s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/95fc9a6a656746c480ef8205b42d84dc : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":25},{"id":"0444de23ae0c3680aa687dea8cf7238e060643db","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 01:37:41.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"33d55cc70fb1b5236de0651db5aa012c53c98024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 01:54:00.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94fe9e2e8f7a4f5389cd758e02aa19de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94d63340289d4983b76886a631f41795 : SUCCESS in 2m 40s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/45ae662816d24e7c947298963bb7f4e1 : SUCCESS in 15m 13s","accounts_in_message":[],"_revision_number":26},{"id":"593fafb1ac1cb266e6bcec62d0bdcbdc0e984412","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 01:57:43.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"fd34fc7ee89eff6ee307122c32aba0501228fcd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 02:18:15.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/275af4ffd6f24859aa173c80f2ef40e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd665853cadd4fa7ab44c11f9d557bd3 : SUCCESS in 3m 10s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/b37be4a96fa34cd193003ae693e13ec8 : SUCCESS in 19m 26s","accounts_in_message":[],"_revision_number":27},{"id":"899cdb6315293fa199ff9b9b574397b17ea1aff3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 06:55:24.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"9f271754ff8772b7a6d9c2bf3814139d63960b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 07:14:05.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/073a072a5abf45479067976955dac665\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bd3deeb5094717b96c9989f594a963 : SUCCESS in 4m 16s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/17ed546c5274403bbb6ecc04b283865a : SUCCESS in 17m 25s","accounts_in_message":[],"_revision_number":28},{"id":"4c07c449137118e8333d2b1a02956ea78f7fbeef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-13 23:51:19.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"8e6a0c76c7a617f204a57db8452e6366f1b9f3e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 00:20:31.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b3f96329454b61bed03164cd0ca7c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0003c8e4e34c639ee08fc972871859 : SUCCESS in 3m 55s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/d40f1266e0fb44c1a9cdaf7d00e3ec9a : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":29},{"id":"715b3473261daeba58e57abce323662fccb8fea3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-14 00:30:34.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"7ae2797b8e44d25d5f0f2920b90e434f30d9e451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 01:01:08.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76eef81c60c64d1cab66568ff409de22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5d953487454ef7be3d744fa99e26ce : SUCCESS in 3m 21s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/be6e5c9f8dd04ced96c9988511dc23b3 : SUCCESS in 23m 42s","accounts_in_message":[],"_revision_number":30},{"id":"0f5ec62baf80fae7b6d5858a7fc92189d78b4e7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-15 00:06:39.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"849dd805c03fe84caeceb0735e8399f2260b01af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 00:36:08.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6cac28a1b2480f987491ecedae2aff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b043156deba4557b41b298bd5ee8161 : SUCCESS in 3m 19s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/3aba75da90aa47269207789a0aa2bac3 : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":31},{"id":"2881913c0e3a213e70c19e2d496a028bca4e35e7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-20 21:20:36.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"82a85566eaa803e2d2a2caafe3a8559cf5bcbfd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 21:41:59.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bed3fc383f34f09b29a8833f50b1d16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbda92919fe04422aee15563319c4f3d : SUCCESS in 3m 08s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/d6b60af1a35b412fab81553360aba2f2 : SUCCESS in 19m 42s","accounts_in_message":[],"_revision_number":32},{"id":"5869a93bafef3bfe1622ce8a8456a8b7a1ab44a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-21 05:53:45.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"67dec14f332874b4f064643841076d44b971dc49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 06:12:21.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c524112c0444a4b1c4cfabcf20b630\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c769915b904b0fbc26f651d53d0b55 : SUCCESS in 3m 02s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/4af990e88a3f421a87eb053d3c099d35 : SUCCESS in 17m 13s","accounts_in_message":[],"_revision_number":33},{"id":"8ec77eaf233e026222be4d7381e57d4c475ecc6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-22 23:46:49.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"5d8f633f3c243d7661c5414615d2bcceff80a9b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 00:22:48.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3aef0fd05b48a2b6c0c712042b42e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5125a9a32c974fda872917fc799d840e : SUCCESS in 3m 10s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/37cd27896b804b4fb213e23b4e1d512d : SUCCESS in 25m 19s","accounts_in_message":[],"_revision_number":34},{"id":"859dbac80042973083095cf61396e0739a2d14ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-24 02:25:55.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"049f275df9395edd92ed3ddb904d02d74cb42c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 03:09:29.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03c9f05cec7b4e8bbedffb952a71c767\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be45b44c4d649e089ac24f21f015cbe : SUCCESS in 5m 09s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/d088e4de7f5d4ca8b9a99a2c1b502c6b : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":35},{"id":"2b7fcae8d70de922377f2259bbedff39fecc608e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-03 19:33:16.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"f366b2b522e4937710d50553713734ee7028c586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:08:24.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59785517dbe14dc78b479dd5e96aeed6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02b20c4266d495dbe26639021cf0b49 : SUCCESS in 4m 42s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/011e0f1d31a84e549dedb3a136c23be1 : FAILURE in 29m 16s","accounts_in_message":[],"_revision_number":36},{"id":"79e528244ef9d176980c145e5c520c4869505d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-07 14:53:27.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"27f5bf93711b912806a9013cfea9ecaeb15e4f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 15:23:32.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b23f9fba6de4877b0e6d6a4c37163c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334465f1d1d84de6829df137d2f9ae67 : SUCCESS in 4m 32s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/ac7c247141314f7a9e47a5f1e641f0a8 : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":37},{"id":"15f277d6317840809770a6d5e0be3cdb3d1ba1b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 18:51:44.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"b16f2b38800fa3d47d5fb674f184704e675f22a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 19:30:20.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28436ec94ab84c6c8a0c7bfa3207a6ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3092c94a94e3461ebc31b18a44b1abbe : SUCCESS in 4m 27s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/9e4ccb79319249a2b1bf59aaa33d3335 : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":38},{"id":"a8761393b0f0343339023bd0bb5858562e25e958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-24 01:43:51.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"53f220142760fd45c535e3b253ad9186a7a16b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 01:52:50.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b286a4906934f0fac0eb83f2ebe0ce4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d690dbb74644ddaf73df242f2d856d : SUCCESS in 3m 10s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/b9c9551a37584e9ebe8f119a27a809ed : NODE_FAILURE Node request 300-0024555685 failed in 0s","accounts_in_message":[],"_revision_number":39},{"id":"5be002f5280c6d5025a462b1e447b9efe0dab0a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-24 02:15:00.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"ecc9f01956cd5d8cb03595f7977503eea095841e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 02:45:02.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38c3b8d1e0d0465f80a0af119e262b94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b49c05a6174f7c894d5d03317a9baf : SUCCESS in 3m 40s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/bd2eddeecc18419394b4f7fcd44ff781 : FAILURE in 22m 12s","accounts_in_message":[],"_revision_number":40},{"id":"50f7acf8985999ad22a9156adf767f24aacde347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-07-16 04:12:45.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"6c5990f2dd47a4957a9d21017f1a0871dcc3faff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 04:44:36.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9780a17fc96c44e69634485e31ccecf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf47e6117e3421a8173df3f138d1105 : SUCCESS in 3m 07s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/975759a3c9e842e9ab232148d49fa0ae : SUCCESS in 19m 55s","accounts_in_message":[],"_revision_number":41},{"id":"bf6d398f6df556cc1a9ceeca0c89e3601a070f16","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-21 04:50:13.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"f2149d2f1cd1168d5ae4f173271b92fe3f65d603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 04:50:33.000000000","message":"Patch Set 42: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027groups\u0027][0][\u0027node\u0027]\n\nThe problem appears in the the \"requirements-generate-constraints-multinode\" job stanza:\n\n  job:\n      name: requirements-generate-constraints-multinode\n      description: |\n        Generate constraints on multiple nodes and merge into a single unified constraints\n        file\n      parent: base\n      run: playbooks/constraints/run.yaml\n      branches: master\n      attempts: 1\n  ...\n\n  in \"openstack/requirements/.zuul.d/jobs.yaml@master\", line 136","accounts_in_message":[],"_revision_number":42},{"id":"330e601030bcbbd45cab7d0a6b34b6b565afa781","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 04:50:47.000000000","message":"Patch Set 42:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027groups\u0027][0][\u0027node\u0027]\n\nThe problem appears in the the \"requirements-generate-constraints-multinode\" job stanza:\n\n  job:\n      name: requirements-generate-constraints-multinode\n      description: |\n        Generate constraints on multiple nodes and merge into a single unified constraints\n        file\n      parent: base\n      run: playbooks/constraints/run.yaml\n      branches: master\n      attempts: 1\n  ...\n\n  in \"openstack/requirements/.zuul.d/jobs.yaml@master\", line 136","accounts_in_message":[],"_revision_number":42},{"id":"6b3ea63da878e112f590df8974a5b18290f9a40e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-21 04:58:03.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"17758da920b814de5e7f3f12c734241c100264da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 04:58:45.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c329f39e8a447438f8aa8a8e0fa8651\n\n- noop https://zuul.opendev.org/t/openstack/build/a90f85f049b34f69b0d7676af9f074df : SUCCESS in 0s","accounts_in_message":[],"_revision_number":43},{"id":"45504768e2bdd6ab04b2ceaeb0870f16957c2b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 05:07:38.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e6021c05c71482e9926f613a6472b13\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e9e51506c87a4253a86fb391221790f5 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e879688446a24d7b81cf420259009182 : SUCCESS in 3m 47s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/c1e4bebe73674b92b5d47209fd60ee82 : FAILURE in 53s","accounts_in_message":[],"_revision_number":43},{"id":"f0486c8a51b6438671bb35fd55db758a78f92e72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-21 05:14:40.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"1eb47966b1847aff3da54050dffecb80ae425b88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 05:15:14.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5653ece9434b3db51c38152154f6f2\n\n- noop https://zuul.opendev.org/t/openstack/build/8314ba7997eb407394b4936556634699 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":44},{"id":"98ea2068189d3f548ecd5f1a3869bbf71c5523c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 05:24:12.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec9f80b7b5be4e358326b25f987fcd21\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/304a6b811cd24f9ebfcf8f413e880442 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15279d5498e848bb8ced650a01830f92 : SUCCESS in 3m 48s\n- requirements-generate-constraints-multinode https://zuul.opendev.org/t/openstack/build/fff569cb0c7542b089a048d0193a407f : FAILURE in 2m 55s","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"d6cc3739779f05316f529eb4c5419946f3502fee","revisions":{"e34689715437387d2aca2e324fa3ce1f3f4e374e":{"kind":"REWORK","_number":1,"created":"2024-01-18 05:16:07.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/1"}}},"commit":{"parents":[{"commit":"da82fe614247dcab41ba2b8bf7f544404b1e7580","subject":"Merge \"move mistral to nonvoting for eventlet revert\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da82fe614247dcab41ba2b8bf7f544404b1e7580"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e34689715437387d2aca2e324fa3ce1f3f4e374e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e34689715437387d2aca2e324fa3ce1f3f4e374e"}]},"branch":"refs/heads/master"},"9809fe18a7980a18da423b2e38ae99faf7a69795":{"kind":"REWORK","_number":2,"created":"2024-01-18 05:23:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/2"}}},"commit":{"parents":[{"commit":"8d67991af046bb5a5c5a172440588f4b0bcf6249","subject":"update constraint for python-tackerclient to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:23:23.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9809fe18a7980a18da423b2e38ae99faf7a69795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9809fe18a7980a18da423b2e38ae99faf7a69795"}]},"branch":"refs/heads/master"},"ad00e88c096388d5e850de07ba2bb7e57c6d6e3c":{"kind":"REWORK","_number":3,"created":"2024-01-18 05:29:03.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/3"}}},"commit":{"parents":[{"commit":"8d67991af046bb5a5c5a172440588f4b0bcf6249","subject":"update constraint for python-tackerclient to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:28:44.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad00e88c096388d5e850de07ba2bb7e57c6d6e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad00e88c096388d5e850de07ba2bb7e57c6d6e3c"}]},"branch":"refs/heads/master"},"55f5e85f757464b665e2bb4c2ab1e819f5ca4021":{"kind":"REWORK","_number":4,"created":"2024-01-18 08:52:36.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/4"}}},"commit":{"parents":[{"commit":"8d67991af046bb5a5c5a172440588f4b0bcf6249","subject":"update constraint for python-tackerclient to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 08:52:17.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55f5e85f757464b665e2bb4c2ab1e819f5ca4021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55f5e85f757464b665e2bb4c2ab1e819f5ca4021"}]},"branch":"refs/heads/master"},"3bdc1e06113d927d7f0400e5262becb4930d8461":{"kind":"REWORK","_number":5,"created":"2024-01-18 09:51:42.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/5"}}},"commit":{"parents":[{"commit":"8d67991af046bb5a5c5a172440588f4b0bcf6249","subject":"update constraint for python-tackerclient to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 09:51:25.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bdc1e06113d927d7f0400e5262becb4930d8461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bdc1e06113d927d7f0400e5262becb4930d8461"}]},"branch":"refs/heads/master"},"4f656c9087829909c4d065fc94b10a1f15cbb778":{"kind":"REWORK","_number":6,"created":"2024-01-18 10:00:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/6"}}},"commit":{"parents":[{"commit":"8d67991af046bb5a5c5a172440588f4b0bcf6249","subject":"update constraint for python-tackerclient to new release 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d67991af046bb5a5c5a172440588f4b0bcf6249"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 10:00:37.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f656c9087829909c4d065fc94b10a1f15cbb778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f656c9087829909c4d065fc94b10a1f15cbb778"}]},"branch":"refs/heads/master"},"7e74f67309c70a2f4a9a85bec7512590703f66c9":{"kind":"REWORK","_number":7,"created":"2024-01-23 20:36:32.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/7"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-23 20:36:28.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e74f67309c70a2f4a9a85bec7512590703f66c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e74f67309c70a2f4a9a85bec7512590703f66c9"}]},"branch":"refs/heads/master"},"7d3acdf1d58d5d65ada2b4c5417f2a1d0a9e317e":{"kind":"REWORK","_number":8,"created":"2024-01-23 22:56:32.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/8"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-23 22:56:27.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d3acdf1d58d5d65ada2b4c5417f2a1d0a9e317e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d3acdf1d58d5d65ada2b4c5417f2a1d0a9e317e"}]},"branch":"refs/heads/master"},"b4981ee2eae7a8e51360b1c78a312262b1694a1b":{"kind":"REWORK","_number":9,"created":"2024-01-23 22:56:46.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/9"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-23 22:56:42.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4981ee2eae7a8e51360b1c78a312262b1694a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4981ee2eae7a8e51360b1c78a312262b1694a1b"}]},"branch":"refs/heads/master"},"cd2b1505c646e244a855fc97e02fdcd22c91fa63":{"kind":"REWORK","_number":10,"created":"2024-01-23 23:11:41.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/10"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-23 23:11:37.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd2b1505c646e244a855fc97e02fdcd22c91fa63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd2b1505c646e244a855fc97e02fdcd22c91fa63"}]},"branch":"refs/heads/master"},"80c4ed3b75311796ee65abf40360240195ba1562":{"kind":"REWORK","_number":11,"created":"2024-01-23 23:12:15.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/11"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-23 23:12:11.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80c4ed3b75311796ee65abf40360240195ba1562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80c4ed3b75311796ee65abf40360240195ba1562"}]},"branch":"refs/heads/master"},"8c7638ac10a82e3a5b1e4fe6065bb1341714eeef":{"kind":"REWORK","_number":12,"created":"2024-01-24 02:06:46.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/12"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-24 02:06:42.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c7638ac10a82e3a5b1e4fe6065bb1341714eeef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c7638ac10a82e3a5b1e4fe6065bb1341714eeef"}]},"branch":"refs/heads/master"},"8ac366685535d4c6efaca6a715f4f9822e4a34be":{"kind":"REWORK","_number":13,"created":"2024-01-24 03:14:38.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/13"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-24 03:14:34.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ac366685535d4c6efaca6a715f4f9822e4a34be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ac366685535d4c6efaca6a715f4f9822e4a34be"}]},"branch":"refs/heads/master"},"e1247c7976aacd35ff6c7e35f54516ab6c79d0be":{"kind":"REWORK","_number":14,"created":"2024-01-24 05:11:57.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/14"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-24 05:11:53.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1247c7976aacd35ff6c7e35f54516ab6c79d0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1247c7976aacd35ff6c7e35f54516ab6c79d0be"}]},"branch":"refs/heads/master"},"a21c3beb0f7ff15f709b049a647fe3037433b6db":{"kind":"REWORK","_number":15,"created":"2024-01-24 06:49:47.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/15"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-24 06:49:44.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a21c3beb0f7ff15f709b049a647fe3037433b6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a21c3beb0f7ff15f709b049a647fe3037433b6db"}]},"branch":"refs/heads/master"},"496404299925a8bd6f72d7ee45c8cc25c324fb1c":{"kind":"REWORK","_number":16,"created":"2024-01-24 09:58:46.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/16"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-24 09:58:42.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/496404299925a8bd6f72d7ee45c8cc25c324fb1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/496404299925a8bd6f72d7ee45c8cc25c324fb1c"}]},"branch":"refs/heads/master"},"524af7916ae93ac8c74b730d1132458761e57b6e":{"kind":"REWORK","_number":17,"created":"2024-01-24 23:26:20.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/17"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-24 23:26:15.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/524af7916ae93ac8c74b730d1132458761e57b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/524af7916ae93ac8c74b730d1132458761e57b6e"}]},"branch":"refs/heads/master"},"0fe144dddf49ae39b5299b394b489efe10014f34":{"kind":"REWORK","_number":18,"created":"2024-01-25 00:37:33.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/18"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-25 00:37:28.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fe144dddf49ae39b5299b394b489efe10014f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fe144dddf49ae39b5299b394b489efe10014f34"}]},"branch":"refs/heads/master"},"e9a8d5cf1313e2a851b9f09fcdc2266ba78969c3":{"kind":"REWORK","_number":19,"created":"2024-01-25 01:16:37.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/19"}}},"commit":{"parents":[{"commit":"13f27bd15c478df868af6c67f67b1c4f3e782745","subject":"Downgrade pillow to 9.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-25 01:16:33.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9a8d5cf1313e2a851b9f09fcdc2266ba78969c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9a8d5cf1313e2a851b9f09fcdc2266ba78969c3"}]},"branch":"refs/heads/master"},"24e0c5a25c8c842fb082584a344ef536e5e155af":{"kind":"REWORK","_number":20,"created":"2024-01-31 00:30:36.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/20"}}},"commit":{"parents":[{"commit":"2f843f3cfe9a61a44c0b3d3d1135d9231818a211","subject":"Merge \"Allow locally requiring a package for specific python versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f843f3cfe9a61a44c0b3d3d1135d9231818a211"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-31 00:30:32.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24e0c5a25c8c842fb082584a344ef536e5e155af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24e0c5a25c8c842fb082584a344ef536e5e155af"}]},"branch":"refs/heads/master"},"480f2767d510f57fcade3f34323afcab36171126":{"kind":"REWORK","_number":21,"created":"2024-01-31 01:40:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/21"}}},"commit":{"parents":[{"commit":"2f843f3cfe9a61a44c0b3d3d1135d9231818a211","subject":"Merge \"Allow locally requiring a package for specific python versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f843f3cfe9a61a44c0b3d3d1135d9231818a211"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-31 01:40:39.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/480f2767d510f57fcade3f34323afcab36171126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/480f2767d510f57fcade3f34323afcab36171126"}]},"branch":"refs/heads/master"},"804f4c8bc36c12d8196a58e95c205d820a982835":{"kind":"REWORK","_number":22,"created":"2024-02-02 01:19:19.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/22"}}},"commit":{"parents":[{"commit":"2f843f3cfe9a61a44c0b3d3d1135d9231818a211","subject":"Merge \"Allow locally requiring a package for specific python versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f843f3cfe9a61a44c0b3d3d1135d9231818a211"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-02 01:19:15.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/804f4c8bc36c12d8196a58e95c205d820a982835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/804f4c8bc36c12d8196a58e95c205d820a982835"}]},"branch":"refs/heads/master"},"4f431b56b0301d72131a945c7b4586a9e4a8b1c6":{"kind":"REWORK","_number":23,"created":"2024-02-11 23:43:12.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/23"}}},"commit":{"parents":[{"commit":"1d8a9890961c44ac9f1b991db61b87baa653293d","subject":"Remove unused abclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-11 23:43:08.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f431b56b0301d72131a945c7b4586a9e4a8b1c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f431b56b0301d72131a945c7b4586a9e4a8b1c6"}]},"branch":"refs/heads/master"},"0fa0c3ca37eea9046f29db6a373d6484d79732c6":{"kind":"REWORK","_number":24,"created":"2024-02-12 00:33:14.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/24"}}},"commit":{"parents":[{"commit":"1d8a9890961c44ac9f1b991db61b87baa653293d","subject":"Remove unused abclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 00:33:10.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fa0c3ca37eea9046f29db6a373d6484d79732c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fa0c3ca37eea9046f29db6a373d6484d79732c6"}]},"branch":"refs/heads/master"},"b226cd2f7237f3bcc3569a2cf96c646dacbccc7e":{"kind":"REWORK","_number":25,"created":"2024-02-12 01:30:32.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/25"}}},"commit":{"parents":[{"commit":"1d8a9890961c44ac9f1b991db61b87baa653293d","subject":"Remove unused abclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 01:30:28.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b226cd2f7237f3bcc3569a2cf96c646dacbccc7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b226cd2f7237f3bcc3569a2cf96c646dacbccc7e"}]},"branch":"refs/heads/master"},"e8d2c0ace5dcbf1a1d096920a2cc521d0c9006e5":{"kind":"REWORK","_number":26,"created":"2024-02-12 01:37:41.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/26"}}},"commit":{"parents":[{"commit":"1d8a9890961c44ac9f1b991db61b87baa653293d","subject":"Remove unused abclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 01:37:37.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8d2c0ace5dcbf1a1d096920a2cc521d0c9006e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8d2c0ace5dcbf1a1d096920a2cc521d0c9006e5"}]},"branch":"refs/heads/master"},"cf8696fe5a08733a134cbad51b6df0b144e4fff4":{"kind":"REWORK","_number":27,"created":"2024-02-12 01:57:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/27"}}},"commit":{"parents":[{"commit":"1d8a9890961c44ac9f1b991db61b87baa653293d","subject":"Remove unused abclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 01:57:39.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf8696fe5a08733a134cbad51b6df0b144e4fff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf8696fe5a08733a134cbad51b6df0b144e4fff4"}]},"branch":"refs/heads/master"},"241412197760a325196512b3e13b7a5ba1f98585":{"kind":"REWORK","_number":28,"created":"2024-02-12 06:55:24.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/28"}}},"commit":{"parents":[{"commit":"c6ed4ace09be2a62cbbd1747b85bb0db84a4d4e0","subject":"Add merge-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6ed4ace09be2a62cbbd1747b85bb0db84a4d4e0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 06:33:06.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/241412197760a325196512b3e13b7a5ba1f98585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/241412197760a325196512b3e13b7a5ba1f98585"}]},"branch":"refs/heads/master"},"7e2cf6e837cbf62e2b7f4bcddfd848869c0cbea4":{"kind":"REWORK","_number":29,"created":"2024-02-13 23:51:19.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/29"}}},"commit":{"parents":[{"commit":"9e038849f5bad750040060a1395f2a09f5ba5862","subject":"Add merge-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e038849f5bad750040060a1395f2a09f5ba5862"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-13 23:38:17.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e2cf6e837cbf62e2b7f4bcddfd848869c0cbea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e2cf6e837cbf62e2b7f4bcddfd848869c0cbea4"}]},"branch":"refs/heads/master"},"c967dfd672b9e5aa3d7140f48f0797e804ada6da":{"kind":"REWORK","_number":30,"created":"2024-02-14 00:30:34.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/30"}}},"commit":{"parents":[{"commit":"9e038849f5bad750040060a1395f2a09f5ba5862","subject":"Add merge-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e038849f5bad750040060a1395f2a09f5ba5862"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-14 00:30:25.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c967dfd672b9e5aa3d7140f48f0797e804ada6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c967dfd672b9e5aa3d7140f48f0797e804ada6da"}]},"branch":"refs/heads/master"},"1df522c348ed8de68c080916541506b4c05187d9":{"kind":"REWORK","_number":31,"created":"2024-02-15 00:06:39.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/31"}}},"commit":{"parents":[{"commit":"9e038849f5bad750040060a1395f2a09f5ba5862","subject":"Add merge-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e038849f5bad750040060a1395f2a09f5ba5862"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-15 00:05:58.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1df522c348ed8de68c080916541506b4c05187d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1df522c348ed8de68c080916541506b4c05187d9"}]},"branch":"refs/heads/master"},"6df1549bfc4e8f25b83bcd4d9ff836be4d769443":{"kind":"REWORK","_number":32,"created":"2024-02-20 21:20:36.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/32"}}},"commit":{"parents":[{"commit":"9e038849f5bad750040060a1395f2a09f5ba5862","subject":"Add merge-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e038849f5bad750040060a1395f2a09f5ba5862"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-20 21:20:02.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6df1549bfc4e8f25b83bcd4d9ff836be4d769443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6df1549bfc4e8f25b83bcd4d9ff836be4d769443"}]},"branch":"refs/heads/master"},"60364fc5af364fb390d1fe247eeff36f779b6a89":{"kind":"REWORK","_number":33,"created":"2024-02-21 05:53:45.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/33"}}},"commit":{"parents":[{"commit":"9e038849f5bad750040060a1395f2a09f5ba5862","subject":"Add merge-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e038849f5bad750040060a1395f2a09f5ba5862"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-21 05:53:23.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/60364fc5af364fb390d1fe247eeff36f779b6a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/60364fc5af364fb390d1fe247eeff36f779b6a89"}]},"branch":"refs/heads/master"},"756c6fc513385d5261bb1a55a35390698f6a3706":{"kind":"REWORK","_number":34,"created":"2024-02-22 23:46:49.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/34"}}},"commit":{"parents":[{"commit":"cbdb8f38dc70470d5a640760746769c001cba3ba","subject":"Add merge-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbdb8f38dc70470d5a640760746769c001cba3ba"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-22 23:46:31.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nThe basic theory here is that we can deploy a node with known versions\nof python installed.\nOn each node generate a set of constraints for each known version\nthen on a control node merge and freeze those generated constraints\nand ultimately propose that as a new upper-constraints.txt\n\nThis is an enhancement as we\u0027re using *real* versions of python can\nsupport SLURP runtimes more accurately.  At the other end as can\nincorporate unreleased runtimes which will eventually be needed.\n\nThis is very work-in-progress.\nTODO (in no particular order):\n - Verify pause/return can be used for inter node communications\n   - It\u0027s possible that this would be better structured with explict\n     requires/provides and job dependencies ... time and review will\n     tell\n - Actually generate valid venvs and constraints\n - Split the merge function out of generate\n - Incorporate work from https://review.opendev.org/q/topic:requirements-ops\n   for better debugging and human readability\n - review/fix/merge into openstack/project-config\n - run in parallel with existing implementation\n - #profit/relax\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/756c6fc513385d5261bb1a55a35390698f6a3706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/756c6fc513385d5261bb1a55a35390698f6a3706"}]},"branch":"refs/heads/master"},"c89f5099f86a3075406e505c5585d87aeacf26f9":{"kind":"REWORK","_number":35,"created":"2024-02-24 02:25:55.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/35"}}},"commit":{"parents":[{"commit":"74ae1bfdfa53faf89c93ec51a859fcc3af97c781","subject":"[DNM] Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74ae1bfdfa53faf89c93ec51a859fcc3af97c781"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-24 02:25:33.000000000","tz":660},"subject":"[DNM] Testing multinode multiversion python constraints generation","message":"[DNM] Testing multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c89f5099f86a3075406e505c5585d87aeacf26f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c89f5099f86a3075406e505c5585d87aeacf26f9"}]},"branch":"refs/heads/master"},"55973436869c3f35070c728cd442ae1c5b30067d":{"kind":"REWORK","_number":36,"created":"2024-05-03 19:33:16.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/36"}}},"commit":{"parents":[{"commit":"f2607eae3c86d1b0cbb97ba8f1aabbac8584f390","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2607eae3c86d1b0cbb97ba8f1aabbac8584f390"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-03 19:30:18.000000000","tz":-300},"subject":"Testing multinode multiversion python constraints generation","message":"Testing multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55973436869c3f35070c728cd442ae1c5b30067d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55973436869c3f35070c728cd442ae1c5b30067d"}]},"branch":"refs/heads/master"},"1be7cd6c8f9a59bd8abcf7a00452715b93cb419a":{"kind":"REWORK","_number":37,"created":"2024-05-07 14:53:27.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/37"}}},"commit":{"parents":[{"commit":"a4011c16c47648991c7d30e08c28fb37e42fdc07","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4011c16c47648991c7d30e08c28fb37e42fdc07"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-07 14:52:57.000000000","tz":-300},"subject":"Testing multinode multiversion python constraints generation","message":"Testing multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1be7cd6c8f9a59bd8abcf7a00452715b93cb419a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1be7cd6c8f9a59bd8abcf7a00452715b93cb419a"}]},"branch":"refs/heads/master"},"1ff789e7862f106d2783efb1874e74e238f62b78":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2024-05-09 18:51:44.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/38"}}},"commit":{"parents":[{"commit":"ff7708e0b6cdd6cbd6b0808790caa3663073366e","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff7708e0b6cdd6cbd6b0808790caa3663073366e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 18:51:17.000000000","tz":-300},"subject":"Testing multinode multiversion python constraints generation","message":"Testing multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ff789e7862f106d2783efb1874e74e238f62b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ff789e7862f106d2783efb1874e74e238f62b78"}]},"branch":"refs/heads/master"},"0d48916db1a1a803ea96e9457ebdb85aa8bca1fc":{"kind":"REWORK","_number":39,"created":"2024-06-24 01:43:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/39"}}},"commit":{"parents":[{"commit":"e4e441337113d7a3e8a780d9c12ded889edb74cb","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4e441337113d7a3e8a780d9c12ded889edb74cb"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-06-24 01:43:13.000000000","tz":600},"subject":"Testing multinode multiversion python constraints generation","message":"Testing multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d48916db1a1a803ea96e9457ebdb85aa8bca1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d48916db1a1a803ea96e9457ebdb85aa8bca1fc"}]},"branch":"refs/heads/master"},"d76273f062cbdf83afa75b4cc02755d16ee81976":{"kind":"REWORK","_number":40,"created":"2024-06-24 02:15:00.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/40"}}},"commit":{"parents":[{"commit":"e4e441337113d7a3e8a780d9c12ded889edb74cb","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4e441337113d7a3e8a780d9c12ded889edb74cb"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-06-24 02:14:40.000000000","tz":600},"subject":"Testing multinode multiversion python constraints generation","message":"Testing multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d76273f062cbdf83afa75b4cc02755d16ee81976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d76273f062cbdf83afa75b4cc02755d16ee81976"}]},"branch":"refs/heads/master"},"6d0b56461aa53310d3cd062e3c2feb48973f722f":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2024-07-16 04:12:45.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/41"}}},"commit":{"parents":[{"commit":"1faa376b035f107d870ec4d2c2fd1e3a484de8b4","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1faa376b035f107d870ec4d2c2fd1e3a484de8b4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-18 05:05:17.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-07-16 04:12:29.000000000","tz":600},"subject":"Testing multinode multiversion python constraints generation","message":"Testing multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d0b56461aa53310d3cd062e3c2feb48973f722f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d0b56461aa53310d3cd062e3c2feb48973f722f"}]},"branch":"refs/heads/master"},"025369e600773438956f24a6ccd66b6d978ff741":{"kind":"REWORK","_number":42,"created":"2025-10-21 04:50:13.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/42"}}},"commit":{"parents":[{"commit":"afdb761008007830bb831db17840352db9781823","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afdb761008007830bb831db17840352db9781823"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 06:32:30.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-21 04:49:44.000000000","tz":660},"subject":"Many much refactoring needed","message":"Many much refactoring needed\n\nCreate a merge-constraints tool\n\nCurrently we have a tool call generate-constraints that operates like so:\n\nforeach supplied python_version\n  Create a venv\n  Install global-requirements in that venv\n  generate a pip frezze\n\nforeach freeze combine\n  Combine into a single set of constrains.\n\nUse the \u0027proposal\u0027 framework in project-config to generate a gerrit\nchange with the new constraints\n\nThere is also the ability to simulate a python_version, but this isn\u0027t\naccurate.  The major limitation is we can only generate accurate\nconstraints for python versions supplied by the distribution.\n\nThis change and the following changes will alter that process.\n\nStep one: (this change)\n Move the generation and combination code into a library\n Create a new tool call merge-constraints that will take individually\n   create freeze outputs and combine them to generate a single\n   constraints file.\n\nThis is done without disrupting the existing generate-constratints tool\n\nStep two:\n\nAdd a multinode job that will use ansible to generate a freeze for all\navailable python versions.  This is multinode so we can add new OSs and\npython releases somewhat trivially.\n\nAt the time of writing:\n  Use ubuntu-jammy to generate py3.10 and py3.11 freezees\n  Use ubuntu-noble to generate py3.12, py3.13, and py3.14 freezees\n\nStep three:\n\nAdd another job that will use ansible to take the freezes generated in\nstep two and combine them into a single constraints file\n\nStep four, and beyond: (to be written)\n\nAdd the new jobs into the proposal framework alongside the existing\ngenerate-constraints tool for comparison.  Ensure that the bulk of the\ntools and job definitions are in openstack/requirements with only the\nproposal logic in openstack/project-config, it needs to be there to\naccess the zuul secrets.Disable and remove the older tool once we\u0027re\nhappy with the new tool.\n\nChange-Id: If4f9d1adb9c259dc0fbe067ff6cf9e007d697acf\n\nTesting multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/025369e600773438956f24a6ccd66b6d978ff741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/025369e600773438956f24a6ccd66b6d978ff741"}]},"branch":"refs/heads/master"},"49d7a74a4c6e72440820d249d1c6fd63b4536b97":{"kind":"REWORK","_number":43,"created":"2025-10-21 04:58:03.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/43"}}},"commit":{"parents":[{"commit":"afdb761008007830bb831db17840352db9781823","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afdb761008007830bb831db17840352db9781823"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 06:32:30.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-21 04:57:48.000000000","tz":660},"subject":"Many much refactoring needed","message":"Many much refactoring needed\n\nCreate a merge-constraints tool\n\nCurrently we have a tool call generate-constraints that operates like so:\n\nforeach supplied python_version\n  Create a venv\n  Install global-requirements in that venv\n  generate a pip frezze\n\nforeach freeze combine\n  Combine into a single set of constrains.\n\nUse the \u0027proposal\u0027 framework in project-config to generate a gerrit\nchange with the new constraints\n\nThere is also the ability to simulate a python_version, but this isn\u0027t\naccurate.  The major limitation is we can only generate accurate\nconstraints for python versions supplied by the distribution.\n\nThis change and the following changes will alter that process.\n\nStep one: (this change)\n Move the generation and combination code into a library\n Create a new tool call merge-constraints that will take individually\n   create freeze outputs and combine them to generate a single\n   constraints file.\n\nThis is done without disrupting the existing generate-constratints tool\n\nStep two:\n\nAdd a multinode job that will use ansible to generate a freeze for all\navailable python versions.  This is multinode so we can add new OSs and\npython releases somewhat trivially.\n\nAt the time of writing:\n  Use ubuntu-jammy to generate py3.10 and py3.11 freezees\n  Use ubuntu-noble to generate py3.12, py3.13, and py3.14 freezees\n\nStep three:\n\nAdd another job that will use ansible to take the freezes generated in\nstep two and combine them into a single constraints file\n\nStep four, and beyond: (to be written)\n\nAdd the new jobs into the proposal framework alongside the existing\ngenerate-constraints tool for comparison.  Ensure that the bulk of the\ntools and job definitions are in openstack/requirements with only the\nproposal logic in openstack/project-config, it needs to be there to\naccess the zuul secrets.Disable and remove the older tool once we\u0027re\nhappy with the new tool.\n\nChange-Id: If4f9d1adb9c259dc0fbe067ff6cf9e007d697acf\n\nTesting multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49d7a74a4c6e72440820d249d1c6fd63b4536b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49d7a74a4c6e72440820d249d1c6fd63b4536b97"}]},"branch":"refs/heads/master"},"d6cc3739779f05316f529eb4c5419946f3502fee":{"kind":"REWORK","_number":44,"created":"2025-10-21 05:14:40.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/906022/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/906022/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/906022/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/906022/44"}}},"commit":{"parents":[{"commit":"afdb761008007830bb831db17840352db9781823","subject":"Fix sorting in _combine_freezes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afdb761008007830bb831db17840352db9781823"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-02-12 06:32:30.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-21 05:14:32.000000000","tz":660},"subject":"Many much refactoring needed","message":"Many much refactoring needed\n\nCreate a merge-constraints tool\n\nCurrently we have a tool call generate-constraints that operates like so:\n\nforeach supplied python_version\n  Create a venv\n  Install global-requirements in that venv\n  generate a pip frezze\n\nforeach freeze combine\n  Combine into a single set of constrains.\n\nUse the \u0027proposal\u0027 framework in project-config to generate a gerrit\nchange with the new constraints\n\nThere is also the ability to simulate a python_version, but this isn\u0027t\naccurate.  The major limitation is we can only generate accurate\nconstraints for python versions supplied by the distribution.\n\nThis change and the following changes will alter that process.\n\nStep one: (this change)\n Move the generation and combination code into a library\n Create a new tool call merge-constraints that will take individually\n   create freeze outputs and combine them to generate a single\n   constraints file.\n\nThis is done without disrupting the existing generate-constratints tool\n\nStep two:\n\nAdd a multinode job that will use ansible to generate a freeze for all\navailable python versions.  This is multinode so we can add new OSs and\npython releases somewhat trivially.\n\nAt the time of writing:\n  Use ubuntu-jammy to generate py3.10 and py3.11 freezees\n  Use ubuntu-noble to generate py3.12, py3.13, and py3.14 freezees\n\nStep three:\n\nAdd another job that will use ansible to take the freezes generated in\nstep two and combine them into a single constraints file\n\nStep four, and beyond: (to be written)\n\nAdd the new jobs into the proposal framework alongside the existing\ngenerate-constraints tool for comparison.  Ensure that the bulk of the\ntools and job definitions are in openstack/requirements with only the\nproposal logic in openstack/project-config, it needs to be there to\naccess the zuul secrets.Disable and remove the older tool once we\u0027re\nhappy with the new tool.\n\nChange-Id: If4f9d1adb9c259dc0fbe067ff6cf9e007d697acf\n\nTesting multinode multiversion python constraints generation\n\nChange-Id: Ic1501ceaaf701e3b0e294a8535ae2c17d07c1133\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6cc3739779f05316f529eb4c5419946f3502fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6cc3739779f05316f529eb4c5419946f3502fee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
