)]}'
{"id":"openstack%2Frequirements~906110","triplet_id":"openstack%2Frequirements~master~I911ac0fbdf510441510fd626c8094bc8d7e82e07","project":"openstack/requirements","branch":"master","topic":"remove-etcd3","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-01-19 16:40:10.000000000","reason":"removed on reply"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-22 04:20:14.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-22 04:20:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I911ac0fbdf510441510fd626c8094bc8d7e82e07","subject":"Drop etcd3","status":"MERGED","created":"2024-01-19 13:39:10.000000000","updated":"2024-01-22 04:21:18.000000000","submitted":"2024-01-22 04:20:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"906110-remove-etcd3","meta_rev_id":"ad919be48916f4213c82ff2d861e6819ad6a9697","_number":906110,"virtual_id_number":906110,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-22 04:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-19 16:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-22 01:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-01-22 01:53:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-19 15:28:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-19 16:12:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-01-19 16:29:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-01-22 01:53:04.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"a636929cc724aa655eefa9d02e248b955c4b7a05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 13:39:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06bd559ad8b43d59f0a2acb59022cc8963317ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:28:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82bd9ec2267941949a6e02ea99e0b80c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050a9c109a8341a784ced8fd0b366e97 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd73a9d938734915b6a0cd0bbaf27385 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d1a4b94c124099a71586938bc63015 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7014619c64a94192be675e07e9dc640b : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13c2474a6c664c2fb003a35093039d93 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/298633951f8e43aaa9828682f9afe845 : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0dc33cea8614caeb34d583dd1949f7e : SUCCESS in 8m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a732ec7a19474ad4934d177ef172107e : SUCCESS in 14m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b520518aa00142279095857ff5d5fbd0 : SUCCESS in 8m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9f81da243ff34927b9132b0c4b327807 : SUCCESS in 17m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c3169d8daaec4c77b7358bb074627a96 : SUCCESS in 4m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7ba57ceeefdd405abb02aca2c058f224 : SUCCESS in 4m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d3468355b9c24c34bec8ec0ba8c960e1 : SUCCESS in 4m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5031a03698474f918b37287693abb4c2 : SUCCESS in 11m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dabe45b8f2b74846b0c04745839cae07 : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/add5cf3fdf71492191c33b795b4251c7 : SUCCESS in 5m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/44978127527d4e429c465c642ed83582 : SUCCESS in 8m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4a230bd4998046e797f13106e728ded5 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e2a2cde962948d6b7721002a6a42d04 : SUCCESS in 6m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fde68ea7826c4b0ab05e390b4ad1742b : SUCCESS in 8m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6422962062024d3794c52d822b9d1d54 : SUCCESS in 12m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7e016cb2ba37415e9ae8171115228b54 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/04ded8c74002439a84301793b507d893 : SUCCESS in 10m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/647a070a044746408ed6475e07d57d09 : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6c5ed6f9e600451f91e0805077cde872 : SUCCESS in 13m 31s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c8dfe46d217243dbbf8dc9c97cc99409 : SUCCESS in 24m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/51f7bdafaaa14509a5f61ff47b3a9975 : SUCCESS in 6m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f959f8804e5f43f5bc97c9b42acf0ca0 : SUCCESS in 13m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b799c8ce2ad04c588b9d9319737f9758 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b0f3f989340c4f26869267d1af97b542 : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32f00e49c66d4b5b8d1943488721f908 : SUCCESS in 23m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/416682aabcd340c8bcd2c9723aa0540f : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/93156bd1533b46b69f5faa301aa841be : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6e8bc620f1e4ee39145fb83031d2c09 : SUCCESS in 5m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/aee721ab22174844912f376536b33a8c : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/546cf62e719e4dd38d27c00660d9e766 : SUCCESS in 35m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/942c5ec8a4e84a1e90a3f57010249a01 : SUCCESS in 1h 43m 52s","accounts_in_message":[],"_revision_number":1},{"id":"21dbf5fda1783a78c8e953ca9adcde1cad13164f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-19 16:12:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cad56a478beb88e6669ac5f42cab5c8e72ef250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 16:15:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bf09f7be62e3fdc2a82da9e1e21c168979d29905","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 16:15:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f51239b2ffd07915465938cfde5df152733b647f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-19 16:29:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e53c0d1e8ee7e7b137f92957924e379f00f4373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 16:36:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cdc44a4bbfc39ccf9c54c0e8df730bbc52467f26","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-19 16:38:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6554e498888ec2d0f53c43b8b4a66c25f2662b3d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-19 16:40:10.000000000","message":"Patch Set 3: Code-Review+2\n\nfair enough","accounts_in_message":[],"_revision_number":3},{"id":"ebd4c3bd710a3c119eddf57bbed073f6873e5332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 18:43:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43c9de7bd16c41d2a48d6522813fc285\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d51fdbdfdf4a1bb1c04f278995bdcc : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2a7272e50444c0a073ca4975558bed : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/513d47ef89f14913acb9219a4f45c22f : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/581febab9e0d4964b72e6ff8f2f17e7e : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e80e89bf018846d2a7a4e6550cd3d47e : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e54b390f4354ae781f5a6bdc5cb2bf4 : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37e11f4bae954e33af0142b8daf537eb : SUCCESS in 9m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1adc7ba5ed7b44af8c202afd152f60c1 : SUCCESS in 14m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/907595ec33604736bdd3caad16096088 : SUCCESS in 8m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2d41f934fa12435fbd4ca9bcc423a06c : SUCCESS in 15m 16s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5cbd9a7dba064d3c8ea3f8ce12c83861 : SUCCESS in 6m 39s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f73b6572f89a4aa9a20730833f823a6c : SUCCESS in 7m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a6a75f16ac934e4cb6f286d54e38f775 : SUCCESS in 4m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e90c13e9bac74d42b179271f922a1797 : SUCCESS in 13m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/29c635b14a884e5cb29683fe5ae5148d : SUCCESS in 5m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2c5f162208d14d78995ed942d88d3b53 : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/50d2732b3a524e57bb5b0c7743c366d2 : SUCCESS in 8m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b8b7754a1529413fb8ca78a0b12d0207 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6771c805cf8c48ca9331eb4d55e1f719 : SUCCESS in 6m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/358ac7d3cfbb4eb197d1a356eb97fbad : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4dc2b3e9338d429d972eb0233bfa55a8 : SUCCESS in 12m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/18020cb924fa4d67b08cbc95db10901c : SUCCESS in 4m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/512676bca892431a912a2a0985fe4d2c : SUCCESS in 10m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/75f35f5abaf74f91a1b89bb0d1088e04 : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6ce7b3ddad0c4afb9d8dba55f675878d : SUCCESS in 14m 37s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/46be0901f0fb474c9b5fc27edb6c9d55 : SUCCESS in 25m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a557b3757c7040348481ad609b28d35d : SUCCESS in 7m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/67d75b01af254fe7b3807c583b53aee0 : SUCCESS in 13m 49s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d5535a986f1146cca588bbba8ddff8bd : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b3836dc4c7e48489f972ef7f9988062 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8c68533c94443fbbdfa3b4043aab373 : SUCCESS in 23m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/01f4895730d549cdb24b2ab8276e702e : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/881d41e8969344ab80dca4589c79be69 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a34e791268941a6857b045888838ef4 : SUCCESS in 6m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9211f2474cd54b40bd8bf1e818d17a42 : SUCCESS in 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d6a25b59cf94166b794ff99c4498182 : SUCCESS in 31m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9dddee41afd474fa492576815880ce9 : SUCCESS in 1h 56m 34s","accounts_in_message":[],"_revision_number":3},{"id":"5185dcb231b611a20f19acdd677926fffbd06c6f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-22 01:53:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c3410f4f72142ef5a76caa9c62a561409a8607c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 01:53:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"83c316661d9ed1d1a4f34153151739b32ab6ae71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 04:20:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b4974cc79d74922a8f83c24dc1c638c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23321843df324731971dfa9999ef113d : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8be7a76c7d5404689ab9f8567cf0f74 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75467f2746c44d10a114d45c98e856f0 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70ce08ace5544ff2a66e7fb12ddc985c : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47c047cbb7404d2a88cb7874a550125c : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/428d4c1bc1e742008962f7582f46aca4 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64257246ec1343e39e49a4f0e9999643 : SUCCESS in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56ae9a6d342244369578ea2d62baddfa : SUCCESS in 14m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/30a12813c206468e8eece02d29e37edb : SUCCESS in 10m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ecfcc0a41065466881f480fb7a7af497 : SUCCESS in 16m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ed355c1bd48f40c9bd7fb5ef47afbbe2 : SUCCESS in 5m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d092daa37796462f91f9e3986bc52e49 : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e8f7fefbb92a465d922a68fbc0c8fa91 : SUCCESS in 8m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6cccbde5d0934b1298ccc05efecf237d : SUCCESS in 16m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e35296ef7d740a7a3de54131c64d536 : SUCCESS in 7m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/416716a776e24b81862141aa8473add7 : SUCCESS in 9m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d4291db8f1b44948b1bb21247c836234 : SUCCESS in 12m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8974408832ee42f091b149670ee8e25a : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1bbdc771e9c34754a6ab7c51af14eb77 : SUCCESS in 9m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2057889e519040bab66d174c061f21fa : SUCCESS in 29m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5bbc8eda2ce744a7b5ddda0c7e2b2a0e : SUCCESS in 8m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a75ff016ddd84816832e954ecfc65a31 : SUCCESS in 18m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/594d59ab970140329e45e5f8f7bbf1b4 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f265d9dceef94bb48eec444b7f2af0f2 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/382516f98aab444e8302ef18ac033dc7 : SUCCESS in 25m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d541d1296f544611801efa30f7b77a18 : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6718e3e9d6b445e98b0161d98d9e159f : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec405f1e96cc4c7287a9dd81620b536d : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/86617b5539c64fc3acab5b7407b5f84f : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e8eea33a91040e78289face3907602e : SUCCESS in 24m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/676aa25120df45ce94ae47adb1b741e1 : SUCCESS in 1h 15m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fc7663d7d99350d21aa4ff82c041377ad97d347e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 04:20:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ad919be48916f4213c82ff2d861e6819ad6a9697","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 04:21:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d966251336294507a972bee75e8c9897\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8703330f87be4b30ad5ac221bf1e618f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d26730b811e6e8c20387cfe30704d818b934698e","revisions":{"b8be23561eaba8ea9a93e8ab4d796b402a152a12":{"kind":"REWORK","_number":1,"created":"2024-01-19 13:39:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/906110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/906110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/906110/1"}}},"commit":{"parents":[{"commit":"db23c452a54bdc1eb0d312ff92895065e45043ee","subject":"Merge \"update constraint for oslo.limit to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db23c452a54bdc1eb0d312ff92895065e45043ee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-19 13:35:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-19 13:35:33.000000000","tz":540},"subject":"Drop etcd3","message":"Drop etcd3\n\nThis library was added for tooz[1] but we dropped the etcd3 driver from\ntooz during this cycle[2], in favor of alternative options such as\netcd3gw driver and inactivity in the etcd3 library.\n\n[1] dbce88da79ef815fa69cd4e49146c4ce6bddd05b\n[2] 6bc02cda5b452bbf4821621eafc031bd676f8a2f\n\nChange-Id: I911ac0fbdf510441510fd626c8094bc8d7e82e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8be23561eaba8ea9a93e8ab4d796b402a152a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8be23561eaba8ea9a93e8ab4d796b402a152a12"}]},"branch":"refs/heads/master"},"2433648aa76bef126522717026c62271d3fcc38e":{"kind":"REWORK","_number":2,"created":"2024-01-19 16:15:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/906110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/906110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/906110/2"}}},"commit":{"parents":[{"commit":"db23c452a54bdc1eb0d312ff92895065e45043ee","subject":"Merge \"update constraint for oslo.limit to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db23c452a54bdc1eb0d312ff92895065e45043ee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-19 13:35:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-19 16:15:22.000000000","tz":540},"subject":"Drop etcd3","message":"Drop etcd3\n\nThis library was added for tooz[1] but we dropped the etcd3 driver from\ntooz during this cycle[2], in favor of alternative options such as\netcd3gw driver and inactivity in the etcd3 library.\n\n[1] dbce88da79ef815fa69cd4e49146c4ce6bddd05b\n[2] 6bc02cda5b452bbf4821621eafc031bd676f8a2f\n\nChange-Id: I911ac0fbdf510441510fd626c8094bc8d7e82e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2433648aa76bef126522717026c62271d3fcc38e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2433648aa76bef126522717026c62271d3fcc38e"}]},"branch":"refs/heads/master"},"d26730b811e6e8c20387cfe30704d818b934698e":{"kind":"REWORK","_number":3,"created":"2024-01-19 16:36:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/906110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/906110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/906110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/906110/3"}}},"commit":{"parents":[{"commit":"db23c452a54bdc1eb0d312ff92895065e45043ee","subject":"Merge \"update constraint for oslo.limit to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db23c452a54bdc1eb0d312ff92895065e45043ee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-19 13:35:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-19 16:36:45.000000000","tz":540},"subject":"Drop etcd3","message":"Drop etcd3\n\nThis library was added for tooz[1] but we dropped the etcd3 driver from\ntooz during this cycle[2], in favor of alternative options such as\netcd3gw driver and inactivity in the etcd3 library.\n\n[1] dbce88da79ef815fa69cd4e49146c4ce6bddd05b\n[2] 6bc02cda5b452bbf4821621eafc031bd676f8a2f\n\nChange-Id: I911ac0fbdf510441510fd626c8094bc8d7e82e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d26730b811e6e8c20387cfe30704d818b934698e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d26730b811e6e8c20387cfe30704d818b934698e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
