)]}'
{"id":"openstack%2Frequirements~906262","triplet_id":"openstack%2Frequirements~master~I58f0c06f122fd3b5ad349d1bdd25d5417dc0e39d","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-22 21:17:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I58f0c06f122fd3b5ad349d1bdd25d5417dc0e39d","subject":"Downgrade pillow to 9.5.0","status":"MERGED","created":"2024-01-22 12:32:44.000000000","updated":"2024-01-22 21:18:06.000000000","submitted":"2024-01-22 21:17:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906262","meta_rev_id":"8221aae42f0babe2bfd70741f6e7de1b98f99841","_number":906262,"virtual_id_number":906262,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-22 21:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-01-22 12:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-01-22 20:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-22 14:33:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-01-22 12:57:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-22 12:57:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-22 13:22:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2024-01-22 14:01:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 14:33:10.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-01-22 20:07:35.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"10f0280c123a38287c18bd6db56b3c377b3adc7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 12:32:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0da8be590289df1af0dbcba18874b9429f282d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 12:33:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1a52dca486933bd557ff1a60b619fbe46a212692","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 12:48:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7f0d0a75cfc09ef36452382858cd5d6a93ea0b4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-22 12:57:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f767106ee1e90752d089fb16093288d2bcc937d2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-01-22 13:22:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"272b69a5a9d2e1dfd0b300266659220c504932b4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-22 13:46:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33db36294c880be42062f70599784d9488f1d345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 14:01:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/553d6c9bafe747b3992aeb348ced231a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5541b6abaf114ebbab8cd1f915b6e478 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b675aeb60b4848847bc70b928bb1ee : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/246b7d1d50b141cc869a591cae837525 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74433878fa414916a5058bb9ba1e11c9 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c8358fda9534788965bd2c434120f5e : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7e7bb801e534c1bbb6d2ca8f0a5287f : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40c9bf3ed8384a36b564edeeb56a3d35 : SUCCESS in 8m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/647d8bb94be847489ef9f684d78f46ec : SUCCESS in 16m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/db36efdbb68a41138e05bf7264aad81b : SUCCESS in 8m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5ccd895caa1f487db5ce8ef4a08e338e : SUCCESS in 15m 42s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7af18e2edb2d446493698211f611be2e : SUCCESS in 4m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/92761edc425742ea8735bf92e73f44a5 : SUCCESS in 4m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1e0f02f83d344b159ac6fcccb7ad6811 : SUCCESS in 4m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4b1211abbed847f995d63c9ed29c1326 : SUCCESS in 13m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/93ee798449b04d31a3a19484cae7cbcd : SUCCESS in 4m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ed3df8c64c82452f98873d14abd88790 : SUCCESS in 5m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/06cfc40575f24fa8916f90c7c2254f29 : SUCCESS in 7m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/99dc0a525a8d48f48d4c52ea63df8777 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88f3c91e24f14608ba4d2c4f0da73a00 : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5f95136514de4e1b814ca55573259c48 : SUCCESS in 8m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cc99e699b932471e923056bb60762738 : SUCCESS in 10m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4d5d6918c85f4785a15782dba87cd38c : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/95f4c8f41203450eb5110000edb9d4aa : SUCCESS in 9m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/54ee89b2aaee42808b51d0c90c84a18b : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ced38b0385d94133a00ed4eb5c736860 : SUCCESS in 13m 34s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/49803e80e690433e89ddf1501ed9a533 : SUCCESS in 23m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8b615659c47479eb32ebdda2b00bc49 : SUCCESS in 8m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/791aeafebca4455d98b9755b704dbdc6 : SUCCESS in 12m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/46e30d35cb6d4e67be40359b00931ee3 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed73f10cae8844a09a21426d396582b7 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/474968daf055489ebb8200bb16ca9d34 : TIMED_OUT in 40m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c171a5c771c94d5ab2ce5c1803ec0b39 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/30c25ef79058465db8701a9ef4cb11c0 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8a2b9b8175140d29f14014de0bdbdc8 : SUCCESS in 7m 12s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b41eac749ff34eedb2458f1a8c7db5f0 : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e31ccfaa390a45fabc658e8b2ea8e283 : SUCCESS in 36m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df827ed3c3a942cb94662465b0c61830 : SUCCESS in 1h 22m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fc8137a287d21ab5f1d81fc36b194b2a8c6cf1d8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 14:01:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4d3faccce2b8a85aef5a71be01508a818c9b683","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-01-22 14:33:10.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d8034832c93d3c3ac4ac557dcbcbeed017be02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:19:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b04fe348ae464a6493d9d3c375f0db17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e2bfadde0b4cd0b20442329eeca768 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d09285cfe174e75b929746a57fb3a25 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca4c836d19394914b8cc9ac79a5d4b63 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f15e06c040c4566b1c332877f772d15 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/950157b399cd4c93bd3b5157eb9b0a58 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37ba90e31891449aab32350807a19186 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/71a6a76feab6455e9be764b5d4ec7aaa : SUCCESS in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a0059a222729425a9c3046cf4c0eb003 : SUCCESS in 14m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/73dd04f0de2a4735afd183b489472830 : SUCCESS in 12m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f3992c360b0d49c894f54f56f653cd63 : SUCCESS in 18m 12s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fba3d13ca278418eaa143b0bfb1e1700 : SUCCESS in 4m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8dca72308b844910bb1abdfbdd496efb : SUCCESS in 4m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/cbdc31cd23cb447eb019928f441f2c10 : SUCCESS in 4m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d9d4c8d57f704718824682fae76d067b : SUCCESS in 10m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/92e4a6212b0743efb09e3e68384a6255 : SUCCESS in 5m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9f7ac4163fac4a109664da3162a1ecea : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5fe9ffc0d78a4f4594bea9f70f7b9d95 : SUCCESS in 8m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/543fa0f309a64d668bdef3c41e2768ef : SUCCESS in 13m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbb1ed92aecc484483493115720f6885 : SUCCESS in 7m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5cc944a92f414ba58e44036a9b6fa04b : SUCCESS in 9m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/408db4b2a9fa4f3c9a134d9697e76250 : SUCCESS in 13m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/33851cba02f54418853da36a48ae67df : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e0bf49ac54084f11bad490919463081c : SUCCESS in 10m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/23860ef111604b03a5742576ce8f1723 : SUCCESS in 4m 31s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5f5e65924b924ab7ac4fb28886e60087 : SUCCESS in 13m 02s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5eac1530e7e04f7884dd0e49eaf5c37d : SUCCESS in 28m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8199ca9260c246c6ba35115205978b60 : SUCCESS in 7m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/87390d1d2af845f58b3024d61577f2fd : SUCCESS in 14m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4804b9557fed431c825355700379950c : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b66fdf751194259b9bdba6d4a480f41 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b351cca552aa4dfcbebb12b1812fc3b1 : SUCCESS in 26m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1e977d897a044e53a81695aa66a36ddc : SUCCESS in 5m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/54fe8066951e4de98c967c5b7829dc28 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11a371d427e24d4288cc70b02fda8f71 : SUCCESS in 6m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3ccf72ff74e944aeb3c98d09dcb8d231 : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9515621fe45b45caa337dd7ccdf932fd : SUCCESS in 39m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/642afa12e2534e54b8570852d2109c6e : TIMED_OUT in 2h 04m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9f49f550b3d56bec3d6861a5719a60e9972ddf9c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 17:03:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc95d17d2fb327a5f0f18329ae45681bf9f1f8d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 19:20:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e821548eebe45bfa9a75819c8d69439\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a58799e3a84b4eae4636136ceda691 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69669123163e4ebbb1e5110029126095 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e0d016b405449eeb07596b7e4c77be2 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9117e98b5c894010998f8b258c7612ae : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40f590f09e564f95bf771342371f26a2 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/869d9f855c3142768ea13967c326cf88 : SUCCESS in 5m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a5868da3791147aea9d007459ebe8abc : SUCCESS in 11m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/055c0505c0244bd58e213e86d05607de : SUCCESS in 15m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/20d237b6bec94795a9ec46e7e0a44c4a : SUCCESS in 10m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/acde250934a64597b65559760516599e : SUCCESS in 19m 29s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d1c2e21c4caf4cceb8d3fae1c58ac272 : SUCCESS in 4m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c19d4968dd064b24916f78d15c98f083 : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a8f4278834fc49b4a1ca451044868ee9 : SUCCESS in 4m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9f3c6d243ecd4a4591bb5cda660994f8 : SUCCESS in 13m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3075ea707ca04e118691ada1de4672e3 : SUCCESS in 7m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3a81b5bd06094a609d4c155aca8365a6 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c458c126556a4f30ae06bb8286408767 : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ff72e1c9ca3244dab79cb203a948a066 : SUCCESS in 13m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b7fedab529e4530ada74f006bb59c30 : SUCCESS in 7m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ad9d2d4a15dd44ea9b48a046d913d2d9 : SUCCESS in 7m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e4860571e7e6481b893b76e34ccfd9f0 : SUCCESS in 12m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/caa27b0bf00b43fea84a11247f5708e3 : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c8942bcacc874391841a33a7541ee2a0 : SUCCESS in 9m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c8b28e0cf30a4c6aa75c14357c36ec76 : SUCCESS in 7m 12s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a5b6092046b841cd8a83decacc19431e : SUCCESS in 12m 01s (non-voting)\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/80bcb63550b94faba81f69412af896b1 : SUCCESS in 25m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0411da82c1ec4d75bc75599345fe9373 : SUCCESS in 7m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/421dae6296c845c0b258457844287a2c : SUCCESS in 17m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/38bbf7043805448ea42303943e4a9365 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5940db3b584e4adea6cc533e59662807 : SUCCESS in 5m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ae69491537446289d536238f5b23299 : SUCCESS in 28m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c410db932075454c91505054422f5808 : SUCCESS in 6m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9a963ab121134748941da2ba3a91f7f7 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ded7d49f54a14995a18e4699950dac03 : SUCCESS in 8m 58s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0934b94720c24d36aeb0e4a5a8dc2be9 : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc0981fcfbaf4cd8b6cc666a79901758 : SUCCESS in 45m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a41399eb81e44fbbbcf9b0d2a3b489c : SUCCESS in 1h 30m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a24109de3f95e9b7872add063336a83852e44337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 19:20:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f171fbca1ae0017c91b0bf2209d7d3775b96168a","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-01-22 20:07:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d99e60c7d09721cd1524a0d10d942cdaeb1deb57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 21:17:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb4a2e9ef9745bc908d246f71d1485c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0970fc78586245d087e48d998faf3d5a : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2242e6d0263745c68ca7d95447d9d4f5 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eacf8bc62e864abba84f63b7657c1fed : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3eed8fe5959d49998d7d1647f909ba4d : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17056d164d764f05bcbe85b0bea521f6 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/471946a6fe06431c97ba6bb60a21bd8c : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae87280d83a4483dae527b25a8e4ee0f : SUCCESS in 10m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d5658e710cc9473f91b312113bbf7d77 : SUCCESS in 15m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b1b31e33c04543f3a7e4e76626436db0 : SUCCESS in 9m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/828632ceb98740e9b703d6385266736a : SUCCESS in 11m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ca73c6bdddbf4689ba14c6498d76e630 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/19dd107f028c4945aaf49bc7f5a72895 : SUCCESS in 4m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/862cfc98ab6b427db6994afe0133d9b1 : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a50147d642af4869880faab66f74edbd : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c941b379b1a6406ab0234d5665018770 : SUCCESS in 7m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cf4ddbfaf2e645e7bfcc22a1a1685f66 : SUCCESS in 7m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5643cc4f2678463cb85fe683bb08b688 : SUCCESS in 12m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/423cef5f3db04235be4ff5f9c1cc507c : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9d2181e157b14035b116bd9bf70ad471 : SUCCESS in 9m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0a7a0ba400df49a8a72c97dac5f584df : SUCCESS in 27m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/08b223e21f684b08ad61c56a5bd849a4 : SUCCESS in 7m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1edca7c6d3ca4685b1abb80ac9e2ebf8 : SUCCESS in 12m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6af374280ce64d12b33b30d69583567d : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3686739df6f46a3b4e2a81f53ca4764 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7c90a9faf224b4a985b7809d68acd1b : SUCCESS in 25m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2746d90969bc4c7b8bf294b8e242057f : SUCCESS in 5m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/769d7ab71bc34727a2491cd70b39dc0e : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68734d887dd44595af1684689184d5ec : SUCCESS in 6m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2491c3b578324f9f892ace4ca4213f94 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/334e17d5ecc84395b3ecaf599ad44a87 : SUCCESS in 36m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5e08c7129a8444fae16704029b7042a : SUCCESS in 1h 35m 19s","accounts_in_message":[],"_revision_number":2},{"id":"fe1cd45273a2c4c0938e28d61d86dd76aef0d641","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 21:17:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8221aae42f0babe2bfd70741f6e7de1b98f99841","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 21:18:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc7b2ef3cad043e5bd48cd5e040f9fa6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629b3b45c7f34f1ebffe524609664959 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13f27bd15c478df868af6c67f67b1c4f3e782745","revisions":{"56f0ec2c8743e121eb352a882e3be50ab698b97e":{"kind":"REWORK","_number":1,"created":"2024-01-22 12:32:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/906262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/906262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/906262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/906262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/906262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/906262/1"}}},"commit":{"parents":[{"commit":"b684251042d3e6342a3d87a76ebf8c0eef5ae17e","subject":"Merge \"Remove directord and task-core\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b684251042d3e6342a3d87a76ebf8c0eef5ae17e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 12:31:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 12:31:33.000000000","tz":540},"subject":"Downgrade pillow to 9.5.0","message":"Downgrade pillow to 9.5.0\n\nThe latest blockdiag is not compatible with pillow 10.y.z[1].\n\nThe library looks unmaintained we may have to consider replacing it,\nbut in the mean time let\u0027s downgrade pillow to fix broken nova doc\nbuilds.\n\n[1] https://github.com/blockdiag/blockdiag/issues/178\n\nChange-Id: I58f0c06f122fd3b5ad349d1bdd25d5417dc0e39d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56f0ec2c8743e121eb352a882e3be50ab698b97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56f0ec2c8743e121eb352a882e3be50ab698b97e"}]},"branch":"refs/heads/master"},"13f27bd15c478df868af6c67f67b1c4f3e782745":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-22 12:33:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/906262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/906262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/906262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/906262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/906262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/906262/2"}}},"commit":{"parents":[{"commit":"b684251042d3e6342a3d87a76ebf8c0eef5ae17e","subject":"Merge \"Remove directord and task-core\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b684251042d3e6342a3d87a76ebf8c0eef5ae17e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 12:31:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 12:33:12.000000000","tz":540},"subject":"Downgrade pillow to 9.5.0","message":"Downgrade pillow to 9.5.0\n\nThe latest blockdiag is not compatible with pillow 10.y.z[1].\n\nThe library looks unmaintained, so we may have to consider replacing\nit, but let\u0027s downgrade pillow to fix broken nova doc builds.\n\n[1] https://github.com/blockdiag/blockdiag/issues/178\n\nChange-Id: I58f0c06f122fd3b5ad349d1bdd25d5417dc0e39d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f27bd15c478df868af6c67f67b1c4f3e782745"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
