)]}'
{"id":"openstack%2Frequirements~906475","triplet_id":"openstack%2Frequirements~stable%2F2023.1~Ia32864c9620434fe4f139ba5c295997265cacc73","project":"openstack/requirements","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-24 21:13:14.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-01-24 21:13:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia32864c9620434fe4f139ba5c295997265cacc73","subject":"[stable-only] Fix gate by capping tox version","status":"MERGED","created":"2024-01-24 09:24:51.000000000","updated":"2024-01-24 21:13:14.000000000","submitted":"2024-01-24 21:13:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906475","meta_rev_id":"384730ed0306b809915163cca1ae57a4dc3d27ac","_number":906475,"virtual_id_number":906475,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-24 21:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-01-24 19:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-01-24 17:28:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-01-24 19:21:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 10:05:52.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-01-24 15:57:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 17:28:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-24 19:21:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b48864b498dc16479ce0da6743c1600289d45746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 09:24:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dcc188d31df8afe29afe07dd75f6c525a704f66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 09:44:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f46f5d2fb923138441d8182c2da2dee6e0898261","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-24 10:05:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b6b8ff7b240b5255aac40b4c536be307a2bf1fd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 10:17:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e518c83a4350fa806a8f78bc69b593762b4b02a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 10:36:52.000000000","message":"Abandoned\n\nhowever this fixes the py38 and pep8 jobs the py3*-check-uc jobs are still failing. (pkg_resources.extern.packaging.version.InvalidVersion: Invalid version: \u0027cpython\u0027)","accounts_in_message":[],"_revision_number":2},{"id":"1c17277515cb5045ee128cabc7d526f1b5c3530f","tag":"autogenerated:gerrit:restore","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 14:33:25.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"321c725635bfa43a139b3f81e060813ef345f729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 14:33:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ddafb9149864ec22163183d1a4034231be9eee39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 15:57:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53a19b870aa44e10a639a9df54b934b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08c9a5caa724b59b9ee46458124bcca : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e07d936bc2454e38bf9743b0747dd5c5 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d31d4d5c64049649c54f05d4ecbcbd0 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a013ddc6de44f15aea78d02e89a1da2 : SUCCESS in 3m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af1e42bb1d20416f8f5c5694a2ee742e : SUCCESS in 1h 01m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2ee5e1ad70d4cfb858525e33e44bc5d : SUCCESS in 1h 18m 27s","accounts_in_message":[],"_revision_number":3},{"id":"c67d9522971f47dfa81b076284029e02502ce370","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-24 17:28:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"796edea7f6eccbc7b7c7d0870ffc8400cb70fbe8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-24 19:21:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cd9659426db28feba5db4364afeacf092e4d5a6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 19:21:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5b55aa30750bd0fe88056c86b7bd19e0859fe723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:13:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a862e9086f624b10be5158ffb9e9c36f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498f65e1ece0438eb5528d82f11543e2 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4deba8e46d3e4da6ab92157916e9569c : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/314ccc9b237049f08c151e955f15def2 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03cb970a806c457d9462cb09cf5bc3be : SUCCESS in 4m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67119b06fc4d4399a30639da1c2aa0fe : SUCCESS in 1h 04m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8993fe932299431ebb6a9024a770572d : SUCCESS in 1h 34m 16s","accounts_in_message":[],"_revision_number":3},{"id":"384730ed0306b809915163cca1ae57a4dc3d27ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:13:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ac6c3f2a9c923ec1d716680fcef06b26c467b9e","revisions":{"a633cef37838ef8f977ef8718fba1f16842a19ff":{"kind":"REWORK","_number":1,"created":"2024-01-24 09:24:51.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/75/906475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/906475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/906475/1"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 09:23:43.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 09:23:43.000000000","tz":60},"subject":"DNM: test requirements gate","message":"DNM: test requirements gate\n\nChange-Id: Ia32864c9620434fe4f139ba5c295997265cacc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a633cef37838ef8f977ef8718fba1f16842a19ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a633cef37838ef8f977ef8718fba1f16842a19ff"}]},"branch":"refs/heads/stable/2023.1"},"a9ecc497e708937d5acd2348893c9f049f67c98b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-24 09:44:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/75/906475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/906475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/906475/2"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 09:23:43.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 09:38:26.000000000","tz":60},"subject":"[stable-only] Fix gate by removing virtualenv cap","message":"[stable-only] Fix gate by removing virtualenv cap\n\npep8 and py38 jobs are failing [1] on this branch for some time. It\nlooks like some not pinned requirements changed in the meantime (like\nsetuptools or other base package) and thus causes the issue, which can\nbe eliminated by removing virtualenv capping which is not needed\nanymore.\n\n[1] https://zuul.opendev.org/t/openstack/build/d60a9b657513486e8bda21019e7cb589/log/job-output.txt#1155-1157\n\nFile \"/home/zuul/src/opendev.org/openstack/requirements/.tox/.tox/lib/python3.8/site-packages/tox/tox_env/python/virtual_env/package/pyproject.py\", line 310, in _ensure_meta_present\n2024-01-15 11:00:35.514374 | ubuntu-focal |     dist_info \u003d end.prepare_metadata_for_build_editable(self.meta_folder, self._wheel_config_settings).metadata\n2024-01-15 11:00:35.514477 | ubuntu-focal | AttributeError: \u0027NoneType\u0027 object has no attribute \u0027metadata\u0027\n\nChange-Id: Ia32864c9620434fe4f139ba5c295997265cacc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9ecc497e708937d5acd2348893c9f049f67c98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9ecc497e708937d5acd2348893c9f049f67c98b"}]},"branch":"refs/heads/stable/2023.1"},"5ac6c3f2a9c923ec1d716680fcef06b26c467b9e":{"kind":"REWORK","_number":3,"created":"2024-01-24 14:33:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/75/906475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/906475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/906475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/906475/3"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 09:23:43.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-01-24 14:30:35.000000000","tz":60},"subject":"[stable-only] Fix gate by capping tox version","message":"[stable-only] Fix gate by capping tox version\n\npep8, py38 and py3*-check-uc jobs are failing [1] on this branch for\nsome time. This patch caps tox version to get back the former behaviour\nfor the jobs and make them pass.\n\n[1] https://zuul.opendev.org/t/openstack/build/d60a9b657513486e8bda21019e7cb589/log/job-output.txt#1155-1157\n\nFile \"/home/zuul/src/opendev.org/openstack/requirements/.tox/.tox/lib/python3.8/site-packages/tox/tox_env/python/virtual_env/package/pyproject.py\", line 310, in _ensure_meta_present\n2024-01-15 11:00:35.514374 | ubuntu-focal |     dist_info \u003d end.prepare_metadata_for_build_editable(self.meta_folder, self._wheel_config_settings).metadata\n2024-01-15 11:00:35.514477 | ubuntu-focal | AttributeError: \u0027NoneType\u0027 object has no attribute \u0027metadata\u0027\n\nChange-Id: Ia32864c9620434fe4f139ba5c295997265cacc73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ac6c3f2a9c923ec1d716680fcef06b26c467b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ac6c3f2a9c923ec1d716680fcef06b26c467b9e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
