)]}'
{"id":"openstack%2Frequirements~907285","triplet_id":"openstack%2Frequirements~master~Icc16ef36b6a7f8a3341108d0bf512604c19a9c77","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-31 19:30:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc16ef36b6a7f8a3341108d0bf512604c19a9c77","subject":"Bump SQLAlchemy to 1.4.51","status":"MERGED","created":"2024-01-31 06:17:07.000000000","updated":"2024-01-31 19:31:44.000000000","submitted":"2024-01-31 19:30:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907285-openstack/requirements/constraints/noclob","meta_rev_id":"4aa79dfc9ddd94f8de92ce7d6c2c19934acc0b4f","_number":907285,"virtual_id_number":907285,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-31 19:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-01-31 11:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-01-31 16:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-01-31 16:46:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-31 07:51:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-31 11:03:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-01-31 16:46:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"46e56b0ba4c9678e44675e6ee9023f6f36c1d45d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-31 06:17:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b6f3dbdcc301744fc4a8872cf0c77f52dcff316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 07:51:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95f2db3efbba48a19872844485cff8f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec16f1fbf621451e9b069adbbb7d0649 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3a7d2ab8fb4b5393ecec7f6484bdfd : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/715ebeee4fce4660a2d688266b9ccc99 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4df94643b69d4682a1b81b6be23fb822 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d96fddff36242c8a4875045b26d87f1 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/900e9d361c2e451cb2f718455180909d : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/978b8c48b8384f7984c1363976da14dc : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9d9653bcb2804b4fb6365abcdf1a7fe7 : SUCCESS in 16m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c88ccc2758a44ad2a34fb30f99cc1df0 : SUCCESS in 10m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/08efd4abecdf4ca38e4a8a5a299f42a8 : SUCCESS in 17m 29s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/77eddb8ed68f402791391c0f692885a8 : SUCCESS in 5m 59s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1debb005d3d24ba88bbbba6c07d9def5 : SUCCESS in 5m 10s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0151760fcb30405eb6ea0ed4fcc97150 : SUCCESS in 3m 58s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/471f19d9e95e45ed838d58524fc8af65 : SUCCESS in 10m 48s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/71802523b64147bd9ffe1d46f3d203ea : SUCCESS in 5m 53s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7538f627d692463a80c307b9d8069fdb : SUCCESS in 6m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/bbaac1af7ceb48cbbd57124fe72d51db : SUCCESS in 7m 32s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/67d2255f416a4c48b20781c89b609e48 : SUCCESS in 15m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ade3c777e18449fbf6c9f58c1952bf5 : SUCCESS in 8m 06s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6229e88df46641989e71e3ecff685d3e : SUCCESS in 8m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/99030465bfda40358625d7f0a87c6031 : SUCCESS in 10m 51s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ff35c3b1f868494ebc22883c5782a05f : SUCCESS in 5m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/98f0abaa332245859f3f6bb05fc9b9a9 : SUCCESS in 8m 04s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/9e48b224db6944b5a85e09d284fb0487 : SUCCESS in 5m 48s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9b94dc42e9354517bec0e0a3fbe8d19b : SUCCESS in 11m 36s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/27f722a2e4d44bbeacbc5333e44f08f6 : SUCCESS in 21m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/614a798fd3ca495c915845d26b4fae7a : SUCCESS in 8m 13s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/437e2a4df4ca4e4583455b5ab1d5ba7b : SUCCESS in 13m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d334247d8067476782857af43680fa3e : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aca177bd969e4e2a922c031d4b1e0e7a : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b49796159934d7a9b286757bf3de0ca : SUCCESS in 19m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c911a7f91c7147409f224f66da5dc138 : SUCCESS in 4m 34s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a0d69e7c811d41d9a407efdef224c93a : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29faa3327bec49488fa07c321350f873 : SUCCESS in 8m 47s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/810e101721fa4758a71d36b7e6ed896f : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ce08c01117d491f92931c4ccf14bf16 : FAILURE in 33m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29a351c8b0494cddbea44651b38f1454 : SUCCESS in 1h 27m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8b76e316c56833b06385efaf39e33af050881419","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-31 08:00:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49e70f3468d5bd37a1900a709bf4b924fe75719f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 09:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f51aecd2f348468faf0da46f21929d6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd02ce617f34b5eaf93b84a286c59b4 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99b8c5ee7b9422bb30b9f679a80be7f : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ccfe6ad34bf4a38bcd152e1bea90ebc : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fd642081dd742c18f82a5937917a654 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b212ac264084ecdb65e34c135e74870 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46a71acd7e0142688c29373d52ededfe : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/77119986decf4cb487c7165e0fbe97f5 : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1816943a59c44a9a91d37a999b336e55 : SUCCESS in 16m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f13036df53d462aad5723bfa2011d1e : SUCCESS in 10m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/98fb2190af6942568b4439cd173b6b6a : SUCCESS in 19m 39s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/63eadae2475f43b5a223574ad3904881 : SUCCESS in 4m 56s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/839c7b13024c42cfbfee3da0704d6140 : SUCCESS in 7m 27s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/fef0f53e3bc742269199c524f3ed9900 : SUCCESS in 4m 18s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5be217e4bf0343e39ab3ba9045e27d1f : SUCCESS in 10m 51s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c1a0690abbd64053a7ca020eb21de8ad : SUCCESS in 6m 10s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ef22d26c53024e6da89262f34ca83579 : SUCCESS in 5m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7a14ad9284d844a890cddb841e70f2ef : SUCCESS in 7m 25s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/93dd4a1b92f44d49961a309a9e947c1b : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e697f1c3b814b509ff3fc2b2c243bcd : SUCCESS in 8m 34s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/04c802e5c5994de18266361de5630056 : SUCCESS in 7m 30s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e09b4accaca94ea597c9e4a8e07e57c1 : SUCCESS in 10m 12s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7b50724b749f4deda54bb33f4f271392 : SUCCESS in 5m 34s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/12f6a3b5b6b04202a1491754ed827c44 : SUCCESS in 7m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0e2f2ce23af14f1b82bf492fdfdca816 : SUCCESS in 5m 05s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/5fea0c2ed7fb43d1a90260a700d74966 : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3fd52aa0103c422ea70c8d50366fdf5f : SUCCESS in 24m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/51577bb70a634e52ae1785231c4de47a : SUCCESS in 6m 39s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c975b02ddb2a43ceb0d6a1c2e0b84872 : SUCCESS in 13m 38s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6b0e51dbb3d2432cbc2b1c0170e0f5e4 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b46324e2bf1e4dd8b0758c13c667cd5f : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fad4f80ff3e04f948660c754f80d905f : SUCCESS in 22m 07s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/79f203b897f143cfaac7d8720f177768 : SUCCESS in 4m 27s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/55352eafbfcc49cc83ab1a15e944b4fc : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d61a302d3c004fc2aaf86136b97f60d0 : SUCCESS in 6m 32s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/232f028dfe364fdc976534dee2aa7785 : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd0eb8ca853d4aabaa64f6410ab11c7f : SUCCESS in 26m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb281f43567e49a38db6fa67c2dbd587 : SUCCESS in 1h 28m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6839e74b85f93cf2a30f6ac7ff4501db2125fb12","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-01-31 11:03:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff932ec8f303dc735a57d2a6b962da4c2df04bc7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-01-31 16:46:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33357bd7463462b41f87b338f2598dd12ba9e995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 16:47:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b74e4d39e1c8171a3e538e89be8c28b0f2e51db3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 19:30:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27087c6f8cad413c99b0aff0b0d230b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f69a6b18d474410af5cb70c3f0daf2c : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e970c33a694773852bf5593a70a3d6 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c106f6d7184437896c3603fd675fb30 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9b208a62d23477188ee6b7a4c825322 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c56424d416834da194fc36d92ac7bd30 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9b01e183af514f388c37d9f248248c4d : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bce818b4154540139beb8a3e579ab1c6 : SUCCESS in 10m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b4a9bbfff63d4163961b59ced4dfd258 : SUCCESS in 15m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d238f5daff6422da14041b4a93c958d : SUCCESS in 9m 58s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e29f8fb1cc5f408d9dc84db72a595015 : SUCCESS in 11m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d3a9f659e20a45599c0008963a116ca1 : SUCCESS in 5m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1f1ca880973240a59937f0018b3a60b0 : SUCCESS in 5m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/60db2c3933564d5cb69677982da934e3 : SUCCESS in 7m 39s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8d405fed4b274c36abe2e69399755346 : SUCCESS in 13m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/efaa34e61e0946e896aa1cdf9f8e18a1 : SUCCESS in 7m 32s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e77b785e4c5c41299dc9dccf09126a4e : SUCCESS in 7m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/7679e2593ab34cffa6cdc7da78b7f4c3 : SUCCESS in 8m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3f542624f4344f7fa0f3da85eda0c6b6 : SUCCESS in 5m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/53ce4be303824e3fa46b6c1546be2df8 : SUCCESS in 9m 47s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c4b95ff0ea1a4c7eaeaf765d34cc07f9 : SUCCESS in 26m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/135e44fcc12245d592e9ae84daa0c6a7 : SUCCESS in 9m 17s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4a13fe0a0a24410884cba5ea611c3c92 : SUCCESS in 13m 55s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/440f636274a8424699e8c6f4ff408d7c : SUCCESS in 6m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea2efb764aaa49059a29af67e9c18141 : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b40f79fc64f42a98af8fb371d56c591 : SUCCESS in 23m 26s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/10faef87869a473997c7045f97557f70 : SUCCESS in 5m 00s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/90f65230962a4ffc9c0f6225aaf86570 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cab2b6fb6c6643e89f712d544e0b0682 : SUCCESS in 5m 42s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c1bf262cc74942d2afb1cbf452f551ef : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf52e7c82e614961a1113e681b5bff25 : SUCCESS in 39m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d617c93c9eb4dc4b937f11e023f0bbd : SUCCESS in 1h 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8fa9b05d72d3c972d01d6df7b14cedcb3a5b26a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 19:30:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4aa79dfc9ddd94f8de92ce7d6c2c19934acc0b4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 19:31:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847ca360f3544b94a31c3c6779baa2c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16084f15a6147948dfb9216529d4298 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f93a560abd0dad447c1904f7bcd43c78fd2f370","revisions":{"3f93a560abd0dad447c1904f7bcd43c78fd2f370":{"kind":"REWORK","_number":1,"created":"2024-01-31 06:17:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/907285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/907285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/907285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/907285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/907285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/907285/1"}}},"commit":{"parents":[{"commit":"a0fe9ec214d026fbfe1564cf1987bb6a92c5095b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0fe9ec214d026fbfe1564cf1987bb6a92c5095b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-01-31 06:16:13.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-01-31 06:16:13.000000000","tz":60},"subject":"Bump SQLAlchemy to 1.4.51","message":"Bump SQLAlchemy to 1.4.51\n\nThis is the most recent version \u003c 2.\n\nChange-Id: Icc16ef36b6a7f8a3341108d0bf512604c19a9c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f93a560abd0dad447c1904f7bcd43c78fd2f370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f93a560abd0dad447c1904f7bcd43c78fd2f370"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
