)]}'
{"id":"openstack%2Frequirements~907412","triplet_id":"openstack%2Frequirements~stable%2F2023.1~Ie87eb651e86c17c6f4cffa9bae5af35ea5dd501e","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"Ie87eb651e86c17c6f4cffa9bae5af35ea5dd501e","subject":"update constraint for python-manilaclient to new release 4.4.1","status":"MERGED","created":"2024-02-01 11:36:15.000000000","updated":"2024-02-01 17:17:06.000000000","submitted":"2024-02-01 17:17:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907412-new-release","meta_rev_id":"72b4c3b1536f6c96baa1bc53f4c765f8c939b1dd","_number":907412,"virtual_id_number":907412,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 17:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-01 15:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-02-01 15:43:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 11:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 13:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 15:43:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8f2a47b2a61df5b8879de1f04eff11a102c74761","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-01 11:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"572b72e995bfe4f798eecfe9e72f7185d02f6044","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:42:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/266a83c8435e4c6cb62f3ec1c6201e90\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7aad191056e8465db86e6e6463b2c244 : FAILURE in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6f5e12bda50b4e9e73d9e7e843336ee859eb6ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 13:02:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25e3424840e4abe9217d4e57ed34c30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3876c8841bee44fd8059071bc916dbaa : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5d8534856a94231aab49b5dc9a7c29d : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/edd4ee7acd43437f855604ee73c122fb : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d53877b62017402da56ecaa5cf67efd7 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0aed7749e52744c8a231f0795a065831 : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bcae1992e4f043b49064cb8739a80e21 : SUCCESS in 17m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9b7cceff7d1d49e2a45ad92af1135a11 : SUCCESS in 7m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/126cdfb887f4487882b284f5f0f4c92a : SUCCESS in 5m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d44267b60f37468ba5686e46ff55b3c3 : SUCCESS in 7m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9f9e6710c5da454a9521d7ec45853ddd : SUCCESS in 11m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bf8add710d98406cb82e1ce1b1b70387 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/74b5f507d14b498b9c75aa2efdca617d : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c6e891a03243494c8a6144bd1e2064ce : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a7f18aa0663f45af8dd1e52f6133dc63 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/976007498cfc4f8bacf490ba4820cc4a : SUCCESS in 8m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7b5969d36bdb498580d9a1c59db88237 : SUCCESS in 9m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/57b747caa4d34196a1fce5ce22cfecb8 : SUCCESS in 12m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d3ebeedd9e4b4517adf62970a64ae5ea : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/33bba35b35ae400c8d51a965d85783c8 : SUCCESS in 10m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fc116b5e0dfa45e08fa9c1cb21702ce1 : SUCCESS in 4m 51s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d91621350f334d09994e034d866ddb1e : SUCCESS in 9m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ddf3cf82046d45be8d0b28d53a128d67 : SUCCESS in 23m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dfe052641c1d4f8b81ed867f0df83ef1 : SUCCESS in 12m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1aafbfd8c2394c069d5af231f1090d62 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43d6c2df2e874511b698bfb0d0d10c85 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/002e03f2d93a4037b9de2ffd2eaa0af1 : SUCCESS in 22m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/abbfc13783c44508aa726992e5aab904 : SUCCESS in 5m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a392ad0fbb1f42f893a0f01ef55fd657 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c462d9052f54707960533cea812eb61 : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b716b9238ef344dd91e6e67e1f62288a : SUCCESS in 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a33c1660ace4c688883afc81a50dcd2 : SUCCESS in 58m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb5ea09cf4d04df58fbe2c26f20b7a89 : SUCCESS in 1h 20m 32s","accounts_in_message":[],"_revision_number":1},{"id":"df0cebb78ff80a899691884a58a5882a8bfa7c62","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-01 15:43:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b247113c61fd197265ce82244cabf91409c5fd0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 15:43:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57b7edc8d876e3cf1a135bfa8f29c87b8a1a5632","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 17:17:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15173decad8a45f482be1cc7f5b02e63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d42bfb6de54ea68489043f65c96ab1 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72477e171d6540049d9609393dd5698e : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7763308ad3fa4eaebcdbac835b8c8dd9 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6ba932b71d04f2397dfa6a374abd2dd : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e93a49ff11f4137b523d94784056498 : SUCCESS in 9m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b90ddc7c4c824ea699a6e620695ff22d : SUCCESS in 13m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/acbce74c18944ea3b8057449f606e599 : SUCCESS in 10m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6dc71f4c09ce4cc38e9e90af6e901e12 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dabf14cd07744a9380f406f516eabff7 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cba2646363ed466faeff71fcc408efa5 : SUCCESS in 9m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b473e542281d4f4297bd111ca4ecad52 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f10535adabf44fee9aa61e85fcdb04d8 : SUCCESS in 9m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/78f5b7401e3a4d89bea1a2a5f40dae18 : SUCCESS in 10m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/05b345e23d274db1813499f8193c512a : SUCCESS in 15m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0c49f1a7dd5b4eb398048bb2e93de33f : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ad424b8f98224dfabff502478b4c1341 : SUCCESS in 13m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c1d84e4fabf2486288b86e243b102f77 : SUCCESS in 4m 44s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/34c4b15c89b648b788b3ef5bf41ff847 : SUCCESS in 28m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ea7c35ecc2d744048ebd9d317d131211 : SUCCESS in 14m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bac92ddacd944521b4e844249ebafbac : SUCCESS in 6m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8e463eb3b264ec295f1fd73182dfa98 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c38e0054a5844f37ad5101f054cfef01 : SUCCESS in 24m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3ac9a111155942b2b65859c2e330d73e : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bb14a170835d47d9a1b7c77dcf0fb7a1 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a98ea0e2ef4463fbe400fe49e82bf38 : SUCCESS in 8m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e086a0ce875442c58f398f1adba4e9d2 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8035a3eee0ed4b71835ab5203b3102db : SUCCESS in 54m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3512ec5eb5244a78097cff05647c00b : SUCCESS in 1h 28m 23s","accounts_in_message":[],"_revision_number":1},{"id":"72b4c3b1536f6c96baa1bc53f4c765f8c939b1dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 17:17:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e66e5ef0a264f37f6608d5db6fb1084b7c421759","revisions":{"e66e5ef0a264f37f6608d5db6fb1084b7c421759":{"kind":"REWORK","_number":1,"created":"2024-02-01 11:36:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/907412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/907412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/907412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/907412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/907412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/907412/1"}}},"commit":{"parents":[{"commit":"2eb660e533ac6fc0e323b10a8bc2030f03ca0b9c","subject":"Merge \"Update OVS to the latest version of 2.17\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2eb660e533ac6fc0e323b10a8bc2030f03ca0b9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-01 11:36:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-01 11:36:14.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.4.1","message":"update constraint for python-manilaclient to new release 4.4.1\n\nmeta: version: 4.4.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: Iac8f54b5b18773f8078c006728523f0486171bea\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+1: WangYun \u003cwangyun@yovole.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie87eb651e86c17c6f4cffa9bae5af35ea5dd501e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e66e5ef0a264f37f6608d5db6fb1084b7c421759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e66e5ef0a264f37f6608d5db6fb1084b7c421759"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
